)]}'
{"id":"openstack%2Fswift~689888","triplet_id":"openstack%2Fswift~stable%2Frocky~I034b0902f620cbb274647e24566dff2031beeae5","project":"openstack/swift","branch":"stable/rocky","topic":"bug/1776476","hashtags":[],"change_id":"I034b0902f620cbb274647e24566dff2031beeae5","subject":"Unquote URL before using splited parts.","status":"MERGED","created":"2019-10-21 22:11:18.000000000","updated":"2019-10-22 17:54:21.000000000","submitted":"2019-10-22 17:54:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689888-1571766861488-f62c7b6a","meta_rev_id":"ad685e5cf97ceb132fc146312b439a98d6cbe9c9","_number":689888,"virtual_id_number":689888,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 17:54:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28111,"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","username":"wangyx"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28111,"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","username":"wangyx"},{"value":2,"date":"2019-10-22 15:20:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-21 22:11:18.000000000","_account_id":28111,"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","username":"wangyx"},{"value":1,"date":"2019-10-22 15:20:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28111,"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","username":"wangyx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 22:11:18.000000000","updated_by":{"_account_id":28111,"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","username":"wangyx"},"reviewer":{"_account_id":28111,"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","username":"wangyx"},"state":"REVIEWER"},{"updated":"2019-10-22 17:54:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba7fcbda229906ce769bd3531d0685084cfca453","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-21 22:11:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1eb866491a4ca2a59f6a3a27ba83f9d4343d73e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 23:44:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9bce77a865c048c3af7be31db39e383d : SUCCESS in 5m 26s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8c605afa78394e809100519c3a28deac : SUCCESS in 8m 59s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/c4099122c07e4660b13cd841c95816a8 : SUCCESS in 4m 41s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/ef08466d226c486a912989f61a1e3a77 : SUCCESS in 9m 25s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/5eecb79120f14c139d159c38c80cdb22 : SUCCESS in 10m 45s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/906c430387dc445abf29ba6da67b82c1 : SUCCESS in 11m 04s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/443b82039936416e9596b2ea53e0c7ea : SUCCESS in 11m 37s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/f9c75b99609f47b294b20531ab717df1 : SUCCESS in 11m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/410c761252a3466498c9ee355e316671 : SUCCESS in 33m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dad3b1de7bd49b8942254709da3b45d : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b5bdb7a3ef4c0cafbee0496c141808 : SUCCESS in 5m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a5e0cc539ae424c9370289ba9562816 : SUCCESS in 29m 43s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/0df9113a17a24220b6e20af768fba07b : SUCCESS in 1h 27m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/64a411504aa44a7aa1e6b63476ac6a16 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"01f5c87a551c91e49e4e85d2e03eaf2f272c7310","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-22 15:20:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport, test pops without changes under swift/, closes a user-facing bug.","accounts_in_message":[],"_revision_number":1},{"id":"31600b608f8c75abfaeadb390fd3a366bf64ec34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 15:20:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9262d35b439b20e18978d03ee566bb78348e7339","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 17:54:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a63a663af1b30981b4ddf72127d856d4e84edc08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 17:54:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/372820a84f0b45bebcdbf5d651de0379 : SUCCESS in 5m 43s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/00dc1204b3e140ecb97b2b7195c01f1e : SUCCESS in 10m 41s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/79c1895e469a4a0f8ea7378a0a147588 : SUCCESS in 5m 14s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/7f0a2779bb85471e8d24f970a3b358c9 : SUCCESS in 11m 47s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/6dca510031b347edb688a5d6996a21c9 : SUCCESS in 12m 46s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/e1ec64088f254c009ae30506f141bb4d : SUCCESS in 9m 58s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/e0e7efde7c9c4518bc19067f15affb39 : SUCCESS in 10m 35s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e910bb34d6224b988e385725fab545c6 : SUCCESS in 9m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/48be9463b8a1457e9e418be7c906ead2 : SUCCESS in 28m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/038efe29ebc047f39cc499b9598c5946 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76d300722bd84296ad701f9ab140507e : SUCCESS in 5m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b9fc79bbda684204b3251ac0aef98b73 : SUCCESS in 26m 41s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/aa89ceebbb53493fa664c4c969b9b502 : SUCCESS in 1h 25m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/46fb291961724e2e931b0337e2318334 : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"760efd1fe5cf913d692edb6e9c3e4f2b28130eb0","revisions":{"760efd1fe5cf913d692edb6e9c3e4f2b28130eb0":{"kind":"REWORK","_number":1,"created":"2019-10-21 22:11:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/88/689888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/689888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/689888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/689888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/689888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/689888/1"}}},"commit":{"parents":[{"commit":"c6e2fbb41775e065be7cdab2366d846785b57804","subject":"Merge \"doc: Fix the swift middleware doc needs more info  to set s3 api\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6e2fbb41775e065be7cdab2366d846785b57804"}]}],"author":{"name":"Yuxin Wang","email":"wang.yuxin@ostorage.com.cn","date":"2018-07-04 09:13:49.000000000","tz":480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-21 22:10:15.000000000","tz":-420},"subject":"Unquote URL before using splited parts.","message":"Unquote URL before using splited parts.\n\nKeystone EC2 access keys may contain URL reserved/unsafe chars,\nthen the access key in PATH_INFO can\u0027t be replace by account name.\n\nie.\n  /v1/abc%3D/c/o  --\u003e /v1/AUTH_PROJECT_ID/c/o\n\nChange-Id: I034b0902f620cbb274647e24566dff2031beeae5\nCloses-Bug: 1776476\n(cherry picked from commit 8c7ca01ea7ee0943d5b3078b941c15932f6eecff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/760efd1fe5cf913d692edb6e9c3e4f2b28130eb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/760efd1fe5cf913d692edb6e9c3e4f2b28130eb0"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
