)]}'
{"id":"openstack%2Fswift~690707","triplet_id":"openstack%2Fswift~stable%2Fqueens~I7c699e4fc46706a4971fce5a85ed335f471d3a2b","project":"openstack/swift","branch":"stable/queens","hashtags":[],"change_id":"I7c699e4fc46706a4971fce5a85ed335f471d3a2b","subject":"Fixing target for cross-ref \u0027get_data_dir\u0027","status":"MERGED","created":"2019-10-23 18:49:27.000000000","updated":"2019-10-23 23:33:24.000000000","submitted":"2019-10-23 23:33:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690707-1571873604087-b61b701d","meta_rev_id":"aa4f2935ca66612d75c043c873e6620ff4d31515","_number":690707,"virtual_id_number":690707,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 23:33:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":2,"date":"2019-10-23 18:55:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-23 18:49:27.000000000","_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":1,"date":"2019-10-23 18:55:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 18:49:27.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-10-23 23:33:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0bec801abca20d22406f383a06d8302b7be6576","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-23 18:49:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc1c94276618dc9350658b28844a07fab826a235","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-23 18:55:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMinor doc fix; if it\u0027s worth backporting to rocky, it\u0027s worth backporting to queens.","accounts_in_message":[],"_revision_number":1},{"id":"917999ee283ff55c51aca9b7261cf4ad5bed7393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 19:52:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ffa008abb3254e50a06386ea30b477a0 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/252bb4fc5b294f7d983d139270b0146d : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"98513dcb454bae8011af38bcb867b90dd570df0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 19:53:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3555ddb0b15c5dd8b4b4e8d2e0f28e541f7fd0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 23:33:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/897a8b32288f47f28efd2d0da7d889ac : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3869b31b4b94d95bb03d29de51267e9 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"aa4f2935ca66612d75c043c873e6620ff4d31515","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 23:33:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"829a5faa17e47afd860651406c36a74c3f46232b","revisions":{"829a5faa17e47afd860651406c36a74c3f46232b":{"kind":"REWORK","_number":1,"created":"2019-10-23 18:49:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/07/690707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/690707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/690707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/690707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/690707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/690707/1"}}},"commit":{"parents":[{"commit":"8d33e22606b31419aaa48cf73d69bc9a9aebf222","subject":"Modifying URL errors in installation deployment documents","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d33e22606b31419aaa48cf73d69bc9a9aebf222"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-02-13 11:19:31.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-23 18:48:45.000000000","tz":-420},"subject":"Fixing target for cross-ref \u0027get_data_dir\u0027","message":"Fixing target for cross-ref \u0027get_data_dir\u0027\n\nAt the moment, the `get_data_dir` ref in [0] is not\npointing to anything. This patch amends that and\nlinks it to (what I think) is the correct policy string.\n\nWithout updating this change, local tox builds\nfor swift documentation fail with the following\nerror: Warning, treated as error: /home/asettle/openstack/swift/doc\n/source/overview_policies.rst:555:more than one target found for\ncross-reference u\u0027get_data_dir\u0027: swift.obj.reconstructor.get_data_dir,\nswift.obj.diskfile.get_data_dir, swift.obj.replicator.get_data_dir\n\n[0] https://docs.openstack.org/swift/rocky/overview_policies.html#object-server\n\nChange-Id: I7c699e4fc46706a4971fce5a85ed335f471d3a2b\n(cherry picked from commit f21f48b73c500022f5ea309fd0a0cd5196a51b9b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/829a5faa17e47afd860651406c36a74c3f46232b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/829a5faa17e47afd860651406c36a74c3f46232b"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
