)]}'
{"id":"openstack%2Fswift~690745","triplet_id":"openstack%2Fswift~stable%2Fqueens~I43eee7029ca884268c3f2ad74300b94b299fd8d2","project":"openstack/swift","branch":"stable/queens","topic":"bug/1831790","hashtags":[],"change_id":"I43eee7029ca884268c3f2ad74300b94b299fd8d2","subject":"dlo: Respond 200 on multi-range GETs","status":"MERGED","created":"2019-10-23 20:50:38.000000000","updated":"2019-10-25 07:41:51.000000000","submitted":"2019-10-25 07:41:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690745-1571989311312-cd6aae0a","meta_rev_id":"0bea5bdecd944ebf5f5558f25fd0ae59bbfe72f0","_number":690745,"virtual_id_number":690745,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 07:41:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-24 21:30:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-24 21:30:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 07:41:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"598307ae1f2bea861a6f527b9ec42a389e336184","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-23 20:50:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dd9b44cd47f782ae40670dbb3628532c8f8618c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 23:40:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/46112aed23fd49e3b300a92d3aaca972 : SUCCESS in 5m 48s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cae7c042661d4b8ea268fd88e46d8771 : SUCCESS in 9m 32s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/bcfb0870a0894b138d1c7ab6576c2aac : SUCCESS in 3m 47s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/298f362164534065b6484bf851666436 : SUCCESS in 11m 38s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/81a369613d5e4542bf35671f2bce4a88 : SUCCESS in 12m 21s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/809cdd1fa1c84200a75b2af6afb6ff1a : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d237391d2ba40cd8f115f93d149abe2 : SUCCESS in 4m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9cf99a5adcd6430a9aba354d3ed42ecd : SUCCESS in 26m 10s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/28333b3e58e345098f4a536ad6211121 : SUCCESS in 1h 20m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d617da6ffd1a4e86bf5abf946814e978 : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a2e6269d2a20e1ebbf14a90fe243efb67e4be455","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-24 21:30:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"589e28352a8fb42c76ff1f3862823f33a85cf054","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 21:31:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3aabf0d5c1a405a2ee7ab137f68174e3fab7a84d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 07:41:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b837471d38826b9f3296af1a858d00ed59961ba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 07:41:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4aab4428158547a0bb7ef3d5d9971ddb : SUCCESS in 7m 17s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/594b4810b38541d9976f0894ac0648d8 : SUCCESS in 9m 33s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/7e08ef42495d42619f422ae07d667da6 : SUCCESS in 3m 39s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/1ce47b23d048414d90bdaf39fa35aeae : SUCCESS in 10m 24s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/d8451b980b424153a004b8dd0efcbc21 : SUCCESS in 24m 43s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/def9b9ff23ab44948ff33421fa5f1aee : SUCCESS in 25m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b63b7cfb01bb41848cc3eee96d12296a : SUCCESS in 5m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f86236a21444e3c86c505b9e72881b0 : SUCCESS in 21m 40s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/66576e228e0d4b268278b16a12b8676b : SUCCESS in 1h 24m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f3a9925fa8614e068dceb69fb0cf97c6 : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e738aa47f89917d39fa09539d51085400cf247c8","revisions":{"e738aa47f89917d39fa09539d51085400cf247c8":{"kind":"REWORK","_number":1,"created":"2019-10-23 20:50:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/45/690745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/690745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/690745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/690745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/690745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/690745/1"}}},"commit":{"parents":[{"commit":"8d33e22606b31419aaa48cf73d69bc9a9aebf222","subject":"Modifying URL errors in installation deployment documents","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d33e22606b31419aaa48cf73d69bc9a9aebf222"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-05 19:02:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-23 20:50:22.000000000","tz":-420},"subject":"dlo: Respond 200 on multi-range GETs","message":"dlo: Respond 200 on multi-range GETs\n\nWe already had a unit test for this, but it was inadvertently retesting\nwhat test_get_suffix_range_many_segments tested.\n\nChange-Id: I43eee7029ca884268c3f2ad74300b94b299fd8d2\nCloses-Bug: #1831790\n(cherry picked from commit a21c0be70ca6c200076d274c54f6906e1cf92a61)\n(cherry picked from commit b3cc7417e04900ff30a422fc5675ce18ab169725)\n(cherry picked from commit 6a02cac444f4d62282d4a73b2227204716ea8813)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e738aa47f89917d39fa09539d51085400cf247c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e738aa47f89917d39fa09539d51085400cf247c8"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
