)]}'
{"id":"openstack%2Fswift~691424","triplet_id":"openstack%2Fswift~master~I61d044a6d778e158af73a3f4d9fc75d642af0beb","project":"openstack/swift","branch":"master","topic":"691424","hashtags":[],"change_id":"I61d044a6d778e158af73a3f4d9fc75d642af0beb","subject":"Refactor timestamp iter in db backend tests","status":"MERGED","created":"2019-10-25 20:11:21.000000000","updated":"2019-10-27 05:18:23.000000000","submitted":"2019-10-27 05:17:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691424-1572153426367-34f7ffe6","meta_rev_id":"040683fbb66c8fb37ff82814604786a84e21b659","_number":691424,"virtual_id_number":691424,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-27 05:17:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-27 04:47:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-27 04:47:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-27 04:47:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-10-27 05:17:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18d62ec23ba21e06e0a8922dbca158c9d61606de","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-10-25 20:11:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b46d3b7deb1b3307963caa0fa77b8609b338360","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-25 20:39:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"090e0364cc38341a1439cb2a8733cb28ab0e98eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 20:50:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb891ca30954aaaa8c5bd6bc0b4f652 : SUCCESS in 7m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/91e4310df48d4228a04024db1d7af980 : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cb3f3dfedfbd4c1783e5f59c7d712017 : SUCCESS in 9m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4c673583c271496e862daea6204bf01e : FAILURE in 14m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2305e653aa5d453f8f0d7e55bd4253d3 : FAILURE in 10m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b832ceb459c247df814db96ffca95a88 : FAILURE in 25m 24s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/57b681f293ac44dbaff7f0dc890f7491 : SUCCESS in 9m 25s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/3ebadab8ba8145718ec4ab02d0a7d519 : SUCCESS in 12m 49s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/3f0f17eb155d4a4698edf7493e1c130a : SUCCESS in 12m 25s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/3f161b8e8a4e43c58f115673ea3d4465 : SUCCESS in 13m 59s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/772cc8cdb42a4657b0e264aeea75f95d : SUCCESS in 15m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6ddd746dcdb746dd9ba367a4b84103e2 : SUCCESS in 23m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/290f2212f021496894c983e83321e7fd : SUCCESS in 11m 23s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/696a06561e134217b0b6ee6d05ef29ad : SUCCESS in 9m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/550d04b365c947538b4e410d2197482e : SUCCESS in 12m 53s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d3a0db2eed5f45b787fb39b0c364d236 : SUCCESS in 16m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac597bd60ba640d58fcae1626c2db24e : SUCCESS in 25m 48s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/23618a8e163546948984a80a055c791a : SUCCESS in 28m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0071484f9b86434cb39c25cc8478b5de : SUCCESS in 29m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e519f3f997924ed390501ba990cbf9c9 : FAILURE in 13m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7628e3f3c1cd49e1a597a49a77bda25d : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b077871fc3b4d6e90d9c29d1324dd4e : SUCCESS in 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1b8316b44df393c24603cf3ca89f666ac2fe6a65","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-10-26 21:50:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a46a7ee2c710f3445a461e742191fc7ce9d1488a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 22:27:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1bce046102a4b1f91517015640da17a : SUCCESS in 7m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ada0c37a4433435d9f9a87fdb8a0fb98 : SUCCESS in 8m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/31b1800fcb434b28a66cdb0c34ecf19c : SUCCESS in 9m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb5c532f75504a51bd5c1e8447c0048d : SUCCESS in 24m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d568479c747649679da218eb67919758 : SUCCESS in 12m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b0d5035208d648aba19e58aa04b3619a : SUCCESS in 12m 53s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/b1d41ad2025a4b019f9d60082e8bf959 : SUCCESS in 20m 39s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/64789a7f1dc44005b8e9ffbe7472ee4f : SUCCESS in 9m 07s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/0e55424ec2974ff3875345defcb318cb : SUCCESS in 10m 10s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/3b9d1e97cd7c4156a657e5c575965765 : SUCCESS in 11m 37s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/d44041a1e86a43f9b3a7140630ed2482 : SUCCESS in 25m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/38ba6b0512eb41318832af7d812a81e1 : SUCCESS in 22m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/017edbce6bae4ec3aed01f6943106c1d : SUCCESS in 12m 57s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ce2558b72b074d9d87a9977a195da932 : SUCCESS in 9m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/057dc4552d044e739a94a1ef103e6630 : SUCCESS in 10m 06s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4733d00ff5f146acb1bc9c6749fc80ec : SUCCESS in 16m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4bae82c84cfc49909fa093154bf0653e : SUCCESS in 28m 27s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4303468dd6ec4997a72c71dc5a9b7056 : SUCCESS in 30m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1c7fdc56185a42c39084d4c476f2a635 : SUCCESS in 30m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aafe88c123d7439590cec381a0881f13 : SUCCESS in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abc12744ef254c3d8ab94f0710b05ff3 : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc2160a28adc4fa19f38af394653b57c : SUCCESS in 20m 41s","accounts_in_message":[],"_revision_number":2},{"id":"0b7dc625a098a2708f1b12caa5756ca6ad46e9a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-27 04:47:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nSolid refactor!","accounts_in_message":[],"_revision_number":2},{"id":"42c03cb03ab9b8e5de8e3c933124681984e15d12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 04:47:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ab28ca4d117d542a989afc4566baffa126c9ff5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 05:17:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8a83ad67059d47dfe185599da8a8e6c24b8f277a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 05:17:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f48d5c3ed2b44f9f98475d894dafd595 : SUCCESS in 7m 18s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/eef5d0d2f0b449e0ac0aea89b570d56c : SUCCESS in 8m 26s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7166aeeebcd540f9844df3bbcde77a1a : SUCCESS in 8m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/03a584f7d4464416adb757fdb3f47ac8 : SUCCESS in 14m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e2604ee38eec4b2dab7d7fc4cff4862d : SUCCESS in 11m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/04e0e12d46014687bc51f3f04b7096bb : SUCCESS in 11m 02s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/4dbd449c6f3046d49a08eef117490055 : SUCCESS in 12m 09s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/42d19225f2224b229073b9770a375d4f : SUCCESS in 8m 59s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/66eb8155880f45f98555455ad44e06fa : SUCCESS in 8m 59s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/3720d427ff554202b088fc4257789a1a : SUCCESS in 13m 17s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/a08c182efae140ac8c464547a423e045 : SUCCESS in 12m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9bebeb8c792b483a9e600a8a035db5a6 : SUCCESS in 9m 08s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/771f6791f4df4c4e8dce7b3c879c997c : SUCCESS in 11m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3d7ba2f3dfe641519cb548ca3c3763d5 : SUCCESS in 9m 45s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/fcbedff3628548e99c658ad47ed53019 : SUCCESS in 17m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f85b6f6165d34845bbca5eb9278600e3 : SUCCESS in 28m 37s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0bdc9fb3ff7a462b948ddc6188420258 : SUCCESS in 27m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fc492505532445098924dd6e62ca82f0 : SUCCESS in 29m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da6b392aae7f48dda0dc3457a2930e59 : SUCCESS in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5d695fcb3b34761891be5360de93b81 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"8d4d9c26fb948942ca9241bd2364e77994d90427","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 05:18:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/860f50ae9053402f8800758e9253bb05 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ed125480d9154f50b6be998e98a202f6 : SUCCESS in 39s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/81c585e449ff4cb6bb58812ff3b678b7 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b75d5939352c405ab82ab52c3a9a1c3ad030aa26","revisions":{"530b4a71deee4ac716c92d1047083e7f1335e3a3":{"kind":"REWORK","_number":1,"created":"2019-10-25 20:11:21.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/24/691424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/691424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/691424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/691424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/691424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/691424/1"}}},"commit":{"parents":[{"commit":"739c3704107bd991493472c18bfe7ac5f933077c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/739c3704107bd991493472c18bfe7ac5f933077c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-10-25 19:34:41.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-10-25 20:11:17.000000000","tz":-300},"subject":"Refactor timestamp iter in db backend tests","message":"Refactor timestamp iter in db backend tests\n\nChange-Id: I61d044a6d778e158af73a3f4d9fc75d642af0beb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/530b4a71deee4ac716c92d1047083e7f1335e3a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/530b4a71deee4ac716c92d1047083e7f1335e3a3"}]},"branch":"refs/heads/master"},"b75d5939352c405ab82ab52c3a9a1c3ad030aa26":{"kind":"REWORK","_number":2,"created":"2019-10-26 21:50:37.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/24/691424/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/691424/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/691424/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/691424/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/691424/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/691424/2"}}},"commit":{"parents":[{"commit":"739c3704107bd991493472c18bfe7ac5f933077c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/739c3704107bd991493472c18bfe7ac5f933077c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-10-25 19:34:41.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-10-26 21:50:32.000000000","tz":-300},"subject":"Refactor timestamp iter in db backend tests","message":"Refactor timestamp iter in db backend tests\n\nChange-Id: I61d044a6d778e158af73a3f4d9fc75d642af0beb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b75d5939352c405ab82ab52c3a9a1c3ad030aa26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b75d5939352c405ab82ab52c3a9a1c3ad030aa26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
