)]}'
{"id":"openstack%2Fswift~692942","triplet_id":"openstack%2Fswift~stable%2Frocky~I0f5e0e410c8297a4898caae00c196fa3b3862100","project":"openstack/swift","branch":"stable/rocky","hashtags":[],"change_id":"I0f5e0e410c8297a4898caae00c196fa3b3862100","subject":"Only url-quote Keys when encoding-type\u003durl","status":"MERGED","created":"2019-11-05 03:04:00.000000000","updated":"2019-11-08 06:16:58.000000000","submitted":"2019-11-08 06:16:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692942-1573193818642-83f09e0d","meta_rev_id":"3776b18f47bdda9d4989ab897fdc54cd20719a1e","_number":692942,"virtual_id_number":692942,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-08 06:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-08 03:58:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-08 03:58:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-08 06:16:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb6065eb95aed55c49211d610df752fefd00f403","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-05 03:04:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d8b905f2ba6d351938294334bc3cea862c3a386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 04:46:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/233b069e708f4f0b9ba078b9dfc6e2f9 : SUCCESS in 5m 14s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ab229defd4904a86890afc6ffa2663c1 : SUCCESS in 10m 46s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/a2acafed441749238aa86ce4c0ece4fd : SUCCESS in 4m 53s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/8cbf26e33d23442ea965e03fbaa468db : SUCCESS in 9m 16s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/678836653eb24fda9d803dfd41b870f8 : SUCCESS in 4m 01s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/bd3a2dc8d96d47868147f24351b757bd : SUCCESS in 4m 14s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/b9a383dc664a4ea49679aca067cd3dd4 : SUCCESS in 3m 56s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/a70cd7bff03244e2bf490c375f224909 : SUCCESS in 4m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/97472db92b94438389f24d458f8e9f21 : SUCCESS in 34m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/231b65d350b44e67ad2a1a007cc4c639 : TIMED_OUT in 41m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0b8ae2c6e654218afdcfca87bf384ab : SUCCESS in 5m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2fe34f22ab547d6bce24cbb7bedfa6a : SUCCESS in 26m 11s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/9e2937433d0d4eca9061c6c9062dbfbb : SUCCESS in 1h 38m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e2259c9e1fb04f25b2cc3d5256e7b0d5 : SUCCESS in 54m 24s","accounts_in_message":[],"_revision_number":1},{"id":"84d636c952a75c62dc3c6bc89d1fb903c104077b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-06 21:34:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"34ae3f30b878f895c48f0419fb8194f553f980cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 23:01:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b92def0af45b417e88f6416f03d4d194 : SUCCESS in 4m 56s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d3d8e008e8964cf286e41d98a65ebdce : SUCCESS in 10m 37s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/d5c10ce55d334102bad8ab3a5ff96f66 : SUCCESS in 4m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f7a590d166014a29b7c4f79040aa2c2e : SUCCESS in 8m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0256a85743504c51b5270dd181083635 : SUCCESS in 11m 04s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1f6393dde607489cbf4e8a400b1cd290 : SUCCESS in 21m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9ee22ade5e3e4225bc12b2964e19cd4c : SUCCESS in 10m 20s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/97ae9cc9248746d7b45874b9804a8a0c : SUCCESS in 9m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c78d3850cbff422b83975c3df4c9995b : SUCCESS in 33m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3f9ba50f8c14af5880c89e26b36c174 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a72298aaaf274379a2a603fb672cb29e : SUCCESS in 5m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/82b16aa269a8444b8aa2052a6ebbf4b2 : SUCCESS in 23m 25s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/691a849803a745eeac08e303024b83d1 : SUCCESS in 1h 21m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3b2456e2a2dd4173b910130d9b5c9cc2 : SUCCESS in 52m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7a1c66b65ff0c1a22c503b06a2e2d7ab6120a0d4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-08 03:58:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nBackport was a little hairy because of the list-versions API that was added in Stein, but looks good. Fixes https://bugs.launchpad.net/swift/+bug/1820847","accounts_in_message":[],"_revision_number":1},{"id":"7f09c90473aa339b8cb293cfb12bd4e9d1c4e6b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 03:58:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77b01c88f366ad14aec9d95296f646a615505197","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 06:16:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bc76b333f6b74409b84b4bc8c40632a4 : SUCCESS in 4m 52s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8417b8812ad9434cb57614989c46d30b : SUCCESS in 11m 48s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/76b77ffe47d34d47a589a8312a945dcf : SUCCESS in 5m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a2b07dda381447db900185952b95b899 : SUCCESS in 10m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/838ebd154c0a4fad9d4d0494b2a4207c : SUCCESS in 16m 56s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1b7c6a3644e54400aa9083c157f15eaf : SUCCESS in 10m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0956bed0d69247fc8aaf1e229046ac0c : SUCCESS in 20m 44s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d8be017655b741d2bde856cfc433320a : SUCCESS in 19m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/95e2eddf5c624362a778504db9d6e773 : SUCCESS in 31m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8673044c50f434da73142a030c9b487 : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8908bdf53d824a92a875dde82e214ec7 : SUCCESS in 4m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d6ded4c37d854817ba552d3bda877d4d : SUCCESS in 21m 36s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/32369253fab54ec2b8299f8818c38e8d : SUCCESS in 1h 24m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8b0cff4711e14e08b6e9433bd829fce5 : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3776b18f47bdda9d4989ab897fdc54cd20719a1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 06:16:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"345f2b542bd307ee039141ee6e535653f3d7a2be","revisions":{"345f2b542bd307ee039141ee6e535653f3d7a2be":{"kind":"REWORK","_number":1,"created":"2019-11-05 03:04:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/692942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/692942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/692942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/692942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/692942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/692942/1"}}},"commit":{"parents":[{"commit":"da83e066aaeca60441b48ad01e7a52f66a364cbc","subject":"Merge \"dlo: Respond 200 on multi-range GETs\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da83e066aaeca60441b48ad01e7a52f66a364cbc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-10-18 21:32:36.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-05 02:45:32.000000000","tz":0},"subject":"Only url-quote Keys when encoding-type\u003durl","message":"Only url-quote Keys when encoding-type\u003durl\n\nPreviously, we\u0027d even url-quote ETags, leading to weird things like\n\n    \u003cETag\u003e%22a27a822070b843ed6407f4f38afdaa20%22\u003c/ETag\u003e\n\nin listings, where those \u0027%22\u0027s should be double quotes.\n\nChange-Id: I0f5e0e410c8297a4898caae00c196fa3b3862100\n(cherry picked from commit c1c65a7e9f43f1395c63fb7bf22f812962b53714)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/345f2b542bd307ee039141ee6e535653f3d7a2be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/345f2b542bd307ee039141ee6e535653f3d7a2be"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
