)]}'
{"id":"openstack%2Fswift~692958","triplet_id":"openstack%2Fswift~stable%2Ftrain~I96ad9e5ed87a7e08d06e889233730cec476d3af7","project":"openstack/swift","branch":"stable/train","hashtags":[],"change_id":"I96ad9e5ed87a7e08d06e889233730cec476d3af7","subject":"Make our py2 func tests more explicitly py2","status":"MERGED","created":"2019-11-05 08:34:50.000000000","updated":"2019-11-05 11:45:08.000000000","submitted":"2019-11-05 11:41:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692958-1572954085111-7361ecdf","meta_rev_id":"b89786ca8909fe4906c88dc6b3dcfb5af57029ae","_number":692958,"virtual_id_number":692958,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-05 11:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-05 10:00:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-05 10:00:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 11:41:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8de2d15c0739095885e25af6debffa459209ba2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-05 08:34:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cdd9640cfd1b0c9500f25ff86ef7d57130721230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 09:55:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afc98c6e3bcb4a54b3703ba083395bfa : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bef855c4a8e241e389db5c3b9bb097b1 : SUCCESS in 7m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b32d80db5d6d44958ed5417b75010935 : SUCCESS in 55m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6053368f703a45a78c66c99fd18703a0 : SUCCESS in 48m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6c272a5bc9c47358cc1e1393f04d742 : SUCCESS in 8m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/203acc0c62d44646b9dfca1405a14563 : SUCCESS in 14m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/496386160166451e8363d81d71354d06 : SUCCESS in 12m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7bf23b4bb6a44e029746d9cbef3bb766 : SUCCESS in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d4a5be7676a14c57918abe461c8f11c4 : SUCCESS in 12m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ec3a7e9f34484b609e3c08873b38f397 : SUCCESS in 9m 33s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5a3f58d204f145f29f11a61b1ad75581 : SUCCESS in 13m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f548c3715749401c895af46c2e9e614f : SUCCESS in 10m 18s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/b85281b5f0e44ccc9622a91a1806d252 : SUCCESS in 14m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bbec8d754bc84af09db655971f113be3 : SUCCESS in 11m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1d6c21c205154562bb6c292f008774b7 : SUCCESS in 12m 56s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f09491ae7e9d44cd85e50be577df6741 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4c1f70903c5848a79cef41b98f318330 : SUCCESS in 26m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/85622ef75e1440e68b24351ad600f234 : SUCCESS in 20m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ac26c1c93b3047d785e6b9bd518200aa : SUCCESS in 27m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/70f428ad6b694e83aff45bef3dd4f708 : SUCCESS in 39m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1f0fdf67ede41f09c96858a78d5bbfe : SUCCESS in 48m 38s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fc80eb0249914822904489b34bea8243 : SUCCESS in 38m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e04f648c3cdf4e5caf9d2a89c023f812 : SUCCESS in 48m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/583e74477d914f0da9ad81e230b20a37 : SUCCESS in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f59d8223d49409999abde415fb00cca : SUCCESS in 7m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65a8a563930b411a8c6a3db2c57654be : SUCCESS in 26m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24df4a53049940e6a2779f540537e59e : SUCCESS in 1h 18m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ab5215dbf88f8da721a1c39d0beeaa22165e8d98","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-05 10:00:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport, more explicit, has us *actually functionally testing py2* again...","accounts_in_message":[],"_revision_number":1},{"id":"ebdbc895b6e7a542fa30f809c7d5906cdcc594f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 10:00:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3182b25ce75469982215e8659ee1d68972907b8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 11:41:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ff178344d24d848165e8b2b72713dd : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec02fd0e14f9469fa9109644ef079269 : SUCCESS in 7m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ddd7a49c4c924d5c9cdce338bd77e9eb : SUCCESS in 54m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/98f5817ecd29425b8e4b6e2035e08179 : SUCCESS in 55m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/de19797723f646a1ad4e7269457923c3 : SUCCESS in 10m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8fc7ceb7098b45108e41c06ff8d6da48 : SUCCESS in 13m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/89b5a64d8d89441589dd1950dbda1619 : SUCCESS in 11m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/47553f61d2744ebb875a56a46dc75aea : SUCCESS in 14m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/158e4205070843d5bdb517a01ec5fe70 : SUCCESS in 9m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a7b9d91ac57c4f8fb25f978ce37323ee : SUCCESS in 16m 19s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/3a460730c8f14ca8945d85e9c974aa34 : SUCCESS in 9m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/648ed9007c6e4935acb112cf99e017e5 : SUCCESS in 10m 33s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0e8cda93b67841bfbb8cb6da92d30155 : SUCCESS in 18m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f44d4aa341b2479ab0a69bc48e858866 : SUCCESS in 10m 45s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0c75f5525fc8427c86d7beead9d0b534 : SUCCESS in 13m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1940ccbe007545f48ebaabe75ef63431 : SUCCESS in 17m 05s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c3df887173474304b928d0e30a62a954 : SUCCESS in 13m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a622e2385978401592aae2ef814e6bd6 : SUCCESS in 35m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/18c7efe9566c4e41b6e2f8a557de9999 : SUCCESS in 28m 57s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/44f86777f42c4071942325158e7f6d01 : SUCCESS in 28m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5aae5720e19241d58b67c6cfa97d67eb : SUCCESS in 39m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da3e5dfc4c614ad383634c56d3548ec2 : SUCCESS in 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3186e07497ed4e668049db56a4cba3e0 : SUCCESS in 5m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/defbd619f4bc450496f85f4e2a112455 : SUCCESS in 1h 04m 57s","accounts_in_message":[],"_revision_number":1},{"id":"497aebc8cba35d0c5b282924bab64bce7517c5f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 11:41:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b89786ca8909fe4906c88dc6b3dcfb5af57029ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 11:45:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa857279b4ef40f3b9b539e80508b23d : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/799689fe0e074d238407c3cd663d3205 : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0a2c30a7e3cf41eb98aa3aca917cda71 : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a96e8c05c6268891c0bac3dfc7fb9ac50b2c51c3","revisions":{"a96e8c05c6268891c0bac3dfc7fb9ac50b2c51c3":{"kind":"REWORK","_number":1,"created":"2019-11-05 08:34:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/692958/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/692958/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/692958/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/692958/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/692958/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/692958/1"}}},"commit":{"parents":[{"commit":"c23f4fe24122522617e217780374e899a98a4747","subject":"pin lower-constraints to run against python 2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c23f4fe24122522617e217780374e899a98a4747"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-05 01:45:26.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim@swiftstack.com","date":"2019-11-05 08:34:50.000000000","tz":0},"subject":"Make our py2 func tests more explicitly py2","message":"Make our py2 func tests more explicitly py2\n\n...because otherwise, I think they might be running under py3 right now.\n\nChange-Id: I96ad9e5ed87a7e08d06e889233730cec476d3af7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a96e8c05c6268891c0bac3dfc7fb9ac50b2c51c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a96e8c05c6268891c0bac3dfc7fb9ac50b2c51c3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
