)]}'
{"id":"openstack%2Fswift~693007","triplet_id":"openstack%2Fswift~stable%2Fstein~I4e7e71cbaeda9e6bfcf4e415dd5f859df4835376","project":"openstack/swift","branch":"stable/stein","topic":"692963","hashtags":[],"change_id":"I4e7e71cbaeda9e6bfcf4e415dd5f859df4835376","subject":"Fix stable gate","status":"MERGED","created":"2019-11-05 13:57:14.000000000","updated":"2019-11-06 00:52:50.000000000","submitted":"2019-11-06 00:51:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693007-1573001461203-83f09e0d","meta_rev_id":"f7aa47331f481754b4f95ef1f8dc3f704f988156","_number":693007,"virtual_id_number":693007,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-06 00:51:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-05 22:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-05 22:17:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-06 00:51:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e2b24fe88ff762a9c29eea100f5fdbcaee10147","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-05 13:57:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e83efedc2938288c97280453f3e7b467559d274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 14:47:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/847fa245accf47d39a9dbdc06ca707b2 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40b9c67094c9401485187c93e0621c79 : SUCCESS in 6m 53s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c1cf967b48854dd3aff7fbf5f6bbd127 : SUCCESS in 15m 06s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/ebf03e3422d04e49843df4d10f377726 : FAILURE in 17m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c95609f1b4c641f49d28d6dc0918268d : SUCCESS in 10m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/837d8948ad2b4ead8c197fd40a54b170 : SUCCESS in 8m 55s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fdd40e57f1824d40bad279b8e5324532 : SUCCESS in 8m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1ac1362f67ca4d7f8e959e0d12afb919 : SUCCESS in 9m 49s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/98af7e75f05c4a47a70ae6ec52fb6aa4 : SUCCESS in 17m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3a37bbead0544ceaaeff9a151b430b07 : SUCCESS in 11m 22s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/406eae364454436aaa7fdff497fd2459 : SUCCESS in 22m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/856f9d39335b4f248859798933bcf84b : SUCCESS in 16m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bf1334d1101c412bb13eae249422eb89 : FAILURE in 38m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/78e3ddcf94064c859b3a56878a1599d7 : SUCCESS in 28m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3777383cafb947209d7aac15af663aec : SUCCESS in 27m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/866f031e05114aeea09e756e5bbf35da : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf4d506c062d4a36842824eb3308f576 : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21c84c5d76b54a40b873b5b0325bce9e : FAILURE in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f2e6c0e2a6e4539075a43f2a78099c9c0062ac6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-05 22:17:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a030252fc8fe9812a4a0c9b05efc8652405067e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 23:01:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a5a1f3589f4808ae125cc84af49065 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7adcc1d000264fa49d4b816fa766e5b8 : SUCCESS in 6m 37s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e0f1f61ea9cf413f96c4eae9213a2e64 : SUCCESS in 18m 27s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/54ff063f0e784d729e28fda32e0916dc : FAILURE in 14m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e0ba99eaea334e51a915a92b85d2ffff : SUCCESS in 14m 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/84c04ca04cd94fe0b2e674c628915eec : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/96f703b2351d4129a1939781901cd14f : SUCCESS in 10m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69988f2324b14cef89cd6bdb029470c8 : SUCCESS in 12m 21s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/def4774a995b4fe5bf89e407d5697699 : SUCCESS in 13m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4ebaddfe048d478db465b36f9d1df63b : SUCCESS in 11m 36s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5285002881564571bd7539e65fcbf175 : SUCCESS in 21m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/52e83fcc43414b87ae959f14ac543257 : SUCCESS in 20m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/694f611b1ac040d79e2dc26988d1929e : SUCCESS in 30m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/271523d142c14a598d8f17fa9c9af071 : SUCCESS in 33m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab8fbc6c121c434f8384bbb06ef47e1f : SUCCESS in 33m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/651d2e9338e749efa0ad25f9a857d7ba : SUCCESS in 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/458e10eea67b4bdab30cf33804bbc78c : SUCCESS in 8m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d4e4c22ebf64ea689c43666af34aa47 : FAILURE in 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04f1b03cdf5b6f696a65f95b8437685445abfa67","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-05 23:42:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f821d6a948f9f36841d500dd760e954f9d48c563","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 00:16:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ad21280165475da5503184d0865e33 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9abedc9d0e414e1e867e1525b948edbe : SUCCESS in 8m 16s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/80068113349a4a429af79626b4e9613c : SUCCESS in 11m 59s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/fec23330377f4c02852e6df0fb3dedb3 : SUCCESS in 10m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aa5adcad05a849438096a014205aa30e : SUCCESS in 8m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6927f4547bb64461a92474f918fada1d : SUCCESS in 10m 36s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a4c456769e444107a27a336b84835d72 : SUCCESS in 11m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b40c6d521ea84fada79db595c42abd79 : SUCCESS in 8m 38s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/eb04766d061b4168ad6fe3d339584e66 : SUCCESS in 10m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b5bfc1cfc17d4be499d686b6dd1fdc38 : SUCCESS in 11m 04s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/bdcd32f3c0c240a8b3c384a65f095221 : SUCCESS in 9m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4fbbae178ccf40ada568b761e1ae6fe5 : SUCCESS in 11m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1d3e14d4e2634ddf8a97400e1cd920c3 : SUCCESS in 32m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/620740fa3e694ef9966e89fa1af261aa : SUCCESS in 28m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/13aedc257c654112abda8fb46075001c : SUCCESS in 23m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed175a089e6a4609af660f389447d05b : SUCCESS in 12m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d2d4617bea642a3af923952539e5e05 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65d373c2a25049f89a9a2a4c073ed580 : FAILURE in 19m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"270feccd32673ac70baf406716f629770a9e133e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 00:17:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12965e72fb3915ce0e8b09e09e4bc57272eea122","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 00:51:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cf3dbd8575240bf996e17c2f28fe9e6 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff62df4f2e81459480b1b53f99cb2553 : SUCCESS in 6m 29s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/efc0c666977643fc8bed512f79e3525a : SUCCESS in 14m 14s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/a94a366038d04a7d93500aa8c484bf34 : SUCCESS in 10m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/da460ef07ef94405a3c25651d72597de : SUCCESS in 9m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c432a2447df04102ba5c284c2245c9c0 : SUCCESS in 11m 21s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/34c9629777864ca8be546828f5ae72a4 : SUCCESS in 11m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a8329d00e134a3cad6712730c6ffb49 : SUCCESS in 8m 51s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/96e0ac3627be4aadbddcb8abb4354d8e : SUCCESS in 8m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/520dd13d993a439790b6f07e2bf2b769 : SUCCESS in 33m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c2950781228490fb34cb17b0baca41b : SUCCESS in 24m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77fb5767c1f245668bb4aed2ab8f7567 : SUCCESS in 25m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3a4e9d6766243439903847062e820ea : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b657d6220448e281858f9d90f18b10 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ce195dde6b6940b8733ba97c9f24905f90709507","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 00:51:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6ac2602fe99cd4ea01a3aa49ea9bd1bf55fac30a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 00:52:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/215af38f2bb1420ba8fbaf04928d5eef : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3207e8f37b04693babfa7e3c0201a2a : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56ae0c7debd174bd5676ce937482bf9414efa3de","revisions":{"56ae0c7debd174bd5676ce937482bf9414efa3de":{"kind":"REWORK","_number":1,"created":"2019-11-05 13:57:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/07/693007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/693007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/693007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/693007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/693007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/693007/1"}}},"commit":{"parents":[{"commit":"12fd2392740aa343c356895c0aa10d75e25cdd6c","subject":"Merge \"Sharding: Clean up old CleaveConext\u0027s during audit\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12fd2392740aa343c356895c0aa10d75e25cdd6c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-05 01:45:26.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-05 13:56:29.000000000","tz":0},"subject":"Fix stable gate","message":"Fix stable gate\n\nThis is a combination of 2 commits.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nMake our py2 func tests more explicitly py2\n\n...because otherwise, I think they might be running under py3 right now.\n\n(cherry picked from commit c512574e110c214396658743e75ae497ba76e2ff)\n\n----------\n\npin lower-constraints to run against python 2.7\n\n(cherry picked from commit 281ffab2a52f348aa3f8ad39e800a43c6047e1cf)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nChange-Id: I4e7e71cbaeda9e6bfcf4e415dd5f859df4835376\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56ae0c7debd174bd5676ce937482bf9414efa3de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56ae0c7debd174bd5676ce937482bf9414efa3de"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
