)]}'
{"id":"openstack%2Fswift~693557","triplet_id":"openstack%2Fswift~master~I6b968d3707fede6d62974c31e1b741c81ed66907","project":"openstack/swift","branch":"master","topic":"bindep_docs","hashtags":[],"change_id":"I6b968d3707fede6d62974c31e1b741c81ed66907","subject":"Add libsrvg2 packages to bindep","status":"NEW","created":"2019-11-08 16:18:44.000000000","updated":"2019-11-08 17:33:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"830c956ff7dbba9428d1c4e9e31ceded454acc71","_number":693557,"virtual_id_number":693557,"owner":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-08 17:33:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-08 17:33:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6922e33bdb0fa4b5ba4afcce1a8131470ee7f12","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-11-08 16:18:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"634475513d87a30fe170702de3e64ddc47cce7d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 17:33:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6b98168e76014b11b3fa09374dd38cd6 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04fdffb91d384af69ee0ce841e66d47e : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21b5d39ebc2e4c039978e704620bca42 : SUCCESS in 7m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6de0f52a6d054c9f86aa1d224272285b : SUCCESS in 51m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3051611ab07542878079796f48b77494 : SUCCESS in 50m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c6661fc6138b4921bede2fcb62e9b7bd : SUCCESS in 7m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a4d710c5849d4a4dac0e59c44ed29f34 : SUCCESS in 11m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/67813c9e91d845c99fbfb915a8329940 : SUCCESS in 15m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c4a01bcce5814d9195b4c012c83cc660 : SUCCESS in 17m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4eadef2775324831bdd88043c89493ec : SUCCESS in 26m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f4546a7f21f948758c73d11970244ff5 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c7c65a1feb91464a87e5e794a7b38d5c : SUCCESS in 10m 09s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/997b36b54d9445eb8b1df8642ca0f860 : SUCCESS in 10m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9be65b72e275447a98b146767f6fd55a : SUCCESS in 15m 23s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ae9338c378e9491dbc85bd1b62b6b951 : SUCCESS in 14m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6083934a0d414951a3b7d5a7b1f59550 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/946d2478bf9d43518894ace9bd090a45 : SUCCESS in 11m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5f94fbc039024b4a9b8f9775d55f044d : SUCCESS in 25m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4d5adf59c3ac411a8dbd32ec5cde8d58 : SUCCESS in 17m 41s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/701d774221a2465f847516fd69ef8eaf : SUCCESS in 13m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/22612cd0709246488bf4f5c1e26cdbc5 : SUCCESS in 15m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bca9a98a54a94b4583ca148ec01fcc4e : SUCCESS in 36m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ee235acdf5c40d6a6bac220074b86b7 : SUCCESS in 44m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f5f7891d972d4f858988aec5481605af : SUCCESS in 33m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e2e4477d46f4703b31e989dd29dc722 : SUCCESS in 30m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6c1ebd8edc042df9c55d55d4cd64aaa : SUCCESS in 27m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f80ca0b09b9546d9b6a9e9dde89dcceb : SUCCESS in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39ba5754d5df4b03b3e6955bb0991ae3 : SUCCESS in 19m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17e64fdd8a9d4a32b263b8839ef02f34 : SUCCESS in 1h 10m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ce803d1283a6d25e8962052db201817a82d2633","revisions":{"9ce803d1283a6d25e8962052db201817a82d2633":{"kind":"REWORK","_number":1,"created":"2019-11-08 16:18:44.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/57/693557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/693557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/693557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/693557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/693557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/693557/1"}}},"commit":{"parents":[{"commit":"4b78d66ff6f6be09db6d6aa23edc1083ebbac3cf","subject":"Merge \"print deleted flag in account and container info cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b78d66ff6f6be09db6d6aa23edc1083ebbac3cf"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-11-08 16:17:49.000000000","tz":120},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-11-08 16:17:49.000000000","tz":120},"subject":"Add libsrvg2 packages to bindep","message":"Add libsrvg2 packages to bindep\n\nlibsrvg2 is needed for sphinxcontrib-svg2pdfconverter in docs builds.\n\nChange-Id: I6b968d3707fede6d62974c31e1b741c81ed66907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ce803d1283a6d25e8962052db201817a82d2633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ce803d1283a6d25e8962052db201817a82d2633"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
