)]}'
{"id":"openstack%2Fswift~693893","triplet_id":"openstack%2Fswift~master~I4211cfbd23eeb7badd834c1108823b416a186265","project":"openstack/swift","branch":"master","topic":"bug/1634382","hashtags":[],"change_id":"I4211cfbd23eeb7badd834c1108823b416a186265","subject":"Put storage policy index in object-server responses","status":"ABANDONED","created":"2019-11-12 17:32:58.000000000","updated":"2024-01-13 05:49:03.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e0bc75536c0602fba6969c9d414a06b1c43d63dc","_number":693893,"virtual_id_number":693893,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-12 18:34:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-11-12 22:35:33.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 18:34:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0798d874859ac8353a62b4698482861b9ee6bb9e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-12 17:32:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50ffa174140b16334d73ed28143083a7b74d9c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 18:34:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8a0c46369b4dda9e30d0671231c1c8 : SUCCESS in 11m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3e9aa7d7b2574365a537226427f6da8a : SUCCESS in 58m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20128a406699476e9f9ce419ac5b029f : SUCCESS in 50m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a0295f6a489545398a78e5c44ef12f61 : SUCCESS in 7m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8730731ac3be4272bfe5344101cdb5af : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c3f0f5b2d72b4bab955b78f9eb3ef0a4 : SUCCESS in 13m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/750847603c8c4740b745f51b5a48883d : SUCCESS in 11m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0a6e84d30e8d4350a1202b8a99664843 : SUCCESS in 13m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/af4f4defdc154fe582b973531767e141 : SUCCESS in 10m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6e46d3ca05f54ca7b422ffcf0b0a83ab : SUCCESS in 10m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/cd0edd1cb479409c80c516c5b243759e : SUCCESS in 10m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e53b7d673eda4a8c9a6050e6356d8b8e : SUCCESS in 11m 46s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/fc629992a6eb4fa89e7bce90ea79874d : SUCCESS in 13m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a3effd0ed9824b8cb896da10e1ad9420 : SUCCESS in 11m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7877bedc64eb49a1a324cea978672955 : SUCCESS in 10m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/55d9ee16be0448679c8e402b33e41bf4 : SUCCESS in 10m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/832da28c9e654fafb5a57c65bf6dd8b8 : SUCCESS in 10m 19s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/de0f9c28ccab44b29fd0e901d0962012 : SUCCESS in 26m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4769851580784a2b9913399ab1fd5b25 : SUCCESS in 12m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/587f8f2a2b8a4e058ad29400d6add10d : SUCCESS in 31m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ae140613a9a46b7a59a886bbb955eb7 : SUCCESS in 30m 38s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/51712eb631014cf38aa548e771920686 : SUCCESS in 29m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/158623517d1746069baeb23bc43e14ec : SUCCESS in 27m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e48b08d0268c4b518d12533c997a15bf : SUCCESS in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/475c53a1dbb24facb2c6198b58f93cb2 : SUCCESS in 7m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e9e3875a8b847c9bc074c008e10f6e9 : FAILURE in 18m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efebb029b79e43a188dd81554b27bf8f : SUCCESS in 53m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b85f732320870d9698983ac6d626634ff085dd2f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-12 22:35:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0bc75536c0602fba6969c9d414a06b1c43d63dc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-13 05:49:03.000000000","message":"Abandoned\n\nSee https://review.opendev.org/c/openstack/swift/+/898783 for a more recent take on the problem","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c55b49e6911204d0559cf1b9b126636f24dfe34","revisions":{"6c55b49e6911204d0559cf1b9b126636f24dfe34":{"kind":"REWORK","_number":1,"created":"2019-11-12 17:32:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/693893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/693893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/693893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/693893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/693893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/693893/1"}}},"commit":{"parents":[{"commit":"7f6b09dc930706d63a943dc239dff84ddae52014","subject":"Add delete_account in InternalClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f6b09dc930706d63a943dc239dff84ddae52014"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-09 03:51:00.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-09 03:51:54.000000000","tz":0},"subject":"Put storage policy index in object-server responses","message":"Put storage policy index in object-server responses\n\nNot going to worry about 507s and the like, but 2xx, 404, etc. could\nbenefit from returning that info.\n\nCo-Authored-By: Christian Hugo \u003chugo.christian@web.de\u003e\nChange-Id: I4211cfbd23eeb7badd834c1108823b416a186265\nCloses-Bug: #1634382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c55b49e6911204d0559cf1b9b126636f24dfe34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c55b49e6911204d0559cf1b9b126636f24dfe34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
