)]}'
{"id":"openstack%2Fswift~694650","triplet_id":"openstack%2Fswift~master~I3833843cd8d23c498a2afe6c68a3f0afe26343c0","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I3833843cd8d23c498a2afe6c68a3f0afe26343c0","subject":"Update known-failures and config for up-rev\u0027ed ceph/s3tests","status":"MERGED","created":"2019-11-16 00:20:59.000000000","updated":"2019-12-04 19:20:18.000000000","submitted":"2019-11-18 23:39:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694650-1574120361597-8d7acffc","meta_rev_id":"196ef247857e9d3585094e441a952dd6b7a7ab77","_number":694650,"virtual_id_number":694650,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-18 23:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-18 16:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-18 16:24:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 23:39:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c624ac37509d3b2abfbe159f37dae9f9bd59dbdb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-16 00:20:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7843bd547d270ff5dcba6b34ca742f1235eaf2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 01:30:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/377be61c65534ea89911e36bcb6fe868 : SUCCESS in 6m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6788180165b14870b9cffb0625592f78 : SUCCESS in 55m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cc121b0e9773433096258b2d31c1dae9 : SUCCESS in 53m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/549fde4f97e9455bbc938b342f256214 : SUCCESS in 8m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4b1766d07d24434c9fbdaa1eb6d6e0c3 : SUCCESS in 11m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/99c7f68caeee476bb2b1ec046af16e91 : SUCCESS in 12m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ba966365bf3049dbb1a719348c7bd617 : SUCCESS in 11m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5a314f1ce599494da86878c81bd5d353 : SUCCESS in 10m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/35d8671227404c3687137fa43a01b943 : SUCCESS in 10m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5f4f8ab8e2954002838e8bb3187a7664 : SUCCESS in 10m 12s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/caede79bdb1c43d1a8e3878fecba4bc0 : SUCCESS in 9m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f211e52b38324536bb748176844c1007 : SUCCESS in 18m 16s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/55ea8614dd974087a826848db4f69178 : SUCCESS in 14m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bc333df70cec49a7b64e3ced729014f6 : SUCCESS in 11m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/eaffe5059e0d4aaaa58814cc22258ee9 : SUCCESS in 10m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/747807b1505341a587dcc8d37c4ba4cb : SUCCESS in 11m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f18ba9f87e5a4979a15f79f0a5314144 : SUCCESS in 12m 31s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/94cafb3aeb644934baff014a3340f165 : SUCCESS in 15m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/813545cb02cf41cbb5280d62030d160b : FAILURE in 25m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0a13131ea74b4470913916dc4cc22869 : SUCCESS in 33m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/735ead02198c485e80bea33ca4f7274a : SUCCESS in 27m 21s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e1030893f4f144ae98ce3fc16725e573 : SUCCESS in 29m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b6d3192ceb6461e96e3b395bab65a2c : SUCCESS in 28m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19de7bb247584bea967c971c95c88654 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/943f17e791ee4c35ac788f11c656e969 : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/57360d7414ce4d7fa63cd1f397e874e6 : SUCCESS in 18m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/324b05109d8f42a981abcb4a4ad526aa : SUCCESS in 1h 05m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9598de280f523e0a7b809496a05d069c8d1d8de1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-16 22:24:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"00be02a7600995ee7412c75d45294f0d776cfc23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 23:34:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb270f667740460da1aac6fd20161684 : SUCCESS in 7m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ba526051cbaa416c8917ce54311db8f0 : SUCCESS in 50m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/23b806fb12244662a1f73859adae8dae : SUCCESS in 52m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87ee350459f04cd0913593afed0a615d : SUCCESS in 8m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c7eb0803c74045d8ada6182ea132717f : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/617d4c933d3749ba8ceec559115c1b5b : SUCCESS in 15m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a292463448ee4012878e312d85b2d15c : SUCCESS in 14m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9e450e82b119417e82d74c35722446e9 : SUCCESS in 11m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/112531753b224412b71d1ff482bb216a : SUCCESS in 8m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b8c2e01011941008cb0c947fc42d271 : SUCCESS in 9m 33s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a4dbb4f7f615449290bc12fde28b6c23 : SUCCESS in 8m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a6151bc2b44e4a24a2b189700d629f6b : SUCCESS in 9m 02s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a64b06c9389f41daa62c3165c337c2f9 : SUCCESS in 12m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6054caf34ab647beaa700c50380d7c82 : SUCCESS in 9m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a457f4eb54c942318807b932c0f3af51 : SUCCESS in 14m 24s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4f51bc52ea0040f59ed7b4f7c41a4779 : SUCCESS in 9m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6740ccce197f4663840cf2900e21e9db : SUCCESS in 9m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c4a842c2582241cfade796bbe2c2b874 : SUCCESS in 12m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1241eb6f2aa74752bc916a73254d2486 : SUCCESS in 14m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/44401fc11ea24aec86c670104c2a529e : SUCCESS in 33m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/576a5bb61d9746a69446513503d8c472 : SUCCESS in 26m 38s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4ef8128fae0a4b97a78a138dd7297edf : SUCCESS in 27m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/173d07a5f70f4e328c70d9437102c8d0 : SUCCESS in 26m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/934780c8aa4643cfb3521066f2760c8d : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7acac92213346c3b51268ce15d20d9c : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b674d73e18c46ca9766cbdd1286b3aa : SUCCESS in 15m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85fe3bd563334a0f822b2c3756b71fd6 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":2},{"id":"26ff7a2b212b893413e229e48e0c0d2da7412d31","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-18 16:24:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nTests pass again; let\u0027s merge it!","accounts_in_message":[],"_revision_number":2},{"id":"ca1baf10701070ca906e831d269b3ca28356a5ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 16:24:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf7f1c365bdc4264b0117a3362aae1cda6fb765e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 17:43:57.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a60325f17be4dbb9d3c9822e026306f : SUCCESS in 7m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/990b79cc419f4862a62d85400e299c63 : SUCCESS in 58m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dd1e1f6264854e6db28d9c9c4c9d310b : SUCCESS in 54m 18s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/edee575c61cc42389aa4ccea5cb9229d : SUCCESS in 10m 07s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/08a3177e765a4fd1be4094ca0d52b5ca : SUCCESS in 11m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a8cf120c86b84a0ba9d4cda337a61b34 : SUCCESS in 12m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f02c4e2844c48ffbddeed1ff363ea37 : SUCCESS in 11m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/43f1a9829e154b11ab17d2cb4a0da882 : SUCCESS in 11m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19027544338d45a7b46bb51b8623fadf : SUCCESS in 11m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bac69950e7414c139378609e1a024c53 : SUCCESS in 13m 28s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/612a40b7c3044673b3039d605cac91bc : SUCCESS in 10m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b544db5d7da54373a469b10d7db140b8 : SUCCESS in 10m 36s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/18d99e6d7fe6441385264f9555795c09 : SUCCESS in 12m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ff64228b9a024488add39561d7213fab : SUCCESS in 14m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d9bf80fbbeea4a57a0cb5f6fbdd42d72 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fd70931e739e4e47b87181f29453d5ee : SUCCESS in 11m 12s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/edea75e4b043461589bd9102ef715bb1 : SUCCESS in 14m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fec76ddea60a4ef1b62cc04726c177f0 : SUCCESS in 30m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed5d908d772f485ba38f258f8475b522 : SUCCESS in 37m 36s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ee953217aff642d9b2b1d690121f7955 : TIMED_OUT in 49m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/66821f640c4b46fda5a83cc3cd23f0f6 : SUCCESS in 29m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f2e8118d0954073bb4352b9de03b4d5 : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5377ff5dc9a408a980f0293f2af18e6 : SUCCESS in 5m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc840e0aa7254bf18e52a0122c31e492 : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":2},{"id":"72800f0c7f517af08993d2f4da13825f55e086c0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-18 18:17:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6745be2c45aa5c46df6c6f822aaf237d84fb5cf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 20:53:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f0d550dba5442ebb9ab6515112e7a8e : SUCCESS in 8m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b256a89d5f844d5aa64d6651e5306386 : SUCCESS in 1h 08m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a4971b2301646cc9df124b4b9fa5815 : SUCCESS in 1h 00m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a428233284a847ecaa50b5fcfb783e7e : SUCCESS in 8m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b1842b81dc5b4d0c9863ad1f90deebb2 : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9f576ed3acb643ac9c59495b54b94c74 : SUCCESS in 13m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a9ff1fbd92842d5804dfbdbbe786ea2 : SUCCESS in 12m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/da4a5343f8644a938bc2fa0f634e70d8 : SUCCESS in 13m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fce75c1b08f74b1dbc19da5c81b57230 : SUCCESS in 19m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69aec10a95254714b245c13a11759645 : SUCCESS in 10m 00s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4f721da9734246c18a112011a9757745 : SUCCESS in 9m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c5c8f7145b78416087c61af27623720d : SUCCESS in 11m 16s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/322b065199824dce928828f4420a255c : SUCCESS in 15m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/495bf79c67534e29847286cf5069cbbe : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4146e04258e142f0bf52e3749399fc35 : SUCCESS in 12m 48s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1dce14afa63f4629b65620c57f6e0778 : SUCCESS in 12m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/67dc8057f9064acd8046e9495ef7c6ca : SUCCESS in 12m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/cf4a593dd82443dcb6c93618221fea20 : SUCCESS in 15m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d3d664944cd4f17b8f88e9b7d972452 : SUCCESS in 17m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4be66ee23fd94091a5e64d406b265d4d : FAILURE in 34m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e57a07a01ed64e7d9a63c353b6154a31 : SUCCESS in 28m 36s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b4d2db13c4234087a6ee81202a027fc3 : SUCCESS in 31m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/afd1da8af7d249e5aeeb3ae43dbc8e22 : SUCCESS in 39m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a0dbe77c0764e30a993d03e4b586f05 : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a8a3059492e4a129f806c1f625d791c : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f0e936ff728a4df09321e7b234c45fab : SUCCESS in 19m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/110e658e58674492abb8a14ef34722e4 : SUCCESS in 1h 13m 50s","accounts_in_message":[],"_revision_number":2},{"id":"4389ea9af72d7c04f3cfd9082fde886999c91ab5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-18 20:55:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7df16c8c2b0d4460d51279795a55e06be3d4e122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 22:04:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b4105163c84740a4d988eaf9a5a3ce : SUCCESS in 8m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/209df8073db94fa089776fa93774271c : SUCCESS in 54m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/73540409f091459dac190d4a223b488b : SUCCESS in 53m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e119968e73794b70b7168d54eb262bb0 : SUCCESS in 8m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1fa439a6b17f47d98e1d33163827adb2 : SUCCESS in 10m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9b68d0f1bf1e407b9b05d0adc1249656 : SUCCESS in 14m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/160025b357254c7db515d7718fdf0c2b : SUCCESS in 13m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5962d749d4b742f886814d6709d9f4e8 : SUCCESS in 18m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/58d741fe2cfe4becb1a340ab858c7e7b : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ab057e60ebd4f4e886c63eac80ad4a0 : SUCCESS in 10m 05s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/940dbf846352475fb6d599c554684df8 : SUCCESS in 10m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b40b9db9f31446cfbb1fefd6b90876fd : SUCCESS in 10m 19s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6017a5b0c7d545178350e22ddee1f2d0 : SUCCESS in 12m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e713ed9b378d4bb3b0b4cc4c4e241c02 : SUCCESS in 9m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/415924432bfe4911b30b1a4ee006a728 : SUCCESS in 10m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f198e7435da641a7a9a85c177a1e877a : SUCCESS in 10m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2c6082af51ee46caae6e74eb01d78c59 : SUCCESS in 11m 16s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/342be398b8f244ee99992bde384e7eee : SUCCESS in 19m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/401c0351b8b04676a2399b0a17a4889c : SUCCESS in 15m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c67411cc644a4bb0b0472b28e64505e6 : SUCCESS in 30m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f0b1d3fb993445c4a9800b248fa1d702 : SUCCESS in 29m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/3031f123ed5642c3a6976b4adf3e9bb4 : SUCCESS in 28m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1cdc98abf8e846c7844c472dce52c39f : SUCCESS in 32m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85ee235440b742eb828326128f717b96 : SUCCESS in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4eb989957154325b682ed70b67cebaa : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dfac8b9deb024c5580af4a0a1c3d8991 : SUCCESS in 17m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9380d51a20a44772a5be9060c5eea06a : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":2},{"id":"31d769bcf0fa34f44161a9549bed0f0d4c307205","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 22:04:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0b802049f381da0a53661fc7d6ce60e2e79a7aec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 23:39:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3128e98769b4e2085a1c36b900b628d : SUCCESS in 8m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f9917ad79a8d4d46a12d3815fdd19c35 : SUCCESS in 59m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7d0d703583da44989868b5bfa1fe52c0 : SUCCESS in 1h 02m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0d6c22a79a6f43329873a9c0be9fd3b7 : SUCCESS in 9m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/bf1cc2a6cf0844fb9956a4702c6740fb : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f0f931dc3e4142c59886098d6f77eeec : SUCCESS in 12m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d338546d18144e9cb1c711a288442f97 : SUCCESS in 13m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9d0d3cec4b644293a163d24e0d912eb8 : SUCCESS in 12m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e2d510784ee741b982654b7f23723732 : SUCCESS in 9m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cadb77d3782d4468bb68efb03efec5ca : SUCCESS in 9m 05s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8029e74365a34df0a15f82496e9d58a4 : SUCCESS in 9m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/354f334a744d4314988db02c62ac8db6 : SUCCESS in 12m 10s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/912f988c30ec44de86b25eed0d07b71d : SUCCESS in 13m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e2938026ea4b4b80928792d8572fe8e3 : SUCCESS in 9m 27s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1de6f7f8757e42c2aff699ec30dbdab4 : SUCCESS in 10m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/71567abe7c1f4055909ad9cc88110d9a : SUCCESS in 11m 14s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4f1d60d4f515429fa7d5b92623901533 : SUCCESS in 13m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e102fc27d68f49b2803239b32e56ba53 : SUCCESS in 33m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e5176d7ca79143afa5511068834dbf9c : SUCCESS in 29m 32s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/bf8f4d9c65d349519a9091702796b4d1 : SUCCESS in 36m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0b3e40b40d4148718d0aaffa8d31d495 : SUCCESS in 33m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ba647caa39d46f4a40e8e8d17726106 : SUCCESS in 14m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa399c0a2c3e4e689b66de7d99fef771 : SUCCESS in 5m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55eeacb273164c8fa4553df9d86de4d3 : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":2},{"id":"dd728434ce3237856f6407a01b0f76503e966684","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 23:39:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"03ce58862fa2c84545af4c1ddcc1bb399020c741","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 23:40:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4515cba6584affb615c96f60e321b2 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/531a2f920a354f808f0a122fad8462a0 : SUCCESS in 49s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/335ba22d0fe340018a96f2c703230b48 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"196ef247857e9d3585094e441a952dd6b7a7ab77","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-04 19:20:18.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 4963aebd3b2d6958a864f84c7390e8ac44bc4273","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cfb3ae6019d9a4437675ff9d6161d67071fe528d","revisions":{"83c87e6c7032da2870b66a5e2cae5fd1dee407f8":{"kind":"REWORK","_number":1,"created":"2019-11-16 00:20:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/694650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/694650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/694650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/694650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/694650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/694650/1"}}},"commit":{"parents":[{"commit":"9fa0b211a9b48c6fa00654386a615094f750a151","subject":"Merge \"Seamlessly reload servers with SIGUSR1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fa0b211a9b48c6fa00654386a615094f750a151"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-16 00:09:21.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-16 00:10:02.000000000","tz":-480},"subject":"Update known-failures and config for up-rev\u0027ed ceph/s3tests","message":"Update known-failures and config for up-rev\u0027ed ceph/s3tests\n\nChange-Id: I3833843cd8d23c498a2afe6c68a3f0afe26343c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83c87e6c7032da2870b66a5e2cae5fd1dee407f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83c87e6c7032da2870b66a5e2cae5fd1dee407f8"}]},"branch":"refs/heads/master"},"cfb3ae6019d9a4437675ff9d6161d67071fe528d":{"kind":"REWORK","_number":2,"created":"2019-11-16 22:24:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/694650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/694650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/694650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/694650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/694650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/694650/2"}}},"commit":{"parents":[{"commit":"9fa0b211a9b48c6fa00654386a615094f750a151","subject":"Merge \"Seamlessly reload servers with SIGUSR1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fa0b211a9b48c6fa00654386a615094f750a151"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-16 00:09:21.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-16 22:24:08.000000000","tz":-480},"subject":"Update known-failures and config for up-rev\u0027ed ceph/s3tests","message":"Update known-failures and config for up-rev\u0027ed ceph/s3tests\n\nChange-Id: I3833843cd8d23c498a2afe6c68a3f0afe26343c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfb3ae6019d9a4437675ff9d6161d67071fe528d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfb3ae6019d9a4437675ff9d6161d67071fe528d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
