)]}'
{"id":"openstack%2Fswift~695053","triplet_id":"openstack%2Fswift~master~Ifb15ce50fc3bcfda9532a2c3dec542c272ea4933","project":"openstack/swift","branch":"master","topic":"695053","hashtags":[],"change_id":"Ifb15ce50fc3bcfda9532a2c3dec542c272ea4933","subject":"py3: Fix s3api header casing","status":"MERGED","created":"2019-11-19 17:57:56.000000000","updated":"2019-12-04 23:10:45.000000000","submitted":"2019-11-21 21:56:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695053-1574373385929-de65efff","meta_rev_id":"5b1ea89f0ae6ec8686e15d85f83f742f88c84107","_number":695053,"virtual_id_number":695053,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 21:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-20 23:57:22.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2019-11-21 20:42:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2019-11-21 20:42:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-21 20:42:31.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-11-21 21:56:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6163229329e939105a6b0ea10746d6c12dce6f9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-19 17:57:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae488cceed24be9e03d986cc50fbf2e401a6e8eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 19:56:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635f19751cd4458295a6ff54ef32c9cd : SUCCESS in 8m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/85bda1305fce4ff59dd7a46132ec1e08 : SUCCESS in 1h 02m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a4308c8ea774691a687e0a8807d6c22 : SUCCESS in 56m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/440a0901ec4f4e23acb6c0c05ecb43e4 : SUCCESS in 11m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1bdf468fb39e4405b446c30d4677e404 : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/05c3e07d61af46ee9646ba296f1200a5 : SUCCESS in 11m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c12c91350d4f4189b012e9016eb2c3a0 : SUCCESS in 15m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fdace7ac99d94360a4acd61ae7a7dcd0 : SUCCESS in 16m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4cd0fe870a174f65ac4e50bdfc80dd89 : SUCCESS in 9m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/596931c3e4c0496787c749c244fe113b : SUCCESS in 11m 58s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/953e2d8d29894d26925e212a44126428 : SUCCESS in 11m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/658e80fdb5f147f099df2f37cd75746a : FAILURE in 20m 40s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0acd27a0194347109e00e515857107dc : SUCCESS in 12m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e0755d3a7b5c4965a873c16c6579d75e : SUCCESS in 10m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/053a3e9ca4dd4a98b0ecab74261ffb7f : SUCCESS in 12m 15s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9f8c1c384fba47398a901218ed3006ac : SUCCESS in 14m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c5109dad5a6d49cdae734480b8ef8dfb : SUCCESS in 12m 01s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/dca09bbdcee842bebe95183cef8fc7d4 : SUCCESS in 14m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9385c2d9d8ad4dcea770251383153c14 : SUCCESS in 21m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7ea727fb45a148ef87daabaab771d6a9 : SUCCESS in 34m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/09838341bf064a8687be4fa28575464b : SUCCESS in 34m 37s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1beb6ebc324e45b8ac282f8e377965df : SUCCESS in 35m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7aba3ed8de644f8a37bc9ad9807f5bc : SUCCESS in 37m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d727cafe72c458da23f2096fefa43b2 : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be37af5a17c0440a985a21d9b6a8cd90 : SUCCESS in 8m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c682fa2baea49618d40e05fab2da668 : SUCCESS in 21m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29fdf59d5233480bb499a26602270904 : SUCCESS in 1h 05m 56s","accounts_in_message":[],"_revision_number":1},{"id":"04faa36f4c7cf355a28bb8d8d5270e920756d416","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-19 22:06:56.000000000","message":"Patch Set 1:\n\nrecheck\n\nThis is needed to get ceph\u0027s s3tests passing against py3 swift. (Or at least, passing as well as they do against py2 swift.)","accounts_in_message":[],"_revision_number":1},{"id":"863014a3f1b7650a1330d4b695733723ef6c9e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 23:19:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd66c34b2f249ab94fd0ad0837e7205 : SUCCESS in 8m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f393b3d53c8e42fb90e4323b6cec9356 : SUCCESS in 1h 04m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/80f990ccc28a4ad7881e98a345693101 : SUCCESS in 55m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/09ba6131db554252b31a48bf0f368d9e : SUCCESS in 10m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b118a3947ceb4294a8b386f184338b75 : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e201f3f00b241a6a112164315399a56 : SUCCESS in 11m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/40bca61c7c7e4a3b9ca8091bf338d649 : SUCCESS in 23m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/386f058e21e04e8c954c899e25e455a8 : SUCCESS in 11m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c432af48aa21452ea5e66b349941f83f : SUCCESS in 21m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e752cf07b98049789da61f31d64c2adb : SUCCESS in 9m 45s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/9c3f4f1b9f58439e82615ab6a99fb92d : FAILURE in 20m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5969586071a54bbdb016ddfb834aed9b : SUCCESS in 11m 02s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/2e09f13dfd2145419806c52ca0f263bf : SUCCESS in 25m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/23b433a7df23418d8e749ea02c79e673 : SUCCESS in 10m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/00cce3d191f944fc9ccfd1dd5ab48db2 : SUCCESS in 10m 11s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/aeebe6f91cbc4d5fa9a2bf3262347662 : SUCCESS in 10m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e2de545f12fb48ff8fce866837d4b9d8 : SUCCESS in 22m 06s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/56aef8bb6a5748ba93aab3476f054a1f : SUCCESS in 14m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/204df591c2b5446fab53ba12bd528046 : SUCCESS in 16m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/814499f85d1a49ddba2af591fddbe5c1 : SUCCESS in 34m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/524541a347db44e495f077c38a04db8c : SUCCESS in 33m 05s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c4eb6005193a49f4ba5b33ae1293eac8 : SUCCESS in 30m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c755df24f39046938dc2c5f5a8b06c21 : SUCCESS in 30m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65a7d3e8d1e248238a0d11ab8097de23 : SUCCESS in 23m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21f066f6c02d4d61a42c5eb7342b080a : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/99e96a95a6ee4846af8de410de2b42c1 : SUCCESS in 19m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2d01dc771e144a1a17256f1a1fa1d39 : SUCCESS in 1h 06m 33s","accounts_in_message":[],"_revision_number":1},{"id":"63713de5ad2b5b6e08ac1459fbdc80f19c361db9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-20 19:14:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aa08a67331085660c905b6f3f507da9421017f3b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-20 20:15:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ead46897dcd4cb7bdac9fe6bb3f064cd72fa07c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 22:04:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc3530c988b54255a7992d42527c1efc : SUCCESS in 10m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cacfef52c4814fb195affc69f3e0d1d7 : SUCCESS in 1h 06m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1caa2d3d21164ea29526e9aa9c8333a6 : SUCCESS in 55m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/48fb2324c2ff4f9f833dac803d40b02e : SUCCESS in 11m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2e5db33370364deca4d7e3801d0b80bb : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5237646e43ec4c5f9549c6ad5be130ea : SUCCESS in 11m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4372bc7ef1df4c0ab07e3389d621ef1c : SUCCESS in 15m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/56595ea84dd4402cafe9a3021c9d7410 : SUCCESS in 15m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/707a89c5f3034dd5a949454ceb2e63a7 : FAILURE in 15m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b95096db492f4ff3aa41e1dfacb8c375 : SUCCESS in 12m 54s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/2341a78a128d4a459b5e88364adeda8c : SUCCESS in 13m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/24255b7a12104d0c89e43e58aede2666 : SUCCESS in 16m 35s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ab88aceb262f40cb82b532d885ed2c80 : SUCCESS in 24m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/21e8b35b018c4ac59207ba64454245ee : SUCCESS in 11m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a839ad7b4d914ab88f422afcdeac1e1c : SUCCESS in 13m 35s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3c07523553ab480a90f18fdbfaac520f : SUCCESS in 9m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b8fc6fddc2274202be43e46eaf9f5da7 : SUCCESS in 12m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/63b6f5a2c930406bbe398ee4eee80dde : SUCCESS in 13m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6eb8ce3e220b44d9ac4da26030da324d : SUCCESS in 18m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e35f69033e984c4c917cdceb52ec59f4 : SUCCESS in 34m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55937f6bf00c4c6b8a0e47795e6befe6 : SUCCESS in 28m 42s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/149327b0b1e1405b99a04e7428aaa7b2 : SUCCESS in 38m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f5307f04820445869ea239da5e87d0af : SUCCESS in 29m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2981ad0964443fda6e94ca9c5d9b8ce : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe14f29786d74a0d92afb0dfa90fad99 : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9d3e30f55b5445399ad9a4849ae39ee0 : SUCCESS in 20m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98284e3454b74c03b4a008784cc46223 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":2},{"id":"9fa97a456c7e81313bd95324eae57213924390f4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-20 22:23:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c929d73707be01f20117fa09334d4a2988a1fa14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 23:45:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/832defcf349c4bb38ec0cff5542f2960 : SUCCESS in 9m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4fec86f14eea4bcfaed9348b52e06278 : SUCCESS in 52m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd12fb193e1041a2bbaba6eeab034493 : SUCCESS in 52m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/798343cb5b50405998f0c4039e4f87aa : SUCCESS in 8m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/45e8c19823974c74bf6fa20888d3cca6 : SUCCESS in 9m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9a6de98706064e64bd3c8d4f08782d4a : SUCCESS in 11m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac9afc1debd04a1a88be4b1ca3647c59 : SUCCESS in 14m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/746473b50f4a4ffe88e33c29e7faa235 : TIMED_OUT in 40m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9b2ec3ae4dab42299b567371c6aa0556 : SUCCESS in 10m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/96947f5c5c764c24a8fd9e2b1d13b8c9 : SUCCESS in 9m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4fd91f6aaa774f5b89b4e4c8a6b04919 : SUCCESS in 10m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ac13a0977d1840c3829c02db2410ffd0 : SUCCESS in 11m 04s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/9dbf06cb766c47a6bd84db7b5611e2ab : SUCCESS in 12m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2be9802250b5474f87dfd497847e7edf : SUCCESS in 9m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c70bfad063c6450f9d9f5fbb0622512e : SUCCESS in 11m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/db3d24c2c77f47788b7c7142ee439dc8 : SUCCESS in 10m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0386f92f72c14a5791ac6780012dc566 : SUCCESS in 10m 19s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/bd2f42f2283346bd9090a342ad0fbd22 : SUCCESS in 14m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e22871b6c5884431acb604d658bd5361 : SUCCESS in 23m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a2b60967034641fe8f6d0433a1fb9e48 : SUCCESS in 34m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05b0f815e0b74878a16cbf00bfbaeb28 : SUCCESS in 37m 13s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/6094cec995904ba59fa7446c02874b30 : SUCCESS in 30m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c4e0a6d86a2f448cbc916f4d403d3f83 : SUCCESS in 28m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/168c32d67e464159ad7e6de0c8b57b02 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4999122cbbf42d5afb86e5228813d0d : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5641c1091b6742348ac32e7a92a1fb48 : SUCCESS in 19m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14523f31025c4fa2ba4c447fcbdd4a14 : SUCCESS in 1h 16m 47s","accounts_in_message":[],"_revision_number":2},{"id":"363820f3cebd04078cd305ad28def675262009eb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-20 23:57:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0d542ce598feee5d216e3fd0da8bca7eab85350f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 01:03:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eff5a3bd45254849ae9b2e802df8aec9 : SUCCESS in 7m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/29ebdeac861f495e85d256967c344db2 : SUCCESS in 1h 04m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dfccab62cb4f4b42bf608302fcd9519f : SUCCESS in 56m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/da7425f935ea46bc917ce535336f390d : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/738250aff90040039fbc3267df822113 : SUCCESS in 7m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c09fae990654c828be1a76edbd62ab6 : SUCCESS in 15m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/beefd24ff66b48a4b26371809714e437 : SUCCESS in 12m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9f8dda2a41194456bb14dbd7f5c1bf26 : SUCCESS in 16m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d91bb7257a9b4ef99add93f4f1088a8c : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/da92f1bc44e04553b90ab5ad2cf7bb56 : SUCCESS in 13m 43s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/bc3deca9cfe043179b32d78a776e8578 : SUCCESS in 11m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c0725086422a4fadb57751ef42fe9205 : SUCCESS in 19m 09s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/8121f093364d459bbce513362d10450e : SUCCESS in 12m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0d490a815c67408494a5823332268c97 : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8e476dc6de6e491ca99b65ec4f36de48 : SUCCESS in 9m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0cc9b57e05e8427ba510a548e6fc0716 : SUCCESS in 9m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ff00041328dd40ffb7d55bf1266396e2 : SUCCESS in 10m 13s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/9eca6042da824261823005252face315 : SUCCESS in 12m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/75afd43dc82b4e4391e4edbb11e3a34a : SUCCESS in 15m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cc49c85c9d86416db521bcde0772ba68 : SUCCESS in 31m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/28a538abea7f4b28819c80708f27a89a : SUCCESS in 30m 13s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/cff5b1904c6a4fc0a5c3885a859678bc : SUCCESS in 27m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e480ee7f0f754cd3931941ff6c597bf3 : SUCCESS in 28m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d52b852a67b8469c976b02ff4cb18503 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3ec76236f04eadb3ecfc0a14dd8ae8 : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/78e5ad7a1f934b8cb4b9dc11aa9a6b8b : SUCCESS in 17m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7f4286c960e4c898c6311ec63d9c15a : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":2},{"id":"feda25a0f2726701ca5abbb0cd277a2ded723e7c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-11-21 20:42:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI think this looks cleaner all around!  bonus that\u0027s it\u0027s also more correct ;)","accounts_in_message":[],"_revision_number":2},{"id":"97290e22db15d6a8bde4f32638b466eaeb2b1df7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 20:42:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"161a4978b6dca8f29e75321cbab7517267106e4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 21:56:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecf0f0952e7747c697f90916a28dd588 : SUCCESS in 10m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/892aec2312bb4d8f9be4cf7b89daa93d : SUCCESS in 56m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1ea5ea885d34f7ab625b81380b4f5f5 : SUCCESS in 53m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6d4ad277abb744dbb27f5161f970af08 : SUCCESS in 9m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/831dce1ac6544879ba033c29ca25be93 : SUCCESS in 9m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/41e547a6db534b09902b06d11ba3c9ad : SUCCESS in 11m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/043a565fa2d1428d8601cee63e404a74 : SUCCESS in 13m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1f686196d1ee46aea4e4c15f753ca6ab : SUCCESS in 11m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a2d7adf1da884ed199647db985ba3fa7 : SUCCESS in 11m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb3d0621c3914a2bb8de69d98ab49c9f : SUCCESS in 9m 31s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/7747760a113c452bad85a44a14349588 : SUCCESS in 11m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8af6f3fd68d04b0b991696529902fded : SUCCESS in 11m 56s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/44d24b1426484c5e934dcc09038644ca : SUCCESS in 12m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9cb882efd61c4d5db6a93c0cfab411de : SUCCESS in 8m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b06afd212550447fb21abb048a04cc6c : SUCCESS in 12m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/84727fa2ead645029b6da023418d4c17 : SUCCESS in 14m 58s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/023e00e1f6ee47839b833a63c855333f : SUCCESS in 13m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b9994195cee34d95a21980835addc284 : SUCCESS in 33m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2fe97533f3454b038054020df56ba280 : SUCCESS in 29m 41s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/57dbc66819c345aaa84486b232027686 : SUCCESS in 30m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de819430cc7e4f2f9dd5f4539301b136 : SUCCESS in 29m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e57ece50ab264be08732795a57693cbb : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a463aa90c1445d28bb70cb203595252 : SUCCESS in 5m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b11971c5112e4385a0d0b7ae2e4468b1 : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":2},{"id":"4c7de06e678484bacdb6b7830e30be3e5e25937b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 21:56:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2b6e7095336ddff395456ff611d59ccf9a2a1e4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 21:57:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa9a425d1df45a292463e342566b805 : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c26e539051f64cbca4e5acb549a7aec6 : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/622adcefb3f1473b9ba32cb64fb45a22 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4d923c9c50e83eb30b83b59b74ea303e6ffdf7b6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-04 23:10:45.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit f98396e6a2e3f87983fa4ba49bfbd19bd4e9607f","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b5c9dc1c9f68afe8083b74eaf8080e17a9ef8f8b","revisions":{"fd5300d3fa978ea9b00667b98b5c8dac0d5b1313":{"kind":"REWORK","_number":1,"created":"2019-11-19 17:57:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/695053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/695053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/695053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/695053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/695053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/695053/1"}}},"commit":{"parents":[{"commit":"cfb3ae6019d9a4437675ff9d6161d67071fe528d","subject":"Update known-failures and config for up-rev\u0027ed ceph/s3tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfb3ae6019d9a4437675ff9d6161d67071fe528d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-19 17:02:27.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-19 17:02:27.000000000","tz":-480},"subject":"py3: Fix s3api header casing","message":"py3: Fix s3api header casing\n\nChange-Id: Ifb15ce50fc3bcfda9532a2c3dec542c272ea4933\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd5300d3fa978ea9b00667b98b5c8dac0d5b1313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd5300d3fa978ea9b00667b98b5c8dac0d5b1313"}]},"branch":"refs/heads/master"},"b5c9dc1c9f68afe8083b74eaf8080e17a9ef8f8b":{"kind":"REWORK","_number":2,"created":"2019-11-20 20:15:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/695053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/695053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/695053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/695053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/695053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/695053/2"}}},"commit":{"parents":[{"commit":"cfb3ae6019d9a4437675ff9d6161d67071fe528d","subject":"Update known-failures and config for up-rev\u0027ed ceph/s3tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfb3ae6019d9a4437675ff9d6161d67071fe528d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-19 17:02:27.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-20 20:14:35.000000000","tz":-480},"subject":"py3: Fix s3api header casing","message":"py3: Fix s3api header casing\n\nCloses-Bug: #1853367\n\nChange-Id: Ifb15ce50fc3bcfda9532a2c3dec542c272ea4933\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5c9dc1c9f68afe8083b74eaf8080e17a9ef8f8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5c9dc1c9f68afe8083b74eaf8080e17a9ef8f8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
