)]}'
{"id":"openstack%2Fswift~695057","triplet_id":"openstack%2Fswift~master~I488f88a54948663c682b35dc976b62db6e72eef7","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I488f88a54948663c682b35dc976b62db6e72eef7","subject":"Switch py2 DSVM jobs to only run swift under py2","status":"MERGED","created":"2019-11-19 18:17:25.000000000","updated":"2019-11-22 00:23:08.000000000","submitted":"2019-11-22 00:21:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695057-1574382070528-9f2e12eb","meta_rev_id":"c17c63647927c963f0b159c51b25cc83c13b1ece","_number":695057,"virtual_id_number":695057,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-22 00:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-21 22:20:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-21 22:20:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 00:21:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21195cd3ee50e2d05946a21f99f2ac8a6ab03a72","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-19 18:17:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8c9ccd701439168edfe7c048910b0915c873a3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 20:09:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52932b12afc04e999e72ef031701ac9b : SUCCESS in 7m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0a068f8067c14c33b3464a961ab4f6e6 : SUCCESS in 1h 14m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7636fcd4888243689b8bc4cb3c278ed8 : SUCCESS in 1h 19m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a679560f9c394b38acb97a0ea02fe7ae : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d11001f2bb80465394716da305c97f70 : SUCCESS in 12m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/89be580c402b47f2be162e43f9052d6a : SUCCESS in 16m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8fdc0fc202ad4e898a05778c9f3e59ed : SUCCESS in 12m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/25bf1f511c47405aa6dcf32dccf20041 : SUCCESS in 13m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/21e24ab7cb064861bda4bea6547580e0 : SUCCESS in 11m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/77b933a6c0d2410c93bf0ebae2f38b1b : SUCCESS in 10m 54s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/3c2591f249f5453e877b0de898380fb4 : SUCCESS in 10m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/81143e31d80a4dcf87e104f77554fac0 : SUCCESS in 11m 17s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/b1677ac412d144afbe9df773b466702b : SUCCESS in 13m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d5a729d20f6c4defa7e6e629ee797c0c : SUCCESS in 11m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/185b092d6a1945b5b21d51ec1569d5ac : SUCCESS in 9m 50s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/cf97da680c184264b8dd88a0386b8ddd : SUCCESS in 10m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8f6ebe20a71b4c80b23fab680c1b9828 : SUCCESS in 11m 04s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/16583c4f931a4d54b5b18a5334248327 : SUCCESS in 12m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/841953486e6f406584d120f5b447d604 : SUCCESS in 17m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/53dab6d18a904c12a9abec7f76dccb22 : SUCCESS in 34m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/786d6afcda5841bfb34f857286fff741 : SUCCESS in 31m 36s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5bf1a1e87de74a5ea575bee765cdb80d : SUCCESS in 30m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/72f628cc1a084450a63b8be1e2b5b831 : SUCCESS in 34m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66164bd8a98c4296a6be0b4df5f29556 : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9e2a863c0574e2fb377458f5f39ebf3 : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/207ad24e261040eaaa2b6b2a9137ae6a : SUCCESS in 20m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d991510aaea483bab865f9988920259 : FAILURE in 1h 03m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b738d6046220a3ed856caf81ee7526c0919a66b6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-19 22:08:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a7f2498248a6cc22fee24ebe8f74e71399933075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 23:12:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/797817fb401445309107817528efd223 : SUCCESS in 7m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4cb4c2afeaea4232b4036e0e56809250 : SUCCESS in 50m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f33b8001357b45fbafe38f1981501311 : SUCCESS in 57m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4da0008f7af4a71b901e90e155aef79 : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9e9744ebb8db44d78e2705a3d547fc12 : SUCCESS in 10m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/64bc66d919314616b134c3c56fc9b780 : SUCCESS in 14m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/090151f9ca6a4b50bf9ef29bbdb6f4c2 : SUCCESS in 13m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b1d53b3f04274c0db328602327221d1b : SUCCESS in 15m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d7da2159afc44474904b840753820f35 : SUCCESS in 12m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8daf2410c7324c47aaffa05a313448ba : SUCCESS in 11m 39s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/0f0fa5d3a2d84cd2972d91e263591fb3 : SUCCESS in 11m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd027fc8f4a74e56945bae3e9d1950d3 : SUCCESS in 12m 11s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e8d2f01d963445f0beb35fd10c9c9eab : SUCCESS in 14m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1225696b79bf433bb728beaee598a77d : SUCCESS in 11m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4590cd4215444dd9844e101bbbd9f6c9 : SUCCESS in 12m 49s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a9e4268640fa44609bad12436fbf6a1e : SUCCESS in 11m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fc9de3b59c864a78a9be4229faeee3f9 : SUCCESS in 12m 00s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/edfa5c1e1b9a447f93e97c7ba76ed2e0 : SUCCESS in 15m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/25cf6367998d4bed807abb873654cf9c : SUCCESS in 26m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/241fe58df30b4e1aa9c43e7365653129 : FAILURE in 32m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e2e52fced3b414e832046f807905412 : SUCCESS in 31m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/034cde7208084e0d95a15b6150325fda : SUCCESS in 27m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a3575898e6d84cfab8fd1b4ab2811f0c : SUCCESS in 30m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09acd06b0892463d928f8c2030c52019 : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddc961c93f974c928962fcaffc189cea : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a3520b0dd054510a16661afce86cbbd : SUCCESS in 16m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cff6bc8c926d4b9c86aa1a0318ee5311 : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":1},{"id":"aef151d380006fe51bbd97fe535fb064aa752e1c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-19 23:53:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ec3ae13ab23264634e71cc65973b4bf35788c9b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 01:02:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b28ef336f994383874ba47bd5aa342d : SUCCESS in 9m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3bcfab890030407984cd052bc073198b : SUCCESS in 1h 06m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fcf711fd4de94f3f9cff4a703ca7a9a5 : SUCCESS in 59m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e5e6571ccd5245938f66480e039ab837 : SUCCESS in 11m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/951b72d14d9a4c1cb512acd810b6e51b : SUCCESS in 10m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94ddb529122442aa81b1964c22699c9b : SUCCESS in 11m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3dbcb227f60d4db3831f92cd764612fd : SUCCESS in 12m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ff0a1858cffd48e28859c90d6d91e459 : SUCCESS in 18m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ec06546952af43f587431ede35435af6 : SUCCESS in 22m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/66923e71346248689614076d73f5e63f : SUCCESS in 9m 00s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ae78b4426d70417aae08bbff5402199a : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/023aa85ab111404eb3392da366bc9258 : SUCCESS in 19m 01s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/497995cf523e4eb9aab9139381d00a84 : SUCCESS in 12m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ed4518e92d5e415f8ecdaa466404c52b : FAILURE in 17m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d106333596214540a613a4308d98c0c9 : SUCCESS in 9m 44s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/926c3b86a9d04cb49e2bfa54fc2b5dd6 : FAILURE in 22m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/54b2abe2868c48a4a8791280b41f7e25 : SUCCESS in 12m 43s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/60413ff3e45142c0a81c130e5907a5f8 : SUCCESS in 15m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eb93819ca5b54035bfe99f137237e967 : SUCCESS in 19m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0b7fc0aa8b2045ed85a635dc2257d6f8 : SUCCESS in 37m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7bc56caa27ca4dbfaa71ff62f075ff7b : SUCCESS in 34m 20s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ddee3bd9c69c4189b9e93081fcaf8286 : SUCCESS in 34m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea199ad337a84d72b549248956b01229 : SUCCESS in 30m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d38b57fb45f046a5ae3b57fcc9d8b23c : FAILURE in 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69772e689fc4bd585e942a6344c58f3 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c9c6f658a55048d787de65143ac6e495 : SUCCESS in 17m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/712215844c644abcb0bbbbfebc643162 : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8b6ff3958d00507fb1c87d4752fe14a29447dbc2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-20 19:37:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"99f74c9789451a49930d08c8b656d85887fb2ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 21:41:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c5d9b5a51d471394c7c0e8596a88f9 : SUCCESS in 7m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5781b5c53ae647239fadc25100be054e : SUCCESS in 1h 07m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/530cdd95d5144b3981bd5fe56c1248ac : SUCCESS in 53m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6cd6cfb4270e43ecb8beb49c3cd7d0a4 : SUCCESS in 9m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/14e87ec7745b4e198fdee3f186469753 : SUCCESS in 10m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ebdf93b911ef409bb8452d8bbae415c8 : SUCCESS in 12m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/23539397b78f43f282a434e1ee0a0694 : SUCCESS in 18m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dfaf7a8c777646aa8e13b859a78ed0aa : SUCCESS in 12m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9c24bb59249e42d991e60854f3b72a57 : SUCCESS in 10m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9107af53eef4490c9f5190782d0b71d9 : SUCCESS in 14m 50s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/780aaa2a90014ed5840dccb7a1253231 : SUCCESS in 11m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b68afdcfd707480ab04ddb85509f5fb1 : SUCCESS in 11m 51s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/42bc48fe34bf433db9966defbfd43324 : SUCCESS in 13m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e8f976d06c464a42b83e15be1b687684 : SUCCESS in 11m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e6b7b6a31df64c63adac3f231d8fee35 : SUCCESS in 13m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4eae0cec6a8e4ab09cbed6abaab28865 : SUCCESS in 11m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fe1ed4162d2b4166afd4b8d70857006f : SUCCESS in 15m 23s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4bf3b7a5c4f2429c8b18b75e15ff0287 : SUCCESS in 12m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e931722bb118444eb5d9e18d78f6847d : SUCCESS in 18m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/455772ee94a04a2fa3606033da9910ee : SUCCESS in 39m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8f3a3e668c24d588c719d15b6ed6ff7 : SUCCESS in 29m 44s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/850507b0cab1412c9ec6f51d11fc5ded : SUCCESS in 35m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d846e2d1c59466283b4a543471056bf : SUCCESS in 43m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bbe47fa0bf94e0cb3ff3ef0b4ad7904 : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d73ac2ba863f4e85a9a97adfd65d16ed : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f775698e9534413b592dc3ac8016fa2 : SUCCESS in 29m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/268fc7be13704fef86df00f9138456d2 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3c3cccc1c78a6ca24c6bb5b897db7e653ba76bfd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-21 22:20:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks like devstack\u0027s default has switched; see https://storage.gra1.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_6de/682382/44/check/swift-dsvm-functional/6deecc3/controller/logs/pip3-freeze.txt.gz for example (where swift is installed under py3 when the job was supposed to be py2). Merging so we don\u0027t lose coverage.","accounts_in_message":[],"_revision_number":1},{"id":"9f6668e281c3cedf2a1e7e486fa6508050628ee4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 22:21:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d2a190acbd3d5e2e41f9f0a74b340c4a2ee8d8a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-21 22:21:57.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 6e4fe1d5fa54ff8853b6bfe3924d5504fdb168e2","accounts_in_message":[],"_revision_number":1},{"id":"9fa086509a730dfd65603ec70c184dd8c96cdaae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 00:21:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ccb4a10da541bdb0bfcb1978fa2cf5 : SUCCESS in 12m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3ed57e6516d644a5a92da9f53cf43fdc : SUCCESS in 55m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e8db68c1e5cc466a860eb8dd608ff93a : SUCCESS in 53m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4138edbb4d5b4db3b74f3e845cec4929 : SUCCESS in 14m 36s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1cb14bed950a4f2cbafed30ce15d2921 : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28fce5b5a1c04739a4007c6e003ad03d : SUCCESS in 12m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/24ff0c8336bd43ddbdce89a55e71f57e : SUCCESS in 12m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cb4b8536669f44e58bcc63828128ac82 : SUCCESS in 15m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/447e94f76c944b99bb7da1c8a03e19ae : SUCCESS in 20m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a87470685c3f4b618c5afb98860dcb65 : SUCCESS in 11m 42s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ecdb6800977c4db19610ce14cd2a02c9 : SUCCESS in 20m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf8a7dcf6ccc4daba79bad95eedda4fc : SUCCESS in 14m 24s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6b4a132bf68a49fd96611dd1c3cec722 : SUCCESS in 25m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c468136ed36c47fdb750122905ee760d : SUCCESS in 14m 30s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/078a331a07ec44919fe48e745c5fb72e : SUCCESS in 19m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7ba692c94e82408ca570ef5c50a990e3 : SUCCESS in 21m 23s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a6f7625dc4164c6da0941446f833d2ba : SUCCESS in 14m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/13eb9fd145db4a3abbf185cbef0a9450 : SUCCESS in 36m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/13361604b956477081d54fa8d45037b0 : SUCCESS in 29m 53s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/64ea88a303914dc8b48413c490690d0f : SUCCESS in 41m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bfdfec435af342219698abb326e1a6e1 : SUCCESS in 22m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80aea4c8cb284d8c8e4fbde825b3ab74 : SUCCESS in 15m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a27010ff1e034b4bbe36b72042050d07 : SUCCESS in 6m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60219edddc134ed6a2e190fd79960264 : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":1},{"id":"011f89039e7f0e2de24b4ae47c4b5d46839096a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 00:21:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c17c63647927c963f0b159c51b25cc83c13b1ece","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 00:23:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090d20cba3d5422587eca323f3464113 : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/066758db007a40f88ae3f4a11d2e7dfc : SUCCESS in 37s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/16b3f07483fb42f7b77a2e303e3dfe99 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5dad445b258b6fd823bddb7b1e3ca10064ba3e8","revisions":{"e5dad445b258b6fd823bddb7b1e3ca10064ba3e8":{"kind":"REWORK","_number":1,"created":"2019-11-19 18:17:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/695057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/695057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/695057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/695057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/695057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/695057/1"}}},"commit":{"parents":[{"commit":"089b1a8b9a819ed9163e76e3db91b2a0fd32b61c","subject":"Merge \"Update known-failures and config for up-rev\u0027ed ceph/s3tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/089b1a8b9a819ed9163e76e3db91b2a0fd32b61c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-19 18:16:28.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-19 18:16:37.000000000","tz":-480},"subject":"Switch py2 DSVM jobs to only run swift under py2","message":"Switch py2 DSVM jobs to only run swift under py2\n\nOther services are fine to move on to py3. We still want to test py2,\nthough.\n\nChange-Id: I488f88a54948663c682b35dc976b62db6e72eef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5dad445b258b6fd823bddb7b1e3ca10064ba3e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5dad445b258b6fd823bddb7b1e3ca10064ba3e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
