)]}'
{"id":"openstack%2Fswift~695109","triplet_id":"openstack%2Fswift~master~I775a03e0ba1e10982eb6f7ef52be773c8831b1ec","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I775a03e0ba1e10982eb6f7ef52be773c8831b1ec","subject":"Remove a bunch of known-failures that moved from boto to boto3","status":"MERGED","created":"2019-11-19 22:57:30.000000000","updated":"2019-11-20 00:39:26.000000000","submitted":"2019-11-20 00:37:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695109-1574210275728-8d7acffc","meta_rev_id":"3f866dfa8e7e6bb0373ba8fa2a437aecdf4775ea","_number":695109,"virtual_id_number":695109,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-20 00:37:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-20 00:12:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-20 00:12:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-20 00:37:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd75422e764075ad7d48b6c6c3cc4109ba55f5dc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-19 22:57:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d201c28668c6a8b389c546b5e4442efd43b26f91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 23:22:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf3607286544f1f92302fe42cbc364f : SUCCESS in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b97aa40fb4774dd693dafd0baac3da48 : SUCCESS in 24m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98e6ceb96609f118d88bf5d8151728b853deb951","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-20 00:12:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nNo unexpected failures! Merge it!","accounts_in_message":[],"_revision_number":1},{"id":"c25f1cf71f2a4961baab294bf52bcdf4dc1c3e22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 00:12:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca336e9cd4d6216f35916293d84b0e7b8ba855a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 00:37:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34f7c668df948fb8654fb5998dfd16b : SUCCESS in 7m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6fca98c9485b4cbaab00c4707b1b8caf : SUCCESS in 7m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4854906381c74f99b70e26d5f7812505 : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cda33a96cf32b6285a27d247c1fcabadd4869cf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 00:37:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3f866dfa8e7e6bb0373ba8fa2a437aecdf4775ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 00:39:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e71f767f003644718134e15da848a0be : SUCCESS in 1m 11s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/895bc62e61ba4d57b163d4b9c0d9a774 : SUCCESS in 37s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1ba61757a5814d6eafb7e303156a48cc : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"249a302d0c43bf7d847b3691b795a7ba3c5b6c10","revisions":{"249a302d0c43bf7d847b3691b795a7ba3c5b6c10":{"kind":"REWORK","_number":1,"created":"2019-11-19 22:57:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/09/695109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/695109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/695109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/695109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/695109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/695109/1"}}},"commit":{"parents":[{"commit":"29019985ed63790851df7a2a0ed75593b11906b4","subject":"Start README.rst with a better title","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29019985ed63790851df7a2a0ed75593b11906b4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-19 22:57:02.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-19 22:57:02.000000000","tz":-480},"subject":"Remove a bunch of known-failures that moved from boto to boto3","message":"Remove a bunch of known-failures that moved from boto to boto3\n\nChange-Id: I775a03e0ba1e10982eb6f7ef52be773c8831b1ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/249a302d0c43bf7d847b3691b795a7ba3c5b6c10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/249a302d0c43bf7d847b3691b795a7ba3c5b6c10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
