)]}'
{"id":"openstack%2Fswift~695596","triplet_id":"openstack%2Fswift~stable%2Ftrain~I488f88a54948663c682b35dc976b62db6e72eef7","project":"openstack/swift","branch":"stable/train","hashtags":[],"change_id":"I488f88a54948663c682b35dc976b62db6e72eef7","subject":"Switch py2 DSVM jobs to only run swift under py2","status":"MERGED","created":"2019-11-21 22:21:57.000000000","updated":"2019-12-04 21:09:15.000000000","submitted":"2019-12-04 21:04:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695596-1575493447785-612455a9","meta_rev_id":"c58cb95f9a5f3d82de0c1f5f983240958ec38f41","_number":695596,"virtual_id_number":695596,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-04 21:04:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-04 19:17:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-04 19:17:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-04 21:04:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"145c49c4a1cbbf26dfedd1ba449bc6faa3c42762","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-21 22:21:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d4044c909588b0da76259aaf38306298871dc4bf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-11-21 22:22:09.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 19ed6048a0f517332f63cf58102cd2ce33fec4c0","accounts_in_message":[],"_revision_number":1},{"id":"d12a6abbd9d3d295baae22bae5148aa634479c8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 23:22:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c4b16f9a4b2471eb7c40ef2477def46 : SUCCESS in 4m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/85062ffc25a04be8bbc7889baa1c5d58 : SUCCESS in 54m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dc5fe50d4f7d4597a1b28d463382b6dd : SUCCESS in 49m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1c90d9e8a4748d4b385dd06226b608c : SUCCESS in 7m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b98001c1d2034929983f6186314febaa : SUCCESS in 13m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9e26c41fb3a14e5698cf6dcd38115217 : SUCCESS in 15m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9e75a07859b8486a8d06a14228fe02d9 : SUCCESS in 14m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2fcceff61e1a4f83a6e9a3e8b8f64f5d : SUCCESS in 11m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/31972002c2f34723ab1e58ae6a01fd16 : SUCCESS in 11m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/bafd92ae7e5a4d04a1a747f57879b644 : SUCCESS in 11m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/efa3c4b4b80d4b0ea578537c91c3dac4 : SUCCESS in 12m 29s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/b90d50d926a54ef89795fbe69bde9e8d : SUCCESS in 14m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f9d31832e9524cf6a78116b7fae29f43 : SUCCESS in 10m 50s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/16d3738c0b13462fb1c3d15337552b6b : SUCCESS in 11m 29s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/cf4ef5280aeb47e696a31ae626395748 : SUCCESS in 12m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d1af8a2143024afaae9602f78877cd1c : SUCCESS in 12m 45s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1c85cdbae09f435180fc9bd34426c44b : SUCCESS in 15m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/115eac416ea041a8bb3e7543d26e2a39 : FAILURE in 8m 27s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1e075266bd274d7d9932dbfaa88061d5 : SUCCESS in 32m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/deb946cd7c2348fdbedab647a3d70d76 : SUCCESS in 27m 50s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c52c751eaf394db98d5e6446724522f2 : SUCCESS in 27m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e72ff9b3c1443d38ac56d10eed4aab7 : SUCCESS in 27m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e97a1b8851234582b84dc9bad977019f : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c19c2bc194f41e9b32dfcc382eac1a2 : SUCCESS in 5m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/924c118ef85646ceb9b5a6ca6553e468 : SUCCESS in 20m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/238cf3cf9eda43acb890a109db3ac26d : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8282325ae5b8db9c10029cd1e4130c1e0f980128","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-04 19:17:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2ac4935393d0379f7c365ec3dd325dfd2113c69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 19:18:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4535673d21c0fdcc91bd7bf37d5df412c251638","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 21:04:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c73b9018bfece70257299d6c12fa7c06b9b90914","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 21:04:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5866b9ddb98e466a82b11957e0daa62b : SUCCESS in 6m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a4ac3b6f42674dc39a4ac38358d95348 : SUCCESS in 59m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/219f592d272d4e99ab6f92eccf32b7cd : SUCCESS in 53m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/01042e052fba4c20bfc7ede1e6153b84 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/71cc5bca31c24cf1a06490b7405937ce : SUCCESS in 12m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/48beb97908e7444d94e7e239a71a6aa6 : SUCCESS in 12m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b0aa173426434ecba87d06fca4a57b3b : SUCCESS in 15m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/64c6c04599f04d158f63077e9414ac0e : SUCCESS in 11m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8ae6275ea33f4766a9e99020dd8cd5e0 : SUCCESS in 9m 53s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4ea3b033cc6c4d58824f399cfea73bc3 : SUCCESS in 10m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6423af9eb5304ea9b11e22c5f581c6eb : SUCCESS in 11m 35s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0f79ec97cda240d799f030051010d675 : SUCCESS in 15m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/596e93fc4fc244a89e8c58ad658c6a5e : SUCCESS in 10m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fa7b1c1d10c94f9990023991d9311d66 : SUCCESS in 10m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6fa1b9bc2c224f12bc577795ced3fe0f : SUCCESS in 11m 36s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f1428ffe77f241469f9279f6e50e9d0a : SUCCESS in 13m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/68dfbc37e1ed43f8b203301eafc98299 : SUCCESS in 34m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d03e4a1f1414990ac2ee01b1fd5ea7c : SUCCESS in 35m 27s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4b3243ba108b4581b0b1cd5730b3400f : SUCCESS in 29m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c2ddae6a637f4e40b0050160be8d1ea8 : SUCCESS in 33m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29ca1c5cf0ce449ea092b241b12d4dda : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e35451cce0cc4accbd7169572c6bc020 : SUCCESS in 6m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2e167e7a6d14d4eaaf523ca4f65a8a4 : SUCCESS in 1h 05m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c58cb95f9a5f3d82de0c1f5f983240958ec38f41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 21:09:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97e02b9c1b46499588662421aa07ad01 : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3407f65d45a24913bae71d312d37b072 : SUCCESS in 52s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e4fe1d5fa54ff8853b6bfe3924d5504fdb168e2","revisions":{"6e4fe1d5fa54ff8853b6bfe3924d5504fdb168e2":{"kind":"REWORK","_number":1,"created":"2019-11-21 22:21:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/96/695596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/695596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/695596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/695596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/695596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/695596/1"}}},"commit":{"parents":[{"commit":"e1e1f4d4670fd5918661dbabe15e5921131fd4c5","subject":"Merge \"Fix misleading error msg if swift.conf unreadable\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1e1f4d4670fd5918661dbabe15e5921131fd4c5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-11-19 18:16:28.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim@swiftstack.com","date":"2019-11-21 22:21:57.000000000","tz":0},"subject":"Switch py2 DSVM jobs to only run swift under py2","message":"Switch py2 DSVM jobs to only run swift under py2\n\nOther services are fine to move on to py3. We still want to test py2,\nthough.\n\nChange-Id: I488f88a54948663c682b35dc976b62db6e72eef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e4fe1d5fa54ff8853b6bfe3924d5504fdb168e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e4fe1d5fa54ff8853b6bfe3924d5504fdb168e2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
