)]}'
{"id":"openstack%2Fswift~697535","triplet_id":"openstack%2Fswift~master~I36f0954fd9949d7d1404a0c381b917d1cfb17ec5","project":"openstack/swift","branch":"master","topic":"bug/1669888","hashtags":[],"change_id":"I36f0954fd9949d7d1404a0c381b917d1cfb17ec5","subject":"s3api: Better handle 498/429 responses","status":"MERGED","created":"2019-12-05 17:54:02.000000000","updated":"2020-01-28 22:32:55.000000000","submitted":"2020-01-28 22:29:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697535-1580250567849-159bbb0d","meta_rev_id":"992e2601d510045a31725ec6b5073dcab7377159","_number":697535,"virtual_id_number":697535,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 22:29:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-28 17:55:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-28 17:55:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 22:29:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04c049266295dc41dbd2554e1cbc98dcb8c5718a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-05 17:54:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a7354f82fea43e8a743fcaa0594aa924d9e52af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 19:36:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e79415fe47849a8bb8d0a78712118f3 : SUCCESS in 8m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8c986fe53d244257a40d1eedbe49bdfb : SUCCESS in 57m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89bb40a57b72453bb70be56d7eeeea71 : SUCCESS in 59m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6fe1ae59dd5240e6b778f8f0801c2cb0 : SUCCESS in 8m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4803173227d64aaea1249d235af99b80 : SUCCESS in 9m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f5bf10d30e8745cd97b58a1a9f708fef : SUCCESS in 14m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ebff332c40e4a0089a6de3a1b88231e : SUCCESS in 12m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0f862faa01ef40348f6ed3b59236ffb8 : SUCCESS in 12m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2b14255809f742c1ae4fe72c958aebc6 : SUCCESS in 11m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3958be39f9f84fa28c57cfbb8d4b3f31 : SUCCESS in 9m 44s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d1da21cacf50438ab3c7a891b2469060 : SUCCESS in 10m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae95d91a819449ea9158db3615315ac2 : SUCCESS in 10m 54s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/bf6b7f7ca2aa4df48dfa6b7dc312f988 : SUCCESS in 13m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3c3b4d2c4d70488986253c875a22a686 : SUCCESS in 10m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4aa2cde91e4747e7b91aa124fdacf4d6 : SUCCESS in 9m 27s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d87952f618224e609eb10223fe47cb5f : SUCCESS in 15m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/81beb379cfe74b7e8f782d65c57a7a7d : SUCCESS in 14m 43s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/04612540f33c48ffade1f505154603a3 : SUCCESS in 15m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cc30d74ce27b48dca95ef0a6f3b7cad5 : SUCCESS in 16m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a89adf568cd34db4af3448b2b09c1b58 : SUCCESS in 34m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76161222c82a47dbb41fc68f941bede6 : SUCCESS in 30m 52s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/87daa7d0262648aeb904112d550defd3 : SUCCESS in 30m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d72d9c28aabc4972aef5c5485eb1bce8 : SUCCESS in 32m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c79e5e6b565d4b55acfb5294a883814e : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26550921bf7e4545be5b8f293d1e385e : SUCCESS in 7m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b18ffb6593e740a78e074987eedaea03 : SUCCESS in 22m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad30429b3a964233a747a9aba1280d4e : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fbdff54220a5c26e2007951dfb7ddb6fe4fbe14d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-28 17:55:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1c79ff4adb7ffbbff265897f157d931704f1820","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:56:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56aeda1fbda791d0afa5557088aa810180fd3df3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 22:29:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a58ee1ca3c724f4183ada73a1c9b0a1d : SUCCESS in 12m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c1dad6485e854b1b9bba6837402c3f93 : SUCCESS in 1h 15m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8379fc6176554fdabdb882ca7c40c61d : SUCCESS in 1h 15m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f90eab46e43a4645ac597d840e19f63d : SUCCESS in 13m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/710d6dda3a5c40258805ab4d65bb2bc3 : SUCCESS in 14m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4454c67d34864ebc881dc488a658bdcc : SUCCESS in 16m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ba100075ce74a48b1b51ae1afe33c56 : SUCCESS in 15m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b3c367beb8064eb49b44e4d515a940ff : SUCCESS in 16m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b22c1f5694b64352993b9c6815ce8455 : SUCCESS in 12m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b7760f79724d43e0bd31021c94088bb8 : SUCCESS in 18m 39s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/051ddd6fb3104d17b0b919610df4e236 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/166764815aba4dd2a71005f3838e82cb : SUCCESS in 17m 55s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/579c43bd65e9446698819dd4157415b4 : SUCCESS in 20m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2e68930f3c4347eb969ea78bced0fb08 : SUCCESS in 17m 08s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/686d9c53426e4cb8bbd8e57c98516e83 : SUCCESS in 20m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b67b1ecb9d9e4ec3acd02e2a176a086b : SUCCESS in 17m 55s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ecc1885b47e84f21bd58a79f1bc58bf9 : SUCCESS in 20m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/04b8fd840faf4f00986a57f77ff3d636 : SUCCESS in 50m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8324d18ff0b846ea955d58f5c7fe94cf : SUCCESS in 33m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8cbf7a7e03524918b2ef5a0cd8fa7780 : SUCCESS in 49m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d80007ff7c2a4117850b3c8b6e05902b : SUCCESS in 33m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf912da5f808449e9b22c05a46e0f80a : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29ffcdfe4504445b67b7946d04dded9 : SUCCESS in 7m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57d4cd50cf9143d398ac98c2092c7213 : SUCCESS in 1h 17m 10s","accounts_in_message":[],"_revision_number":1},{"id":"84f640effcc76e30ba29fd5b65d780910d524d41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 22:29:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e6cc2313a94f6a4fdaf0c820362614e2fb8f4bfb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 22:32:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc873d2a431a4b9aad842e763e1aeda6 : SUCCESS in 2m 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/75a1ed5c06ba4f22bb75ca76f3153dcc : SUCCESS in 1m 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2f6550edd3d84631ba2737a6ce6bfef3 : SUCCESS in 1m 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f33c061ae92f33ea467c58749d3f15d2b1cc942c","revisions":{"f33c061ae92f33ea467c58749d3f15d2b1cc942c":{"kind":"REWORK","_number":1,"created":"2019-12-05 17:54:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/697535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/697535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/697535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/697535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/697535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/697535/1"}}},"commit":{"parents":[{"commit":"d4b25e3346c2b19bdb8a4ab838e06da13ff1ee43","subject":"Merge \"Turn off dns_compliant_bucket_names for ceph tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4b25e3346c2b19bdb8a4ab838e06da13ff1ee43"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-05 17:42:33.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-05 17:42:33.000000000","tz":-480},"subject":"s3api: Better handle 498/429 responses","message":"s3api: Better handle 498/429 responses\n\nCurrently, they just 500 as an unexpected response status. Much better\nwould be S3\u0027s \u0027503 Slow Down\u0027 response.\n\nOf course, that\u0027s all dependent on where you place ratelimit in your\npipeline -- and we haven\u0027t really given clear guidance on that. I\u0027m not\nactually sure you *want* ratelimit to be after s3api and auth... but if\nyou *do*, let\u0027s at least handle it gracefully.\n\nChange-Id: I36f0954fd9949d7d1404a0c381b917d1cfb17ec5\nRelated-Bug: 1669888\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f33c061ae92f33ea467c58749d3f15d2b1cc942c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f33c061ae92f33ea467c58749d3f15d2b1cc942c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
