)]}'
{"id":"openstack%2Fswift~699022","triplet_id":"openstack%2Fswift~master~I25b26e1fa61f2787e66ad1cd5497de331ec9d5d0","project":"openstack/swift","branch":"master","topic":"etags","hashtags":[],"change_id":"I25b26e1fa61f2787e66ad1cd5497de331ec9d5d0","subject":"Test normalize_etag in swob","status":"ABANDONED","created":"2019-12-13 21:19:49.000000000","updated":"2019-12-14 01:56:08.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9290e8632da09106db03f5555042987974e0b134","_number":699022,"virtual_id_number":699022,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":-2,"date":"2019-12-14 00:15:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-13 21:35:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-13 21:35:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-13 21:35:38.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-12-14 00:15:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"088886838458844fbb9700792a20febba75b68db","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-12-13 21:19:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9b772f974b84ce32b4f3c7979e98875498f220a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-13 21:35:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ec3328d56dcb09661431c1d78b91a7f5ec8f9bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 21:57:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e61a4d0d101d4e3db1161be0abc327df : SUCCESS in 10m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1b5b4fa73bcb4513a593c69c65ce5911 : SUCCESS in 10m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/745b1b1752a7431d881994e6d0eb390c : SUCCESS in 18m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/324958fe38094a829f00fd1f6bf12fc4 : SUCCESS in 19m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee54b532261c43b7aaeb74958cdf5431 : SUCCESS in 25m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1608584229a24ae583c36e8508010c10 : SUCCESS in 21m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b9b5f735d545487a9d55793ef02d6997 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8ee5b7eec33041f8b119cedbf2797b0c : SUCCESS in 12m 59s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8da36d8c5a5346fe81be87c714c07d50 : SUCCESS in 12m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9f03806e134d40bfb11e79b62d105a29 : SUCCESS in 13m 03s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/54b435e609104a929d8956d19dd3979a : SUCCESS in 16m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1ff25e5cfd734c9899f9c0203563118e : SUCCESS in 13m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/439bc1d669b64b62a48825d070f069db : SUCCESS in 14m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/056ec6f39c7144b9b2ab44510f738335 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cbcbcc772d4b4dbc987e1b3ea724b2f0 : SUCCESS in 15m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/990027b1ca504883bdc41dfb283ec978 : SUCCESS in 16m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/096eb5e8771147049ce809c151fbcb68 : SUCCESS in 30m 51s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1a38f4c67c864a3880f5b288f7ac3885 : SUCCESS in 31m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5ece2321fc740c2b4a3ba06438becc8 : SUCCESS in 31m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eac136bd16304a42aa3f643e105294dd : SUCCESS in 14m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9004da4e013b4e69a5395c4a26ae75ea : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/623e64a6bf904c24952ecc7c330ac9d9 : SUCCESS in 24m 40s","accounts_in_message":[],"_revision_number":1},{"id":"731768e90e1d8cb16e4e4a7d68cadb139b53c489","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 22:35:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1053b1d1f5fcd447fe547b65eac4363eac2b51ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 00:15:39.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"6a230fb32c2c36d6bcc9c61fa1b8bb5be4ee10db","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-14 01:56:08.000000000","message":"Abandoned\n\nSquashed into parent:https://review.opendev.org/#/c/698799/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c67545d43fa9eeb8b65181ba48a3f653f83454a1","revisions":{"c67545d43fa9eeb8b65181ba48a3f653f83454a1":{"kind":"REWORK","_number":1,"created":"2019-12-13 21:19:49.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/22/699022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/699022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/699022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/699022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/699022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/699022/1"}}},"commit":{"parents":[{"commit":"0aa888e1242ff25c3bf1b5e171e9288e678cab5e","subject":"Add normalize_etag() helper function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0aa888e1242ff25c3bf1b5e171e9288e678cab5e"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-12-13 21:18:30.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-12-13 21:19:21.000000000","tz":-360},"subject":"Test normalize_etag in swob","message":"Test normalize_etag in swob\n\n... and drive-by a import rename\n\nRelated-Change-Id: I1eaf075ff9855cfa03e7991bdf33375b0e4397e6\nChange-Id: I25b26e1fa61f2787e66ad1cd5497de331ec9d5d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c67545d43fa9eeb8b65181ba48a3f653f83454a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c67545d43fa9eeb8b65181ba48a3f653f83454a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
