)]}'
{"id":"openstack%2Fswift~699514","triplet_id":"openstack%2Fswift~master~Ib2df3ec683c1670cadba72d1fc0d666975f64aab","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ib2df3ec683c1670cadba72d1fc0d666975f64aab","subject":"Forward-port stable-release changelog entries","status":"MERGED","created":"2019-12-17 23:11:56.000000000","updated":"2019-12-18 01:21:59.000000000","submitted":"2019-12-18 01:20:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699514-1576632023913-716aec10","meta_rev_id":"894b8e699a99fd241e7d8be717ff9c34f162013f","_number":699514,"virtual_id_number":699514,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 01:20:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-17 23:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-17 23:46:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 01:20:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4526fc94beec761e98d7e8240377eaf522f00957","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-17 23:11:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0ce92b85bbe2c163122ab3cfb386c3c683695f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 23:30:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33379db3e63f4f67b2dc804355328489 : SUCCESS in 15m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/148218b63ef44b96849dd340e71668fa : SUCCESS in 9m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/39fb4103ca804be9b82bc26eabdc5e5f : SUCCESS in 10m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dcd4868f921f456fa1eed08271a54d16 : SUCCESS in 14m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/54202b081b0645a88f433fbc4cfa1396 : SUCCESS in 15m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/98e8d7328bf747789e3fc1ed5a15fd79 : SUCCESS in 14m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec2a5233c5fb4e1dbcb72b8b04410669 : SUCCESS in 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/018d0c52323d406282fd1fb77bbdca2a : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"46e4c3505b07a4f780c083740456ff68d31d8948","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-17 23:46:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"69c50026501a43fd3f8dba67eec67daa1a0197b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 23:46:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8de6fa9663cabdc9be0c1de1213f18987f92c0c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 01:20:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48534fd83eae4bd3b65972b4f499972f : SUCCESS in 8m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2bcea505f46b420793977b9383fb6301 : SUCCESS in 10m 41s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8ebfc523bebc4da1a7ac60d285c58547 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e43f033316a4f6391689d51a2bef64f : SUCCESS in 14m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0d902c17779243858acee29c389bf75a : SUCCESS in 13m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0480001dece045eebb3838021c3f190e : SUCCESS in 15m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/53bc9a28f2c443dbb6a914c081f92c52 : SUCCESS in 10m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e470e12708d64be0b39127dff110a7bc : SUCCESS in 13m 31s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4b4a1bac14aa48b29de893bd56b26d4b : SUCCESS in 10m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2983b47149524a8aabc43b59a655ef25 : SUCCESS in 11m 38s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a055b53f409c4d25b124b2b820f30f9e : SUCCESS in 13m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/56a89a6aa02f4b7bb021f2c8dbfa3b07 : SUCCESS in 11m 08s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/deee68672e7343d89d5e75159c2e4ada : SUCCESS in 14m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/494af099dc2244809a9800242794d4a0 : SUCCESS in 12m 34s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/bfb81ffe6d0b49d18d6e8db98895a027 : SUCCESS in 16m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0297f53be9064a5fa9f3d7a42220b2c1 : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f663cfd738ca4d6aba721a137c6f142b : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9356fb7cf6a39a651134634e59f58f1afb2fc268","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 01:20:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"894b8e699a99fd241e7d8be717ff9c34f162013f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 01:21:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d01539d83d142b7bae399c9206bb112 : SUCCESS in 1m 14s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/136ded272fac4d58acb2432c1cd1fb0a : SUCCESS in 58s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2eec69340d14453c82d4ff66118d6ca9 : SUCCESS in 52s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c94f625fe463d808d4bae1137739a4dfcf6f4cd8","revisions":{"c94f625fe463d808d4bae1137739a4dfcf6f4cd8":{"kind":"REWORK","_number":1,"created":"2019-12-17 23:11:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/699514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/699514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/699514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/699514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/699514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/699514/1"}}},"commit":{"parents":[{"commit":"e890b0f0fcdf751a2058c1e46da090648d9a0dc8","subject":"Merge \"WSGI server workers must drop_privledges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e890b0f0fcdf751a2058c1e46da090648d9a0dc8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-17 23:11:36.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-17 23:11:36.000000000","tz":-480},"subject":"Forward-port stable-release changelog entries","message":"Forward-port stable-release changelog entries\n\nChange-Id: Ib2df3ec683c1670cadba72d1fc0d666975f64aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c94f625fe463d808d4bae1137739a4dfcf6f4cd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c94f625fe463d808d4bae1137739a4dfcf6f4cd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
