)]}'
{"id":"openstack%2Fswift~700056","triplet_id":"openstack%2Fswift~master~I380c6e34949d857158e11eb428b3eda9975d855d","project":"openstack/swift","branch":"master","topic":"etagquoter","hashtags":[],"change_id":"I380c6e34949d857158e11eb428b3eda9975d855d","subject":"Middleware that allows a user to have quoted Etags","status":"MERGED","created":"2019-12-19 19:26:58.000000000","updated":"2020-01-28 03:17:54.000000000","submitted":"2020-01-28 03:15:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700056-1580181356776-87017a1f","meta_rev_id":"0c39766ee1456d3540d878596a49a7b177309f55","_number":700056,"virtual_id_number":700056,"owner":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 03:15:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":2,"date":"2020-01-27 22:31:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":2,"date":"2020-01-27 20:54:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":1,"date":"2020-01-27 22:31:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 03:57:15.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-01-27 20:54:31.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-01-27 22:31:26.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2020-01-28 03:15:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84994043bbb2b9bf3b62e12839e8a6f9a2f9d706","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2019-12-19 19:26:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df06aa1dade6f5099477352cf9a3d384bec4678a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-19 21:05:01.000000000","message":"Patch Set 1: Code-Review+1\n\n(6 comments)\n\nLoving it! I\u0027ll try to clean it up a bit, too.","accounts_in_message":[],"_revision_number":1},{"id":"d20f5d7bd64653301f89d0eef9bb28df3e29506e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 21:05:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0efd35ca7d342b4bf0171fa29a11983 : SUCCESS in 8m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2f54e941b17f4763ac16e499d9520771 : SUCCESS in 1h 01m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f4bb6beefdb42beb6d10209b27ea4fc : SUCCESS in 58m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6d9e4daa2be144039cec1acb437939d4 : SUCCESS in 13m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/640c5a7735a942caaac11182c605cb7a : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fb7ef0fa98a8470e98bdb434cc13439d : SUCCESS in 11m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35930c02c57e474c9a74801d00ae2324 : SUCCESS in 19m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9651601e84684ed79cbc82b54f4d1086 : SUCCESS in 20m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/26e7eda806214e248070a2dc40c4ae9f : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ae2ecdf6d6fa4642a5adf6cdf8141256 : SUCCESS in 13m 15s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/97a26c4b9eef46f497140b606751126f : SUCCESS in 9m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0450e2eb22ab4ec2a4ea1a08e42c4c71 : SUCCESS in 10m 41s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/dd669f675bf346efb7c6d3d16d566ed5 : SUCCESS in 12m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fa7463f3cbb74e2799fb2614f8a62b1c : SUCCESS in 10m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e26cba534cd3452392007c3be17d93d6 : SUCCESS in 10m 28s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/93ac102990064ec29e4a220b3b9fe566 : SUCCESS in 10m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0522bfb29f284de49eb2ff903bfe99dc : SUCCESS in 15m 10s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/fe4413544ab54b60874481c0f17aff9e : SUCCESS in 15m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cfb280cc05a64fa899c7c4bcecbbc7dd : SUCCESS in 17m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/57ef1ffad52c437e880b311383f39258 : SUCCESS in 42m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c968e1b7909d41728d757506d2b1a0c3 : SUCCESS in 33m 58s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/76905ed6e69b47edbe4299fb7a8e7e32 : SUCCESS in 31m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d4e205fe8473486e9c9d1d0a4707e5d9 : SUCCESS in 32m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4937d2510e6d4225bccc6b015fdabfcf : SUCCESS in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de4db2d2ec8e4572bb9af7b23ea397af : FAILURE in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e9fc56ddc2164d19977be53267f6e68a : SUCCESS in 23m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40648fea13284b379587bd82abae04dd : SUCCESS in 1h 27m 08s","accounts_in_message":[],"_revision_number":1},{"id":"8e3ce8056e7aae09de22cc13918c96688396d2a6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-20 19:09:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"65b9e9ce90084ce182a7d134855119371b8b44cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 20:17:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6ebb2d93083b451bbaf2e420e9e9641b : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b860b57f524123be5938c768bc92e7 : SUCCESS in 9m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fc57275458a041579c8bd50b569ead93 : SUCCESS in 1h 02m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/356dc6261780459188e6123e68abcb4c : SUCCESS in 58m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cc4a758ada934130893d9055c7bb198a : SUCCESS in 8m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/81fb3c8a80d14b5a854ed689d0b1b460 : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0e3ee03f864345f4b333a71206be3e54 : SUCCESS in 11m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/319c67bf124b45f5866946095be64c39 : SUCCESS in 11m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c563d47e82c848a294e5e873c0b8046c : SUCCESS in 20m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9d2cbfe4c1eb4bd29ecdc2667c51fd7c : SUCCESS in 9m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/febd15402f074a85a5a4631802e98f2e : SUCCESS in 11m 08s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/134269cc44fe4e01a69c0a82ef9998ee : SUCCESS in 9m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c914cebd86414888ba8323a5ce629ab7 : SUCCESS in 10m 32s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d65ecc04bd3f45c7b0e76825af0a7658 : SUCCESS in 13m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bd60abc8cf65407ab404626aad88a362 : SUCCESS in 10m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e745721150194403833e6df5089d5733 : SUCCESS in 9m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/45570314d7ac4e52b099de7d583c7cb8 : SUCCESS in 11m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f1fd26daaace4eddbebc6b410e6776fc : SUCCESS in 11m 02s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/122b12fb7638409aa88d0af37b06e645 : SUCCESS in 14m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e218f703dac04b43a294a3330c222b91 : SUCCESS in 17m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/be8c530459c34c34b21460e05cfa2404 : SUCCESS in 33m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57ba99eee2ce420890c8a09e837f9784 : SUCCESS in 34m 54s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5408dd2826614df4a5c3c50ec0154bb9 : SUCCESS in 31m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1f23d6504eb94c52b586c1fdcb62f452 : SUCCESS in 35m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/834d60fcc3dd418aaf35404c7d58982e : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c7ac0be37c14bed892ec7c8989ed74e : SUCCESS in 6m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f59b9c917f0d4fb8bff75a126c2b12a1 : SUCCESS in 17m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8e85ad1ad43426ab65154883202a8df : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":2},{"id":"4737356be2627e8a30b1275e0d34718d34ccc536","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-27 05:39:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f4f28aa70d25419c589957d08479d118856e5dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-27 07:15:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0066a5f234b04cd9889c46dd5da5d398 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39bade071e9a4b759c682cbe23dff401 : SUCCESS in 8m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b3759a8733bf40f3990dc04f1c7e9788 : SUCCESS in 59m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad4b1ac1399f4d20802aa6bece78c76e : SUCCESS in 56m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6175adffe7c44f4aea5b1698b3035bb : SUCCESS in 7m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/61bd4535fbc34a928996e8a3388cafdd : SUCCESS in 10m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8310ce06b89a4b8386eb98fbd9885011 : SUCCESS in 14m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6876f075e0dc4a5c842ac49128bc4752 : SUCCESS in 14m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/092151ae482044aca63559333f69642e : SUCCESS in 14m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c75d74ce799c43ae912f16c8243d22a7 : SUCCESS in 11m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db119e9788d44148ba35fb04940ffcc3 : SUCCESS in 11m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d82595ffe01f492eb044b4f8454d60a7 : SUCCESS in 11m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7ebeba851b744d60b301b280497dcec9 : SUCCESS in 11m 53s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7cb592c93bc44ab7a1e2b77896184336 : SUCCESS in 15m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b2f85589e5a147f6875009856988c109 : SUCCESS in 13m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/64e5af50e44e4d158ea5382550c962e7 : SUCCESS in 13m 05s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e4c09b5f0d5e4c8192d9f3ea80e11212 : SUCCESS in 13m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cd76ca11c5444e2b8f0372325b8b30e2 : SUCCESS in 13m 38s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c9a1316393cb4263b97d270fbee208de : SUCCESS in 16m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3058fa9d8a1b493e88fe301f20d9d552 : SUCCESS in 19m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4080f61261fc4c2b99a05714d453bb9a : SUCCESS in 30m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5dc1ec4d7e134934b269b69dd7d228d0 : SUCCESS in 37m 46s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8e6f0fc0bd5a40d48a8b1f7a1bd4c9f1 : SUCCESS in 36m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f447cd2016541da88e22b1a614979d2 : SUCCESS in 31m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/296a0261232c4666ac0c4704b869a51a : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/661acfd7025348d4849a578e2d85978e : FAILURE in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cee7b0a657bf435c9c8e0134ff51888e : SUCCESS in 29m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef432a9c3f2b44158b6b7ea69fc5124d : SUCCESS in 1h 05m 11s","accounts_in_message":[],"_revision_number":3},{"id":"e648c2cd530c9e0afb0fca5adf04670585a0967e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-27 19:36:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8210f49806c0f5e553af8536ce156c03810dceb4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-27 19:46:06.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6736193d46afdf183eb6eff221b6edc425239b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-27 20:54:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/973da26cee704e71868420baa30bd86d : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9730b974f7c54d32abd48422255918ed : SUCCESS in 7m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5d177daa8d4f458c93ac9a35d0cb73e4 : SUCCESS in 1h 01m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d5069af68c7544ff831185927ec641aa : SUCCESS in 56m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b9f19ea3f9840fbb794f2fa6325f6df : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cc0562ff26284af6a34524368c632e2d : SUCCESS in 10m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a708876a7da4531a9cd3448ba43e857 : SUCCESS in 12m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4399a8fa464549a3b7301913e94a9005 : SUCCESS in 12m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7423d7e7e3894d1e87d70c8f7c93acdd : SUCCESS in 12m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71f585571984486e9c92332c70f963c1 : SUCCESS in 10m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e24e094f686a413186da4629a33da909 : SUCCESS in 10m 04s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/cccf38b4faaf4d7abf855d7dd64139e1 : SUCCESS in 10m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e4eabdbe68dc4f00b346fbcd4a070c44 : SUCCESS in 9m 51s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/fcf3525482264b6da09a7c2505b2676d : SUCCESS in 13m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a9fda69f6310422896da4cf9df061b5e : SUCCESS in 10m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2a440099d9dd4fa6945d9f33197c0c1c : SUCCESS in 10m 35s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/14e0e6e8472b40f9932e1af5eb08420e : SUCCESS in 10m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7aecf081626b4460a2bd58d98d7b78a0 : SUCCESS in 11m 39s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/26b123039a084dc39a059de73093b79c : SUCCESS in 13m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e664adf7ab064811b7e74f44dc1305b1 : SUCCESS in 17m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b3ef5b09a67948f080f156bca85bb8c5 : SUCCESS in 31m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/82dc82aeeda4428fae19d908ebe447f9 : SUCCESS in 32m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/93048071f59f4e15bdb2890ec9582161 : SUCCESS in 31m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea3cca36018f40fa914fe18007f95272 : SUCCESS in 31m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0f0d4ca562a413d8f749834cf35fca3 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f82b117d0841e38d0c307091d9f802 : FAILURE in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/128df1c37e3047e58b52909d5cadd2ef : SUCCESS in 17m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62c0ea935d4445e9a3c294c963750181 : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":4},{"id":"ddbd0718eae5ba344b50957ccc3795b99fd48741","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-03 06:16:01.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c4acd17d7173a627b0bf41947e826151e6c4b20d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-03 06:16:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1e95af9f209a86238039e7d6e31a44c9e0deb699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 07:45:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/52bd8367d31b470e9f20b137539787f3 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec859db13275452c810ddc5eb2a86b98 : SUCCESS in 9m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/79f58dbf0718441d8b13815a93b15bad : SUCCESS in 1h 02m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a7790b24b9be492da6fb21622756ea3d : SUCCESS in 53m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ec350e7e550423cb51a69d4aca98e75 : SUCCESS in 9m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0ac2cdcad19c4ac48c4c9dad70d081f9 : SUCCESS in 10m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/894c8c01305c4ffba12966c6ebd606f0 : SUCCESS in 11m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/273c3f295d274a43ae9388dc07b863f4 : SUCCESS in 17m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9db98cc6953740b18d2bfd3e4ec5e554 : SUCCESS in 15m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6f7ed7cc4e5641acadc6c0762ed46a44 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/59ab0d69516c4cbfa720279b48554584 : SUCCESS in 12m 36s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/142486d74bda490f8fd9ffc8b33853a5 : SUCCESS in 12m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1113b71d58004b7e9034ea57d58cec62 : SUCCESS in 12m 29s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/4ff9979055ba435498b65d3106295794 : SUCCESS in 15m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ec17c11d7aba4f4aa90c700e0071aee8 : SUCCESS in 12m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dfaca2766c224dbb98c86ebfb411c898 : SUCCESS in 12m 09s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3666c11f45ab454c900cdb47f41d0221 : SUCCESS in 11m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8f59fd4828c0438d8830f42bad69cd73 : SUCCESS in 11m 48s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/557f00a13e504bd48d4491b4ae4990c1 : SUCCESS in 17m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6a5f158291214611bb315dbed84c064c : SUCCESS in 19m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c0e384f531ff4270a11b59441d3223be : SUCCESS in 40m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d2c6d0f6bb454b688f5893f4db6a0f55 : SUCCESS in 35m 06s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/cc00f17a4f054ef9b06debf785b41752 : SUCCESS in 40m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f362ae1e00d448089fd0fd21d9bd7c83 : SUCCESS in 31m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2301ad5a565d49a9b3837d14c6931d95 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78b1594895534544821e9d2ba265c6ba : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c255f79c4064625a0fb7ecfa0fc61dd : SUCCESS in 18m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f049eb325724ef881f4775ae2f3d991 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":5},{"id":"9c7d203f1f98b5b3c9a58ceeea8c0273e1f14ecc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-01-08 21:52:12.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)\n\nunittest ratio looks good, maybe some func tests?","accounts_in_message":[],"_revision_number":5},{"id":"d6693935d5808c89c1ec2662e4f9f8d8672f333e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-08 22:05:37.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7303e59e22a3ba418174d6952818e135f2cc2853","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-16 01:19:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e81d79f5edac4e4b71025c3fdf2e7934ec448880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 02:27:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8b3bdfa7a60b48bd836ef0ba145d73a2 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e29a293d584653b26bdc2eeb715132 : SUCCESS in 7m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/be1d4f71295148579d6768b55ef785fb : SUCCESS in 1h 03m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3895347c3478458bb5332c8fddf6d81b : SUCCESS in 56m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e20c55fcb0141999e58fa3660aa8d74 : SUCCESS in 10m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d9509444d2164d079e1fa91542bafeea : SUCCESS in 9m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/da72d0cf44894272953fa0f9dc9029ff : SUCCESS in 12m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8cb39a9af4b94943800083ed2cb96ab4 : SUCCESS in 12m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/196de7767a6e4ac1b81e3ab98613ef3d : SUCCESS in 14m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4dbe53de97ed42108bc050ef7d762e21 : SUCCESS in 11m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df28cddd887340f0bf2223a1bcd0920d : SUCCESS in 11m 25s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ec63727bcdaa4d299ad6332c053a5fbb : SUCCESS in 12m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f52452744d804fbbabb5d0653d293a17 : SUCCESS in 11m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e8886445d37d4f9ebd9c5314d4e22574 : SUCCESS in 14m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8aa9e818017543cebcad0bfc371fb28c : SUCCESS in 9m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fb14e254b5fe49fa94c95412b807fe2c : SUCCESS in 11m 30s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ac1913c9da37449ea1858ef83d03b13e : SUCCESS in 10m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d34817efd58744a790a8cb3166251376 : SUCCESS in 10m 59s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2516b4a46cab4b318ac6bffda7f67f61 : SUCCESS in 12m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3ed984c1b76a48129d33be9dd5f4f8e5 : SUCCESS in 18m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3d20112134de45339bbf2d6127ab4eda : SUCCESS in 44m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/429f8dee66c94beab40d9cd50d00b718 : SUCCESS in 32m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/bc9ab34b89a54e55b0267034dfd3d294 : SUCCESS in 29m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d1d881b4233c42c89692e7955c8600dd : SUCCESS in 33m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b9549193fd443fdb0b2c8518509b661 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4f41427e9204251a23d3f526cc0fc11 : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/732399b4616c483b9f94d3cda0d9f489 : SUCCESS in 20m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75496157a6f1439f818f0f70c052ea92 : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":6},{"id":"b67e8402d9b5d0173660a99520a5d87a2c8f224c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-01-16 03:57:15.000000000","message":"Patch Set 6:\n\nfunctests would be easy right?  you just create a container and turn it on?  maybe upload and object before you turn it on and make sure it still gets quoted?  I guess you could mix in some symlinks or SLOs...","accounts_in_message":[],"_revision_number":6},{"id":"4bcdb6aeda63faeb54b64b4715ff2dc2c6fec43a","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-01-16 04:24:00.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b7ecbd2178d7ef16af6417ba5a178786f34eac44","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-17 21:55:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6d461c38a693f48bba276b929514917e017c574b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-17 21:56:09.000000000","message":"Patch Set 7:\n\n(3 comments)\n\nAdded a simple func test; I think the unit tests are sufficient that we don\u0027t really need a func test for SLOs (but maybe that\u0027s just me).","accounts_in_message":[],"_revision_number":7},{"id":"cdd612a06fe5ed06efe68fca932dbe70baa49a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 01:08:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8f3887f619ab4f9eb2270a6354bb2f23 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a33b48d3629e40938b5220e1265889ca : SUCCESS in 8m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e8acb3bd563948b3bba055997ddf5fa5 : SUCCESS in 1h 02m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e17d099b8674da58c49acec3dd7284f : SUCCESS in 52m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/45fafd8de31b40978742f189ebe7f222 : SUCCESS in 10m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8f60b09b39a144549814bd3f0dc05ceb : FAILURE in 13m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/163ea72adb6b4c9b8c9eff2598f1f6aa : SUCCESS in 13m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1c5dc3dc83fb4b1a93d553655bdef37b : SUCCESS in 16m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4e617ed03ddd4f5cafcf43f25d7e2be4 : SUCCESS in 16m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c2072a3f9394632b6ad63222788f4c4 : SUCCESS in 10m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9e77a05fd5d9425980672ed67a120eb7 : SUCCESS in 11m 29s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/cdc90d0c79654a93b85879dd55d9cb99 : SUCCESS in 12m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7bfb44accb6347afb7023237641e69b9 : SUCCESS in 12m 16s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/9559437a48c943188e29691f83b3d725 : SUCCESS in 17m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b61491a0f95f46a48c932e78e828f661 : SUCCESS in 11m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cb412b108a2749859958c24afcc7b5c0 : SUCCESS in 15m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6bad3bebad484e5e96d33b9338549a55 : SUCCESS in 12m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/729c97a9cfc14b4396a1faaed4cc6af4 : SUCCESS in 11m 39s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3650d2c462e04ea7a221719fd2a0a688 : SUCCESS in 18m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a8eb5ae01bfc4437a859502e62c0d4fe : SUCCESS in 18m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e67416fe18ca4081beb053e42702e908 : SUCCESS in 45m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3bdabd4b25b46369e5990324b5cf252 : SUCCESS in 32m 29s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/84d0269a058a4b31b748d244c4093339 : SUCCESS in 28m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8b4ef81e04874fefb28a3cac6256e9d5 : SUCCESS in 32m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/127ca305dc744f87a247c9aa0801497b : SUCCESS in 14m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/467d8e5186174632ac0a245e6bd1ac20 : SUCCESS in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e4c84abf29364500ae1a6888bf33b266 : SUCCESS in 31m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfb69796a38041dabf67a634a10bc921 : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":7},{"id":"7f9682af76786694496402d6f0c6c8623120514d","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-01-20 05:31:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b34c8011b5a4b8334979b8d062f1bc86d6b911a3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-21 16:47:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9a49c3e184c2e4143926d9bd06ffe10c819bf6f6","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-01-23 21:37:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a496d95bb2ac4dd72d69a432751c9fe11dcd7198","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-25 01:53:02.000000000","message":"Patch Set 7: Code-Review+2\n\nWell, I think I\u0027m on board with this as-is. Could probably still use a bunch of unit tests to make sure I got the setting/clearing of metadata right, but the added config complexity seems worth it for now while we figure out just how disruptive to clients it\u0027d really be to start quoting all etags. Really, I feel like the proper solution to that complexity is something closer to https://review.opendev.org/#/c/504472/\n\nHolding off on a +A since I\u0027ve had a pretty strong hand in the patch.\n\nI\u0027ll see about turning https://review.opendev.org/#/c/695131/ into a patch that gets func tests to pass on a cluster with quoting on-by-default.","accounts_in_message":[],"_revision_number":7},{"id":"fc6b08bd4c034426271cc2ce051d6e483ee707fe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-27 20:53:41.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"654f741bd02c1dddb7f2330cf8b33a90c8b68fd1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-27 20:54:31.000000000","message":"Patch Set 8: Code-Review+2\n\nJust fixing up a conflict in the docs for swift_source.","accounts_in_message":[],"_revision_number":8},{"id":"7b658ef3394104d3fe16fb297cd25d08acfce433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 22:29:20.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ee9d77e745dd41ff9d6a31e53b81d5de : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6677eacdf9647c392c6a4c46ae9205a : SUCCESS in 8m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5c3895176ecc4b76a02d9caeda193a75 : SUCCESS in 1h 00m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5451942079464b898809011a54d8c8af : SUCCESS in 53m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dc4448e3f56240619b21b35bc6297367 : SUCCESS in 12m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/55b5631301794f02a5e067a804cbd711 : SUCCESS in 11m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ab2828996e947a282431b5b0d3349bc : SUCCESS in 19m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b8e378cdd0c84656a924b6df7231a7b8 : TIMED_OUT in 41m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c08338989952417ab4979ebb119f6165 : SUCCESS in 16m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/30383316f320460cb693767b528c0702 : SUCCESS in 10m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9089c673f4e4442c895e78b8658ea876 : SUCCESS in 11m 25s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/66d3aea470e945608b7f03c9acd324b3 : SUCCESS in 12m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1589d6d6822b45d48ff5c3d9b8f9702f : SUCCESS in 13m 25s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/eb42873be66d4438a1434bb58f3de7c8 : SUCCESS in 18m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8e9fd6843cb94c0badadc266a6eaf12c : SUCCESS in 13m 12s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ce86b4cb06f14998a4230c88f65375e2 : SUCCESS in 16m 05s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/dfe556f760d442ef8302e25a2d9c8fde : SUCCESS in 11m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c98aaf35078d43b0a1aa57f051717e91 : SUCCESS in 12m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/07a45380e2d34ce2b7818b98e8829855 : SUCCESS in 18m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7fc365ef478841a497e94b4d8313ef12 : SUCCESS in 22m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e5c2ee18c3744b1abe33d8c863e42fb9 : SUCCESS in 41m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fa506239612e4b1e8380f663bd7f09d5 : SUCCESS in 33m 24s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d46e2feac7d14712bfe5d46084d0e226 : SUCCESS in 29m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3af9bd5fb1de4b8da851368a229c1dc5 : SUCCESS in 36m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22d5852717974469a1e4627c1d73aaec : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5363da75ebb462d9fb34221746fad44 : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/abb7e19bb23541709671652f14332450 : SUCCESS in 23m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e7db491a67b4ebf8f0fd7de87617dfd : SUCCESS in 1h 14m 12s","accounts_in_message":[],"_revision_number":8},{"id":"b7a8094723226670e61251b09ceef61cecbaf66c","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-01-27 22:31:26.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"070f1be0b8ba5dd57db6f301d18ee895574815b3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-27 22:55:06.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"f6f7c41782dfaac3b33e65032f0d7962efef9f2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 23:44:26.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f2a664bdf7244589ae90dc22f6862335 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863e4a011f104cea84f459df63096e6c : SUCCESS in 8m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5d3cbdabead04877812814011bd88865 : SUCCESS in 53m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d455442394b849de911fb7773c080fb8 : SUCCESS in 47m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/44d1aa9954384fd4958848e6468fed56 : SUCCESS in 11m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7256c409f37b4e86a3ab9c52e7655917 : SUCCESS in 9m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/82ec0f4b6ce14a7984e085b88e5ca512 : SUCCESS in 17m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7bf1b064ad56433db94cbc68e8ef00e0 : SUCCESS in 12m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d293f98e22894146b48956b475e7a377 : SUCCESS in 13m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ca502242f6174588aaa0a98c488044c0 : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/74659569a2c543fca997bc3abeb682ea : SUCCESS in 15m 53s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ccff683163e04c278c84b6a70cdb8e46 : SUCCESS in 19m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dc49b0d0dcff4947ad788a1b45a36a1e : SUCCESS in 15m 24s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6af985b63fac4add86de436e6cafc063 : SUCCESS in 30m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c8348fd43f1d415097959679055025a5 : SUCCESS in 24m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3a67fcfc29e144c9843e45094a896991 : SUCCESS in 24m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ba95c996d9774aefb34d999ee265358e : SUCCESS in 13m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/341594f001164af4870622be9f49b5c3 : SUCCESS in 15m 50s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/febb45f84f3944288c800763dc258462 : SUCCESS in 28m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/53f529e091264cc7a3cdadefce304045 : SUCCESS in 27m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cec5bb61380f4f369f8f71cd44d7dd9d : FAILURE in 44m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d6ba4541608040e89758fd564b8512f9 : SUCCESS in 30m 32s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/baa2d7e5f50a4d7ab259c7c01e948b89 : SUCCESS in 44m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6f8f524b80ae497291e36d51de3b04aa : SUCCESS in 38m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5570cd97f85142c38429bc0f0ca5c53f : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ababe887eaf448cf99485fa7c23351b6 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d8fa8e05f6554f5b906c96c982038004 : SUCCESS in 34m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d295cdef6034fb58ec67fe1022fc5b6 : SUCCESS in 1h 05m 03s","accounts_in_message":[],"_revision_number":8},{"id":"0993d8852e84d53e2d706d50ce1025ddc38679a3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-27 23:47:52.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"af7f68edd88b87f744027e21f6d203384ef5c41a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 01:04:59.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f30b00c604464c1bb9fcdb6009c30a5a : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a44b63851824a6c874e237c407d4bea : SUCCESS in 21m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b15c74ab85ad445da637148dcc311b6d : SUCCESS in 1h 01m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bff795b96120456889734d6793c0e837 : SUCCESS in 1h 07m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b692c330e64467e9f8bcb791fb6c990 : SUCCESS in 8m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e02e4cdfca214d2693b1ed9d1d1faed2 : FAILURE in 11m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/506a02d22f18423e8cf0307433323633 : SUCCESS in 16m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/53b538fd3953486184e18b66380539c7 : SUCCESS in 11m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a2b209dfe8f24989b40ae6615388db93 : SUCCESS in 19m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2db66f76ff374daeafd3d514f73c91a9 : FAILURE in 16m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ce954e2fb3dd41989c461a96b8b638d9 : SUCCESS in 22m 19s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/463c195a7a984668afcf86aa8da586ee : SUCCESS in 12m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7e69c9919e134e25a1cff14e4eafe249 : SUCCESS in 17m 37s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/20142c7658284df4b632ecf4da711911 : SUCCESS in 25m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bbbf7832028f474ebe4eb0fbce9cb2a4 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/05569dcdd50749dcbd23692987cc47ea : SUCCESS in 22m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/76d4ee627f5f4bdcb187cfac569b3dbb : SUCCESS in 18m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b1794ffc224d468cbea575fe28b39a8c : SUCCESS in 16m 27s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/7f2de178243041dbb3984e0b5673abbd : SUCCESS in 18m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6b4f144ae7d4ed5a1d37b437b31f650 : TIMED_OUT in 31m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f6d53874cccd485e870b3d7695e4d8ce : SUCCESS in 1h 00m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8abe902d910240108c8396bf6bfca774 : SUCCESS in 35m 41s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0c64b51353d542f3bed5db3e982861f0 : SUCCESS in 39m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a7a47904c0844649c1ab0d1db5edbf8 : SUCCESS in 42m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af88b60c2e0f43f09608bad5b011b580 : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffa71831a4a140d9828ca3aec76c3f12 : SUCCESS in 8m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/73b210414d38460cbd64f359d3c63830 : FAILURE in 27m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b56fdf8d0294719b3a7bdd5e09ee563 : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":8},{"id":"a4003a18f101fb4cdd2bbf8b63562c8d9df3fa8b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-28 01:07:37.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"fbb248bc55f95f405adfe2d4b52a3a1c68ac8e88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 02:18:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ff0bb518359b4ac59604f5c6289e0c0a : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b240935cf81a4ee78e795c62eacd69ef : SUCCESS in 7m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4a450ab5dfe540fb86ff7d109d3ecd62 : SUCCESS in 57m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bce5e85557d240f8b2e3a909b2cba871 : SUCCESS in 50m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f39658c73804cb1bcc689ec49d31969 : SUCCESS in 8m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/40cf5efeb9b24ecf9f10daa283743309 : SUCCESS in 9m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ce5302c52784c31a4659b815cb89d1b : SUCCESS in 15m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2cbabe1082724d6aa1b835ab75aa6b9a : SUCCESS in 15m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1e80f3c41d944c1b9d79883952a9a9a8 : SUCCESS in 16m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/56b75bf4401f403c887440f43fa770f7 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f04bd8134dc04b6f868f81538a53b6d1 : SUCCESS in 15m 43s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/be005eaf634842ab83e473e5f6572795 : SUCCESS in 12m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d8bc066188964631af1009c33161fb7b : SUCCESS in 13m 46s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7a1d4b3ca7ab4a5d8bf12e82e6160757 : SUCCESS in 22m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/72ad1eb109124972b0e60e8b50f7e58e : SUCCESS in 12m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9f182cb8aa15468189e0fd68caa2ce56 : SUCCESS in 15m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/aa08a24c74c545c2b9edb6eec50f00a1 : SUCCESS in 13m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4d4fdca4e9754b9f9a6b757b71124090 : SUCCESS in 16m 52s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f3a01fff28e14a6781e9d1629a862f9f : SUCCESS in 20m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/63f040c6d557459e951b4a43618e0367 : SUCCESS in 21m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3b4915e74e024347a7392fb711d4ee68 : SUCCESS in 45m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f8f916ff1cef46eca6f86149a98ebca6 : SUCCESS in 30m 50s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/cd551d1117db403ca820836921cb09f0 : SUCCESS in 27m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/12acba3001e840df860a6c151d4031d2 : SUCCESS in 34m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77e2ff3bf9cc4455bf9b14983f17c10b : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48140b81c82b4a5ab463d4927722cea6 : SUCCESS in 6m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad6c71640fe04cf79ad1655bc7fc0035 : SUCCESS in 20m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6bda8b0807548d2bf3b4a66c0cca541 : SUCCESS in 1h 07m 48s","accounts_in_message":[],"_revision_number":8},{"id":"a15c6fbf7412cf2b1bb03270189360a1d2d363cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 02:18:35.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"bb88e2c446385496b26355fed919a97e70ff044b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 03:15:56.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5c800a19951a4771ae4748a9889c206d : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e397a46c5b9c44b485c3363dd4eacc95 : SUCCESS in 8m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/16ef83c5cdf24270ad9acbb0e296e2a5 : SUCCESS in 55m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/83b890f3136949f9b38bd9cfebcb9521 : SUCCESS in 51m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d5c4366f6ba646db848e30e556d2dca3 : SUCCESS in 9m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/455fd74c8e8648f69041cc50d329f956 : SUCCESS in 10m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1e2e1cacb19740f4a77f6fb533586f48 : SUCCESS in 13m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7aba8c942e834d99b349b72a635f8574 : SUCCESS in 14m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9f578a5c60e745158248c3f735d799de : SUCCESS in 12m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/abc6b81a98c340a4ab9b74a92c55644a : SUCCESS in 12m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9332f4e304f04ef78cd34ec8c5dd9f7f : SUCCESS in 16m 22s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/3eeff61f28b4407a8834225382e94a13 : SUCCESS in 9m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f96370d2b67241ef89c2f7ec0ba90ea9 : SUCCESS in 13m 00s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/76a548b3b6ae4697967afb6a9d1a8172 : SUCCESS in 19m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/431e8f4fe7cf4e22b55a4b37d87b6cac : SUCCESS in 14m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d6a5bfd0d4014d5aae28a4b3faa9d3db : SUCCESS in 12m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e85ba927abcc40cfa31b6d91f437f735 : SUCCESS in 13m 55s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/beec2c5c739940bf8a3ff0c0ae44602a : SUCCESS in 20m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b7516650e90e44489b624a883696f611 : SUCCESS in 51m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/41293a0dfa8b4a2baee451af5e0c95ac : SUCCESS in 30m 02s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9dd2eb25b0d34d95a7a592fe262ceb73 : SUCCESS in 32m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fc75237a346840cc963c38c4223c4d49 : SUCCESS in 30m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da8665d344b4482fb2dedac9dd98b4a1 : SUCCESS in 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55f0fc0048894bb2919197cf8af40fe9 : SUCCESS in 6m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b7af205dcee44729bf537d6ca4e1137 : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":8},{"id":"b3c465c1d29adb35629d9f2707b86a911a43c01e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 03:15:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"996dece2130ffcfb14836097912818b8d89e14b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 03:17:54.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e24c6c3d8e4f9db61a73554804d063 : SUCCESS in 1m 38s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/33ef5531da3e4b8badd60f25156cd0c9 : SUCCESS in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f456ca3ad80c4864b88724c9bee90d3b : SUCCESS in 1m 04s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"27fd97cef927f39a057f46cf09a6c38ed52c6aa9","revisions":{"959b76f4b3d78b4ee61771cdcba7885e82461c06":{"kind":"REWORK","_number":1,"created":"2019-12-19 19:26:58.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/56/700056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/700056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/700056/1"}}},"commit":{"parents":[{"commit":"5e9d5d101a8103b437c4332f79b16f21bf8dd9a2","subject":"Merge \"Fix type on py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e9d5d101a8103b437c4332f79b16f21bf8dd9a2"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-12-19 19:24:51.000000000","tz":-300},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-12-19 19:26:41.000000000","tz":-300},"subject":"Middleware that allow for a user to quote Etag","message":"Middleware that allow for a user to quote Etag\n\nChange-Id: I380c6e34949d857158e11eb428b3eda9975d855d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/959b76f4b3d78b4ee61771cdcba7885e82461c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/959b76f4b3d78b4ee61771cdcba7885e82461c06"}]},"branch":"refs/heads/master"},"7d69ea690883d6b99058e80ecf8467eeccb16d95":{"kind":"REWORK","_number":2,"created":"2019-12-20 19:09:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/700056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/700056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/700056/2"}}},"commit":{"parents":[{"commit":"5e9d5d101a8103b437c4332f79b16f21bf8dd9a2","subject":"Merge \"Fix type on py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e9d5d101a8103b437c4332f79b16f21bf8dd9a2"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-12-19 19:24:51.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-20 19:09:01.000000000","tz":-480},"subject":"Middleware that allow for a user to quote Etag","message":"Middleware that allow for a user to quote Etag\n\nChange-Id: I380c6e34949d857158e11eb428b3eda9975d855d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d69ea690883d6b99058e80ecf8467eeccb16d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d69ea690883d6b99058e80ecf8467eeccb16d95"}]},"branch":"refs/heads/master"},"8ad5d4d3283733e56302ada6aa8a9dd2cf385cc1":{"kind":"REWORK","_number":3,"created":"2019-12-27 05:39:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/700056/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/700056/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/700056/3"}}},"commit":{"parents":[{"commit":"5e9d5d101a8103b437c4332f79b16f21bf8dd9a2","subject":"Merge \"Fix type on py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e9d5d101a8103b437c4332f79b16f21bf8dd9a2"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-12-19 19:24:51.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-27 05:25:46.000000000","tz":-480},"subject":"Middleware that allow for a user to quote Etag","message":"Middleware that allow for a user to quote Etag\n\nChange-Id: I380c6e34949d857158e11eb428b3eda9975d855d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ad5d4d3283733e56302ada6aa8a9dd2cf385cc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ad5d4d3283733e56302ada6aa8a9dd2cf385cc1"}]},"branch":"refs/heads/master"},"c217836e169bf1cc63998ae728c04cb123c5b7b1":{"kind":"REWORK","_number":4,"created":"2019-12-27 19:36:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/700056/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/700056/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/700056/4"}}},"commit":{"parents":[{"commit":"5e9d5d101a8103b437c4332f79b16f21bf8dd9a2","subject":"Merge \"Fix type on py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e9d5d101a8103b437c4332f79b16f21bf8dd9a2"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-12-19 19:24:51.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-27 19:27:06.000000000","tz":-480},"subject":"Middleware that allows a user to have quoted Etags","message":"Middleware that allows a user to have quoted Etags\n\nUsers have complained for a while that Swift\u0027s ETags don\u0027t match the\nexpected RFC formats. We\u0027ve resisted fixing this for just as long,\nworrying that the fix would break innumerable clients that expect the\nvalue to be a hex-encoded MD5 digest and *nothing else*.\n\nBut, users keep asking for it, and some consumers (including some CDNs)\nbreak if we *don\u0027t* have quoted etags -- so, let\u0027s make it an option.\n\nWith this middleware, Swift users can set metadata per-account or even\nper-container to explicitly request RFC compliant etags or not. Swift\noperators also have get an option to change the default behavior\ncluster-wide; it defaults to the old, non-compliant format.\n\nSee also:\n  - https://tools.ietf.org/html/rfc2616#section-3.11\n  - https://tools.ietf.org/html/rfc7232#section-2.3\n\nCloses-Bug: 1099087\nCloses-Bug: 1424614\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I380c6e34949d857158e11eb428b3eda9975d855d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c217836e169bf1cc63998ae728c04cb123c5b7b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c217836e169bf1cc63998ae728c04cb123c5b7b1"}]},"branch":"refs/heads/master"},"972f69381e4a2bbd66f1f3385ee0076e800d5758":{"kind":"REWORK","_number":5,"created":"2020-01-03 06:16:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/700056/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/700056/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/700056/5"}}},"commit":{"parents":[{"commit":"5e9d5d101a8103b437c4332f79b16f21bf8dd9a2","subject":"Merge \"Fix type on py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e9d5d101a8103b437c4332f79b16f21bf8dd9a2"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-12-19 19:24:51.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-03 06:15:34.000000000","tz":0},"subject":"Middleware that allows a user to have quoted Etags","message":"Middleware that allows a user to have quoted Etags\n\nUsers have complained for a while that Swift\u0027s ETags don\u0027t match the\nexpected RFC formats. We\u0027ve resisted fixing this for just as long,\nworrying that the fix would break innumerable clients that expect the\nvalue to be a hex-encoded MD5 digest and *nothing else*.\n\nBut, users keep asking for it, and some consumers (including some CDNs)\nbreak if we *don\u0027t* have quoted etags -- so, let\u0027s make it an option.\n\nWith this middleware, Swift users can set metadata per-account or even\nper-container to explicitly request RFC compliant etags or not. Swift\noperators also have get an option to change the default behavior\ncluster-wide; it defaults to the old, non-compliant format.\n\nSee also:\n  - https://tools.ietf.org/html/rfc2616#section-3.11\n  - https://tools.ietf.org/html/rfc7232#section-2.3\n\nCloses-Bug: 1099087\nCloses-Bug: 1424614\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I380c6e34949d857158e11eb428b3eda9975d855d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/972f69381e4a2bbd66f1f3385ee0076e800d5758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/972f69381e4a2bbd66f1f3385ee0076e800d5758"}]},"branch":"refs/heads/master"},"ba5f4a8ac75763725469dd471c4c9b85e54a14cc":{"kind":"REWORK","_number":6,"created":"2020-01-16 01:19:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/700056/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/700056/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/700056/6"}}},"commit":{"parents":[{"commit":"5e9d5d101a8103b437c4332f79b16f21bf8dd9a2","subject":"Merge \"Fix type on py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e9d5d101a8103b437c4332f79b16f21bf8dd9a2"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-12-19 19:24:51.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-16 01:19:05.000000000","tz":-480},"subject":"Middleware that allows a user to have quoted Etags","message":"Middleware that allows a user to have quoted Etags\n\nUsers have complained for a while that Swift\u0027s ETags don\u0027t match the\nexpected RFC formats. We\u0027ve resisted fixing this for just as long,\nworrying that the fix would break innumerable clients that expect the\nvalue to be a hex-encoded MD5 digest and *nothing else*.\n\nBut, users keep asking for it, and some consumers (including some CDNs)\nbreak if we *don\u0027t* have quoted etags -- so, let\u0027s make it an option.\n\nWith this middleware, Swift users can set metadata per-account or even\nper-container to explicitly request RFC compliant etags or not. Swift\noperators also have get an option to change the default behavior\ncluster-wide; it defaults to the old, non-compliant format.\n\nSee also:\n  - https://tools.ietf.org/html/rfc2616#section-3.11\n  - https://tools.ietf.org/html/rfc7232#section-2.3\n\nCloses-Bug: 1099087\nCloses-Bug: 1424614\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I380c6e34949d857158e11eb428b3eda9975d855d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba5f4a8ac75763725469dd471c4c9b85e54a14cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba5f4a8ac75763725469dd471c4c9b85e54a14cc"}]},"branch":"refs/heads/master"},"25d5123d25c623ad11ce04d45e7ca3efcada8ecb":{"kind":"REWORK","_number":7,"created":"2020-01-17 21:55:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/700056/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/700056/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/700056/7"}}},"commit":{"parents":[{"commit":"e92288bb61a59616608cacf1993e2d3fd9c4abff","subject":"Merge \"account-server: Add test for leading delimiter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e92288bb61a59616608cacf1993e2d3fd9c4abff"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-12-19 19:24:51.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-17 21:53:43.000000000","tz":-480},"subject":"Middleware that allows a user to have quoted Etags","message":"Middleware that allows a user to have quoted Etags\n\nUsers have complained for a while that Swift\u0027s ETags don\u0027t match the\nexpected RFC formats. We\u0027ve resisted fixing this for just as long,\nworrying that the fix would break innumerable clients that expect the\nvalue to be a hex-encoded MD5 digest and *nothing else*.\n\nBut, users keep asking for it, and some consumers (including some CDNs)\nbreak if we *don\u0027t* have quoted etags -- so, let\u0027s make it an option.\n\nWith this middleware, Swift users can set metadata per-account or even\nper-container to explicitly request RFC compliant etags or not. Swift\noperators also get an option to change the default behavior\ncluster-wide; it defaults to the old, non-compliant format.\n\nSee also:\n  - https://tools.ietf.org/html/rfc2616#section-3.11\n  - https://tools.ietf.org/html/rfc7232#section-2.3\n\nCloses-Bug: 1099087\nCloses-Bug: 1424614\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I380c6e34949d857158e11eb428b3eda9975d855d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25d5123d25c623ad11ce04d45e7ca3efcada8ecb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25d5123d25c623ad11ce04d45e7ca3efcada8ecb"}]},"branch":"refs/heads/master"},"27fd97cef927f39a057f46cf09a6c38ed52c6aa9":{"kind":"REWORK","_number":8,"created":"2020-01-27 20:53:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/700056/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/700056/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/700056/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/700056/8"}}},"commit":{"parents":[{"commit":"742835a6ec12944f606d6cd65ca7f18fdf78a09c","subject":"Merge \"New Object Versioning mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/742835a6ec12944f606d6cd65ca7f18fdf78a09c"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2019-12-19 19:24:51.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-27 20:53:35.000000000","tz":-480},"subject":"Middleware that allows a user to have quoted Etags","message":"Middleware that allows a user to have quoted Etags\n\nUsers have complained for a while that Swift\u0027s ETags don\u0027t match the\nexpected RFC formats. We\u0027ve resisted fixing this for just as long,\nworrying that the fix would break innumerable clients that expect the\nvalue to be a hex-encoded MD5 digest and *nothing else*.\n\nBut, users keep asking for it, and some consumers (including some CDNs)\nbreak if we *don\u0027t* have quoted etags -- so, let\u0027s make it an option.\n\nWith this middleware, Swift users can set metadata per-account or even\nper-container to explicitly request RFC compliant etags or not. Swift\noperators also get an option to change the default behavior\ncluster-wide; it defaults to the old, non-compliant format.\n\nSee also:\n  - https://tools.ietf.org/html/rfc2616#section-3.11\n  - https://tools.ietf.org/html/rfc7232#section-2.3\n\nCloses-Bug: 1099087\nCloses-Bug: 1424614\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I380c6e34949d857158e11eb428b3eda9975d855d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27fd97cef927f39a057f46cf09a6c38ed52c6aa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27fd97cef927f39a057f46cf09a6c38ed52c6aa9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
