)]}'
{"id":"openstack%2Fswift~700111","triplet_id":"openstack%2Fswift~master~I5b391664b9260c762bd122a92387084c0582a362","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I5b391664b9260c762bd122a92387084c0582a362","subject":"py3: Clean up some needlessly-noisy tracebacks","status":"MERGED","created":"2019-12-20 06:10:23.000000000","updated":"2019-12-20 21:19:45.000000000","submitted":"2019-12-20 21:18:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700111-1576876691944-d94f54cc","meta_rev_id":"fb142e7e239638e6abb42b70217da77da4daec6c","_number":700111,"virtual_id_number":700111,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 21:18:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-20 18:47:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-20 18:47:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 21:18:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4c561bb1a7e14560cbd892e98ec5d002ed07afd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-20 06:10:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2901fc6c327ac254a0f23e267906eb466db8fe34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 07:51:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5911ca7cc7f04d129650ba05d0d0ab64 : SUCCESS in 11m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/112ef5d606f142e0b4163371602ebea3 : SUCCESS in 56m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75a067bb15fc4b4abc8a6af263aa042b : SUCCESS in 59m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/82933d0063904607863dca466dce04db : SUCCESS in 10m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/83040a1d932e4320ab680421ff61bf0c : SUCCESS in 11m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/acc46981793b4d218d1daf996d8c6792 : SUCCESS in 12m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9502b4d39ae2491395cb0db021924db4 : SUCCESS in 17m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/07a33340ca3243d9bc77eeb668bc2f8c : SUCCESS in 19m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6ad6f069c7a44e399acba543c8587b93 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3d9dfea71f254f41b456ced7463f87ce : SUCCESS in 14m 37s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/6af6827555ce4419b2e95767dfd5b674 : SUCCESS in 16m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0f125c40d08d4e23a0fcb09ed82eb77b : SUCCESS in 25m 08s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/053ced6be0474734bde89a6b04adbd99 : SUCCESS in 13m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d637dd519dd54da584290998760755d5 : SUCCESS in 11m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7c54b10be70f45e2a12ade658ffb2be4 : SUCCESS in 11m 35s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/27ef12c4517b43d689430c432d37eb9f : SUCCESS in 11m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2eeb230e2b6148dab9bc275f79b42517 : SUCCESS in 13m 09s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8da0ed7556f74d3bac2b64154dc6866b : SUCCESS in 21m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b54c8cc7e7844b6b8672c618364fe133 : SUCCESS in 18m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d9a31ad715254765b5a37884dc8d6e1e : SUCCESS in 35m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/494b2c96163f405bb05c2723462c7b4b : SUCCESS in 48m 05s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/75f1c0d2644a4b9189146738b244d9a9 : SUCCESS in 46m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a9ec846120784f7ba6be0a3f8d3e3977 : SUCCESS in 44m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6d0765a47fd4a30a87c71016c277982 : SUCCESS in 15m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef1dd20c3982433ea886c154a3d48016 : SUCCESS in 15m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7904256872d34ff3912dd71efa0cac5d : SUCCESS in 26m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc75e104bf414a778c8a62a10f047787 : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e88a5d0e242a11d4cd5adc94108cddf00e11f55a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-20 18:47:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c594434e56ee5d06f62fd2ea264ad8816e2615d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 18:47:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07c3cd3f8929f00e712799c055314f8441de47dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 21:18:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c73327741e45048ad222c39570f403 : SUCCESS in 7m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9d691e1d75e942f98aaa0cf81badaf3b : SUCCESS in 51m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9405f07cce6442ada2d4b7716df8b197 : SUCCESS in 52m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/dee5bd5ae7124e9e92d9304dba74dd9d : SUCCESS in 9m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7480a00e8497430e93c887903447997c : SUCCESS in 10m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/32622944e9994749bf95cf3317bcb9f8 : SUCCESS in 11m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/07fe9e954d3d43f3971e702373e81191 : SUCCESS in 15m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a0f5057bcb934e97a361d151cf30bd91 : SUCCESS in 13m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/16940fac211f430081c84441234d9df1 : SUCCESS in 11m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e5e9371a293f4546b82c0e3211de8594 : SUCCESS in 10m 33s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/954a5890856141bfadbecc31eb5066a1 : SUCCESS in 11m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9eea16839bf442f3925c0af338422b66 : SUCCESS in 11m 27s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/1d2044e79d4f419f85d257eeb5661792 : SUCCESS in 15m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/293252fdb3f64f3b82cee8d00513f3d3 : SUCCESS in 10m 25s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3a74d8af79984d518e6a80985df315a5 : SUCCESS in 13m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/765ad07581014372973ec55c31e70f06 : SUCCESS in 12m 55s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/74340e5318604881ae1452e8cd3ad2be : SUCCESS in 16m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/19a972941e0246c9872dba46d5d85a11 : SUCCESS in 36m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d063100f92f4b9eb9fda402481b4640 : SUCCESS in 29m 55s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4bdc7431ac334c58ba235933739c6f31 : SUCCESS in 29m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b879c6c1654349f9bf18ea78f7432184 : SUCCESS in 28m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/778a6ca8c6db477293f2d49d1e1e5a96 : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a89161708aff4712a73e63e5f11b81de : SUCCESS in 5m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/458862e0fcda4e0ea1387176065b73d9 : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7053bfdc2ede2425b42fd5dd1673f8637b8ffb2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 21:18:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fb142e7e239638e6abb42b70217da77da4daec6c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 21:19:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ddb9c1792eb4c34a16e97ffcebab730 : SUCCESS in 1m 12s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/85211df752864ce0b38fe77e29b14f02 : SUCCESS in 53s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/14863e974d2840d19b089c0e4d6b5fd2 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54312b0b79ef2517f87a84a101507038d833104f","revisions":{"54312b0b79ef2517f87a84a101507038d833104f":{"kind":"REWORK","_number":1,"created":"2019-12-20 06:10:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/700111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/700111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/700111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/700111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/700111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/700111/1"}}},"commit":{"parents":[{"commit":"5e9d5d101a8103b437c4332f79b16f21bf8dd9a2","subject":"Merge \"Fix type on py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e9d5d101a8103b437c4332f79b16f21bf8dd9a2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-20 06:09:57.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-20 06:09:57.000000000","tz":-480},"subject":"py3: Clean up some needlessly-noisy tracebacks","message":"py3: Clean up some needlessly-noisy tracebacks\n\nChange-Id: I5b391664b9260c762bd122a92387084c0582a362\nRelated-Change: I5f3bc6416207cab2c7e3a77ee6689360b55990e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54312b0b79ef2517f87a84a101507038d833104f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54312b0b79ef2517f87a84a101507038d833104f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
