)]}'
{"id":"openstack%2Fswift~700454","triplet_id":"openstack%2Fswift~master~I5dac8a35be00f520171e5e179504e8f1c6361988","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I5dac8a35be00f520171e5e179504e8f1c6361988","subject":"WIP: sharding: Fix conainer_name typo in probe test","status":"ABANDONED","created":"2019-12-23 21:51:35.000000000","updated":"2019-12-26 22:57:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d9c4bfe18b76aea1ab3da50c45e795c216197bc5","_number":700454,"virtual_id_number":700454,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-24 01:39:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-12-24 00:29:02.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-23 23:12:51.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-12-24 01:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cbd24f352edc6f57bc03aa23c4ec599ebffbc120","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-23 21:51:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc36c81b9c451fed3aa2acc7f4d85f36bfcf1e57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 23:05:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33a7ddcbb5f74996800d83d11a64728e : SUCCESS in 8m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/eb41902cff75436b8d91acb854f3ec82 : SUCCESS in 1h 07m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d071c68c4e754811a1196cb4ec749cb7 : SUCCESS in 57m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/da9a47478b5e465d9ddf145e670582cc : SUCCESS in 8m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/14a14a3d2626483086e6970d85d52246 : SUCCESS in 8m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f6c7650b714045b989e0f09139c4aee9 : SUCCESS in 11m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8bcad4e070d54368a81bd8bbaa540c22 : SUCCESS in 11m 59s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b1345ea2e0674425a0d739429e8c0c8c : SUCCESS in 11m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/97763608439b4db2a87399aefbba53c2 : SUCCESS in 9m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b8d9e3ed2426449594744682c3073433 : SUCCESS in 13m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/6d8837dd8d0547df9314124ce65993dd : SUCCESS in 9m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b1ac5a7ddd47483697f4b2fb8037b36d : SUCCESS in 9m 41s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/8bdb927609b94239acd1a99eeb5001a1 : SUCCESS in 11m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/719d8469876a4d98930618ba3892916d : SUCCESS in 11m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/58aca1ccaf2443a8a072e7a72e9a0038 : SUCCESS in 10m 51s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/051aa7c424ea4edbbb584d9ae559e9e4 : SUCCESS in 10m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/557ca1d969aa4d21a3d26b8b57002811 : SUCCESS in 10m 45s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6e4a18a0d48244d6b920799955b92520 : SUCCESS in 21m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/91b6207d95b64081ba84711670c686d8 : SUCCESS in 16m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/19ed1b77bab941a4a95ccbe4672e1daf : FAILURE in 35m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0ecfe40ae73349a0a76fbb42ffd74f2d : SUCCESS in 28m 50s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/01695465722e4594943fb472922d02b8 : SUCCESS in 33m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f081841ba90a4e63b900731144c699cd : SUCCESS in 30m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0b61ceda9a74b8497df35d53d52ee1c : SUCCESS in 14m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53e56440c7ed4d7baad5af7901b24280 : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/599fcd8c021d4e439c8b991af9b822b1 : SUCCESS in 17m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4078919bf47f47cea12cb439faa584ff : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":1},{"id":"af90d44a606bf1b3fa981eabced1a62ef793a386","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-12-23 23:12:51.000000000","message":"Patch Set 1: Code-Review+2\n\nI\u0027m okay with this because it only adds one extra resp.read(), after REPLICATE, so users of the direct_client other than sharder aren\u0027t going to be affected. Sharder now decodes a body with UTF-8 while trying to report about problems, which I\u0027m not enthusiastic about, but oh well.","accounts_in_message":[],"_revision_number":1},{"id":"b8e6827f186eea7e40830238e11a970f728e43a8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-24 00:23:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e695f6b41a002a61605cc55794522d0f6caaa768","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-24 00:29:02.000000000","message":"Patch Set 2:\n\nSorry -- I should\u0027ve flagged it as WIP. I half expected the py2 probe test to fail, too; this\u0027ll probably get rolled into https://review.opendev.org/#/c/699892/ so the py3 tests can actually pass.","accounts_in_message":[],"_revision_number":2},{"id":"e195a3e574ada7143bc647dae95d94418aa08c45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 01:39:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d7240b7eb14c7f9e4af3bbc43722d9 : SUCCESS in 10m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d85b8b6e94814a4fa36fe6f8e15b41fc : FAILURE in 1h 11m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f6302c9404ce420496c2b2a4493eb962 : SUCCESS in 1h 06m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/55f8b11610cd44af8c94f13c96e5b42b : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6dc403d619bf4c8dbb2250b91a99a824 : SUCCESS in 11m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a279ee32af4f48968a96fd0f930d84df : SUCCESS in 23m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a6f7a69a0bae42f3b8735439770150d2 : SUCCESS in 22m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/74e5d4ed04834c1494fed5026d14bc56 : SUCCESS in 22m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f0ef7ae4c8bb41c2b116a57ad5e2bbc4 : SUCCESS in 20m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5610904b779544ee93d26051164e3706 : SUCCESS in 17m 52s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/e1ab53726d24497e8446e19557181755 : SUCCESS in 9m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/63ab4df229c7430b8f92d279b37a7e1f : SUCCESS in 9m 26s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3c4742154e2e49d8ae876f2dc678786f : SUCCESS in 12m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9da09cc02fa94203b7407b063e3ca871 : SUCCESS in 12m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/51d93551dddc4eceaa618ac9fe832385 : SUCCESS in 10m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/24390edc0ddb47f383d01af24140760a : SUCCESS in 17m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8da7e60a6d904c96a38ea0959bc8147e : SUCCESS in 12m 06s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2205e45138544fd4b11ba27710b97c88 : SUCCESS in 12m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b3d777eb63354907810f9ca0ddb3391a : SUCCESS in 20m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e19c13abdede4da284f44c830b2a1756 : SUCCESS in 33m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/346b52264ee24063bb8d5f550c4e41f5 : SUCCESS in 29m 12s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/166957c37340487ea47dd2fca703eea9 : SUCCESS in 31m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c5d64ee6d51b4e1aaca99ff1e9466c82 : SUCCESS in 27m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df540d56b0034ecf844a1a3ce94b5c01 : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce7b7d28105e4f39bf3cc03b9a82a581 : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/009be7734cd74b56a615574c2e20def3 : SUCCESS in 20m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31a547f5761b418c9a27fbbb81a9bad7 : SUCCESS in 1h 10m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d9c4bfe18b76aea1ab3da50c45e795c216197bc5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-12-26 22:57:33.000000000","message":"Abandoned\n\nGot the thrust of the fix into https://review.opendev.org/#/c/699892/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"403dabffedd47c4b749ff580100c87e42315d917","revisions":{"75c07a68c6f02e9e98947a4b6f268f56293eb64b":{"kind":"REWORK","_number":1,"created":"2019-12-23 21:51:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/700454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/700454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/700454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/700454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/700454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/700454/1"}}},"commit":{"parents":[{"commit":"ca4e2fdc135d946e7a332300537fadbee43d599e","subject":"Merge \"sharding: Tolerate blank limits when listing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca4e2fdc135d946e7a332300537fadbee43d599e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-23 21:51:14.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-23 21:51:14.000000000","tz":0},"subject":"sharding: Fix conainer_name typo in probe test","message":"sharding: Fix conainer_name typo in probe test\n\nTest now fails on at least py3; will see what the py2 gate job does...\n\nChange-Id: I5dac8a35be00f520171e5e179504e8f1c6361988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75c07a68c6f02e9e98947a4b6f268f56293eb64b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75c07a68c6f02e9e98947a4b6f268f56293eb64b"}]},"branch":"refs/heads/master"},"403dabffedd47c4b749ff580100c87e42315d917":{"kind":"REWORK","_number":2,"created":"2019-12-24 00:23:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/700454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/700454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/700454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/700454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/700454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/700454/2"}}},"commit":{"parents":[{"commit":"ca4e2fdc135d946e7a332300537fadbee43d599e","subject":"Merge \"sharding: Tolerate blank limits when listing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca4e2fdc135d946e7a332300537fadbee43d599e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-23 21:51:14.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-24 00:23:07.000000000","tz":0},"subject":"WIP: sharding: Fix conainer_name typo in probe test","message":"WIP: sharding: Fix conainer_name typo in probe test\n\nTest now fails on at least py3; will see what the py2 gate job does...\n\nChange-Id: I5dac8a35be00f520171e5e179504e8f1c6361988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/403dabffedd47c4b749ff580100c87e42315d917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/403dabffedd47c4b749ff580100c87e42315d917"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
