)]}'
{"id":"openstack%2Fswift~700818","triplet_id":"openstack%2Fswift~master~Ib8f7d08bc48da12be5671abe91a17ae2b49ecfee","project":"openstack/swift","branch":"master","topic":"auto-create-account-prefix","hashtags":[],"change_id":"Ib8f7d08bc48da12be5671abe91a17ae2b49ecfee","subject":"Deprecate per-service auto_create_account_prefix","status":"MERGED","created":"2019-12-31 19:51:39.000000000","updated":"2020-01-07 01:32:10.000000000","submitted":"2020-01-07 01:30:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700818-1578360620260-b2b5e395","meta_rev_id":"c43195d8310ea597af4b9624a36b22688992991a","_number":700818,"virtual_id_number":700818,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-06 09:08:34.000000000","_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"date":"2020-01-05 16:00:57.000000000","_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-07 01:30:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-05 20:37:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-05 20:37:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-05 20:37:01.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-01-06 09:08:34.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2020-01-07 01:30:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c997f7bf11eb8ac593fed4ddfb7d4bcfbb7327ab","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-12-31 19:51:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ad39da016e48d7501f5f6d5c6febf19002c0508","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-12-31 19:53:30.000000000","message":"Patch Set 1:\n\nIf we think this is plausibly a good idea, I\u0027ll rebase https://review.opendev.org/#/c/700449/ on top of it\n\nI think as an enhancement to this it might be cool to the warning becomes an ERROR (don\u0027t start service) if you\u0027ve already configured the setting in constraints.","accounts_in_message":[],"_revision_number":1},{"id":"bdfdefca21f94655edc58cdc19ac6f40af47dbf9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-12-31 19:57:45.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nYou know, honestly - I didn\u0027t actually try to configure my SAIO with a non-default auto_create_account_prefix :smh:\n\nI hate this option :rofl:","accounts_in_message":[],"_revision_number":1},{"id":"c8e02de5b81c2b08af80ce69c8242228a92e1e0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 21:16:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dbf3b4d7b7a4b2ca72822b05e9f51eb : SUCCESS in 7m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fa4f0d6df3c64809942b25749eabec92 : SUCCESS in 54m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e2e14c7bc12499f84549ba6957d85b0 : SUCCESS in 51m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bebc510885c743bb838b82372427ae50 : SUCCESS in 8m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/83bcf10faf934e47823ebbb75740e788 : SUCCESS in 8m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d593fd927e33485ba584e241def2d76c : SUCCESS in 21m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cf14c35df54d4f24b157b95fb208b1e3 : SUCCESS in 13m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c2d2795f1ce54965abae80ff7420156a : SUCCESS in 24m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/923afbb03b754867a3a29a1928940527 : SUCCESS in 19m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/21073e816acd44198227ca17441dc467 : SUCCESS in 12m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/2abe93ab4c334a19aaa33ab63e6351ae : SUCCESS in 20m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7dd4dd8164b4431a83c5f853bd0a3780 : SUCCESS in 22m 21s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/99261422383142779941b873b15c8dd0 : SUCCESS in 24m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ed97bd33bdd84ba999713901572417ea : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/af12ab7073ea4a76ad96256d12767e6d : SUCCESS in 11m 57s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e3a444359aec41508419eb98c9f8e934 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/02ed564b239e4da1a277798cc70a1fc0 : SUCCESS in 18m 10s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6f1a3ca15958485eb045d6463829a543 : SUCCESS in 18m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/38a2a06e98e646c5980fe66b1f3041f2 : TIMED_OUT in 31m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9b50cc1591b4402d8d7ef6c651bff1c8 : SUCCESS in 32m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2009b47c086c493c94ca0f6fad4b1a6c : SUCCESS in 28m 35s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/20cf904453584767b71afe36dba2d9ce : SUCCESS in 28m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b573daeda8e34d20b04c1043cf19f788 : SUCCESS in 26m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/669a4276e0764045bfabe3895a7a4ba7 : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3378de7d48b43b99133c609d9ba5ac9 : FAILURE in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/19325669979b43f6bda5ce282cf9e219 : SUCCESS in 17m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26d389c71a3849f2aca622091f42278a : SUCCESS in 1h 23m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c490f77e85935bbc766e56627538d5733dd88d2b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-01-02 19:12:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ff7c9f7dd3df25a8c2c48480eaba29614e7c41c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-02 19:49:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(10 comments)\n\nAre there any tests that actually exercise having the option in swift.conf?\n\nSeems like a solid plan -- blockers in my mind are:\n\n* updating proxy-server.conf-sample with the new disallowed section\n* making sure that AUTO_CREATE_ACCOUNT_PREFIX is a str (Oh jeez... or does it need to be a wsgi string? :barf: Punt on that question until later, for sure.)","accounts_in_message":[],"_revision_number":2},{"id":"365009ceedaad6da92957334beb79de3ab46bac9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-02 19:52:49.000000000","message":"Patch Set 2:\n\nOh, and we should probably update\n\n* doc/source/deployment_guide.rst\n* doc/manpages/swift.conf.5\n* doc/manpages/object-server.conf.5\n* doc/manpages/proxy-server.conf.5\n* doc/manpages/container-server.conf.5\n* doc/manpages/object-expirer.conf.5\n* doc/manpages/account-server.conf.5\n\nand drop the option from doc/saio/swift/object-expirer.conf","accounts_in_message":[],"_revision_number":2},{"id":"466a99cf302ff15e30db33ec0e8b0f48277bb968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 20:52:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77b2673af9c84799a58eda565aea5cba : SUCCESS in 7m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a4564926ebf741e8924f82875f8a30fd : SUCCESS in 1h 01m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86e9ab043d1141768c1b78a64335970a : SUCCESS in 1h 08m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b063328d63f4be0abb51c61b7be7843 : SUCCESS in 9m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b49f226595dd411dbaeff2c3ef539138 : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0acf61b4cb0348b2986a2e29557bd49d : SUCCESS in 14m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/598f060a97814c5cbc537cf47915d5eb : SUCCESS in 11m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/da17ec5b9b4d4d0ca7780a399a570523 : SUCCESS in 14m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3706caadaec54ab6b6adb6892fbe4751 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7f5d7e55b2d446d7a74dbd37b767de71 : SUCCESS in 12m 07s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/7988d724047e4093bb0f0c067ead076e : SUCCESS in 10m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/61cb3d28b80243b4ab0a7e64bb0a91d1 : SUCCESS in 11m 12s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/279dbd9e4f484e6e9f8b382f1ffb76a4 : FAILURE in 25m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/14834ae2da544b39a8260471a3ce3d06 : SUCCESS in 19m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d2544ab02acd411b9e4b6d631c01b136 : SUCCESS in 9m 56s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/87e8fa87a54142c8b67d0033b3b899de : SUCCESS in 13m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/599cbf4eae7b4f068789e27b428bede9 : SUCCESS in 13m 23s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d3e246c3ee844d0ca58abfb181174ffe : SUCCESS in 16m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e5db69af682d4252893a6606ada6a4a6 : TIMED_OUT in 31m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9e954aa9d95045838ee68cde1451f846 : FAILURE in 35m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1965ddda5d146a7bc60b674af75ca39 : SUCCESS in 28m 13s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c5ce535574fb4136b9115be37464a9b5 : SUCCESS in 28m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d88d87fabe2048b4877e7fa851bff774 : SUCCESS in 48m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4920527a60b047e880418a539d8a74bc : SUCCESS in 17m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f6491f79c84a43af61066e2ce93e26 : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f6f955a30050476b9782527272dd34b0 : SUCCESS in 24m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1e214baa36d4c53af0bdc9a2f7f088d : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":2},{"id":"662e4f3268251570d0938e4b8cd4938dd11513a6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-03 02:12:14.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nFWIW, after fixing up the types in constraints, I saw expirer and sharder probe tests pass while running with\n\n [swift-constraints]\n auto_create_account_prefix \u003d _\n\nReconciler test bombed out though; logs showed it was still trying to list /v1/.misplaced_objects, presumably because we *hardcoded it*: https://github.com/openstack/swift/blob/2.23.0/swift/container/reconciler.py#L34\n\nPresumably we could change that to say\n\n MISPLACED_OBJECTS_ACCOUNT \u003d constraints.AUTO_CREATE_ACCOUNT_PREFIX + \u0027misplaced_objects\u0027\n\nbut I\u0027m out of time to test that today.","accounts_in_message":[],"_revision_number":2},{"id":"e549cd74ac395820f06a0cde682424f1e9b90f3d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-03 05:28:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"48d599173edd438796b79261bd1e41a6292fc4bb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-03 05:36:03.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"41ee87e4ca7b1ed08d3c2807424730eb533fba2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 07:28:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce94221a9b64559b2ec5660e959af8d : SUCCESS in 11m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/be23d8d4fe5a41aa8e50b0912bdee451 : SUCCESS in 1h 03m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ccac477edbad4029b0a8de263ce50fb5 : SUCCESS in 53m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c19c20c084384ea687514902f745a20b : SUCCESS in 8m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/46e9e6a9a1f246e2a02520afbb70fa01 : SUCCESS in 12m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/00763a462e2b43348078881da98e2354 : SUCCESS in 16m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/95959444954242b0825f99a2ca9018fd : SUCCESS in 15m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/24204696fc0d4b1a966cc1214bf2b5e6 : SUCCESS in 19m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d7e0d132d4db44c1ad82e17dfeb323b4 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/77fc7cf54eee47218eb0d25b2e2d45c6 : SUCCESS in 11m 42s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/eff9059d36484bbb96cd2cfeb634a298 : SUCCESS in 12m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/924957ec182c4bc4829a6f5b86324e1c : SUCCESS in 13m 38s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f2135719baea418a977b0fa4fc8e82db : SUCCESS in 14m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b3cf7df954c844988261e38aa31328e0 : SUCCESS in 11m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a23fa15382b04a4e833fd94817ebcf6e : SUCCESS in 12m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/755b9de58cbb4b25b0efe55aeb127d6c : SUCCESS in 14m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5593877ac5ee4808bbe3b3cbe6195588 : SUCCESS in 14m 57s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/41b5bcb2dfb64355a0099dc7756a3f7b : SUCCESS in 16m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1592c8edd48540cd809bfed0f7d6b3a1 : SUCCESS in 19m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c624151997704eaea1f6554edf70ffbc : SUCCESS in 38m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9250cddc45ce456ea2db9b6de23e0983 : SUCCESS in 40m 50s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/563d5da71d074546b56339eac5945bf1 : SUCCESS in 35m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6d7a3f40adf41e5bba36fe37cf04b34 : SUCCESS in 42m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7a1c72ec59241b391e3244dda803932 : SUCCESS in 14m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fecd3cc47e24fb3ba18d05f7f4a0b77 : FAILURE in 8m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e8b1fb1861a492b8158dcd45331838b : SUCCESS in 18m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f71509bb726b4ed8b0d08a0a123c412f : SUCCESS in 1h 11m 50s","accounts_in_message":[],"_revision_number":3},{"id":"ec83e3b0b5993425ec1e180f9b970f64ffd46b99","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-03 18:45:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"45d772f6aef92d1e470d2e09ee1896bce51adb25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 19:57:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2a5f596c6a42e09c69655477e9a543 : SUCCESS in 8m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/56a9e0abfeaa496caea1cd96add1d2c1 : FAILURE in 1h 06m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20e212e6172b4db0b326334aa67c9958 : FAILURE in 42m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b41d216e6a4741319bb3553d8600e9d5 : SUCCESS in 10m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8184a5f6094947598fcb08cb3e0c4a49 : SUCCESS in 9m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/64d1a288dda141bda19646f4115f7716 : SUCCESS in 12m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cf8506b0a18540fe807d024bb6f9e41e : SUCCESS in 17m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0f6e016cf90a437696c9326807ba948a : SUCCESS in 17m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f6a5b0c652de48be8f40a023e17a01fa : SUCCESS in 9m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b9628f5614f645208112bd324c6053a4 : SUCCESS in 10m 07s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/7be073f2469848d0885c80559797b2f9 : SUCCESS in 10m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a6d120447de146d4a1e0b325e1056e48 : SUCCESS in 15m 40s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7a93c33969334061a2b0190efe1a1e9e : SUCCESS in 22m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c2a5f0f13d8a4a468b28b83d37776cdb : SUCCESS in 10m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/aac0f03b9093477c9f196ac12b41e78c : SUCCESS in 16m 02s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/86b0d131ef0840ee8851bfcf2793c799 : SUCCESS in 10m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2ec92d435af74c9a91161085096e9123 : SUCCESS in 11m 50s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f612cb7218af40adbef02de0593c097a : SUCCESS in 18m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/203e991898984659821bef4a9bc45a7e : SUCCESS in 17m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/92481155a76a4c4ca0ccadbc2055ea4f : FAILURE in 35m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ad3f97af03148c4b02e5c56a6775221 : SUCCESS in 29m 41s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0bf6b0cf491e4dc9be95975dfe3ae7be : SUCCESS in 30m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd955b8fd4b645d2a16a32cebb287b1d : SUCCESS in 35m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d83ee3efa93a4943b3a02e84c54b325a : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a223fad2c3914783829d63944867b24d : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e969bc23191044bfb522240acba66b76 : SUCCESS in 17m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/765731ae7b1a4c00bf4cf9721f040115 : SUCCESS in 1h 09m 40s","accounts_in_message":[],"_revision_number":4},{"id":"c1bf25b5fbe79fd884c18637d6f4024a4bf697c9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-01-03 21:43:39.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\ni\u0027m gunna respin this!","accounts_in_message":[],"_revision_number":4},{"id":"50c541c5f414411440b7a3c52c997b9b1707a8ce","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-01-03 23:16:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d2df6797800e21a0b75dbdc96021df5ca7fbb84d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-01-03 23:29:38.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nthis is gunna be great","accounts_in_message":[],"_revision_number":5},{"id":"e9a4744b9bef6a2a102fadddbc298141e11162cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 00:34:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/218bad768a08459792148a14fe299d48 : SUCCESS in 8m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e7e61530ef0b4cb28611215f1e0e5b27 : SUCCESS in 55m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ecaae22eb9d4ffbbea4aae12d7b8653 : SUCCESS in 57m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/07a9218aca14465799569cfe772873be : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/03c355c684da4146ba15349f03eebdee : SUCCESS in 10m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0862e436f47541f5a43503ff3eb13541 : SUCCESS in 15m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2461e85a998343d38c5f67c511d4288a : SUCCESS in 11m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3c481ba426ae41968c6b7ce137212ca4 : SUCCESS in 11m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/82bc43800b8148dfbbf06a43b78c8368 : SUCCESS in 13m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e8f8e05afdc1444abcc746e4813c7c88 : SUCCESS in 12m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4481ffc60af449c8adfceb8051e639f8 : SUCCESS in 12m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/28295dd54ffd48d8bbe03be4cda04bc7 : SUCCESS in 13m 39s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d47586dc947e425d8edcad8ef44320ac : SUCCESS in 22m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9cad80ffe3fc4e09bad119d1c2fb8d84 : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/60ecbd77738e49c39705b197a88ddc26 : SUCCESS in 9m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/60222ddc0c4f4d169bced29874f0c203 : SUCCESS in 9m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e88314182cbc4c88a59b29e39592476c : SUCCESS in 14m 01s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5a6d8b33ac60491ca10053de2481fa12 : SUCCESS in 16m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a94fc5bbd6ca4bc798a7f4885284c630 : SUCCESS in 20m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7731d5f4e8fa46a2ac7082ac41a2e7de : SUCCESS in 29m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/17118c59717e43d1bf5a9c77439d5241 : SUCCESS in 28m 43s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/904fe055406b42cfaa579d593d42db65 : SUCCESS in 28m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b87db24fe4d94ccf9939a4874634c990 : SUCCESS in 47m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7424353aa18e4544ae7f515580909516 : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2ebf34391d042dfb7df1de983731359 : SUCCESS in 7m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/255960fcadeb4249b90d3fb30587792f : SUCCESS in 17m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87c4f4b0a9f743e58c123063800a1a16 : SUCCESS in 1h 17m 04s","accounts_in_message":[],"_revision_number":5},{"id":"bd2b3c343253f410d37c435fb1c0517a491a0ff0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-04 00:35:30.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)\n\nI\u0027m\u0027a mull this over a little more before the +A","accounts_in_message":[],"_revision_number":5},{"id":"ec3f9150067ec6fed4c9fc351420f094afc7647d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-05 09:12:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ca5f42dbce771d5ba42a0eb6c40841fc227e247d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-01-05 14:13:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e8ce04f69319ee80a992028fe4f59eaeefe49730","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-01-05 15:53:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fa81fe63f58c916c0013979c504ec70eb208a5b8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-01-05 16:00:57.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"37979985c7b824f9da53a1767ebbf41e87e2dc5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-05 17:07:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a572a7956244aeba384a04036b9f68e : SUCCESS in 7m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0463e155f2e94223afd7c9be0549be9c : SUCCESS in 1h 03m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/860174949518400ab2ba138c6714e3e5 : SUCCESS in 50m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5b906b237a3145f1b9f932df8d607e54 : SUCCESS in 8m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/22b14227c6ce4c37beebf885da1df5fd : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fade2e0447054095b670232a980dddd9 : SUCCESS in 12m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/84f36818cb414a079cad7fe5114b25b8 : SUCCESS in 12m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/052031029cc7451894c55275ff3ce0e1 : SUCCESS in 12m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2a8afcd746944e329b123cca8ba0a698 : SUCCESS in 9m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8f91958f1cb943188505adfa186efc8c : SUCCESS in 10m 02s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/2ff9d7e5f0c74f4fa3ef7b269e75c643 : SUCCESS in 9m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ce74afb583964e9fb0293ee8f4b0baeb : SUCCESS in 10m 23s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/9c7a6f576cb5411eb7810df767f3a0d3 : SUCCESS in 12m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0d9421bc612248a2bd465fd6814ea843 : SUCCESS in 9m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/619ebedfb97546aaa8ad36e495f2c851 : SUCCESS in 9m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/010dc7d394c04adc98420054308204a5 : SUCCESS in 10m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ec3d436dc04448c889d1542252180432 : SUCCESS in 11m 29s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b98ee629bebf4950bc362f26f97686f9 : SUCCESS in 12m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/85e05e723ad749e98237fae98b2e35be : SUCCESS in 17m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b8f0e7ee328146939ecf4786890e43a8 : SUCCESS in 34m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/45ee5b060d0e45f4b8ec4c6556411ae9 : SUCCESS in 30m 21s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7eb40445bdef4a689f3d4daef0b9f1a1 : SUCCESS in 30m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f435828762074136a9f2bd38b7326de3 : SUCCESS in 30m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c2bc19f0c514bf4a02881e303983c0d : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a71014d9ec84da8a7559b1a5ee3931b : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9538470a2c7f45caaa018ad33aacd63c : SUCCESS in 18m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa9f7f79d1074a8bbe4c34e32868141b : SUCCESS in 1h 12m 05s","accounts_in_message":[],"_revision_number":6},{"id":"6274612f8d78ae09b314b6a003994b1b836748bf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-05 20:37:01.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"07e0c6d3bf62b20d10e50b7dca07f1a3f9494347","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-05 20:37:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"eedb238c7b5018de15d244ede43cb9429be33f1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-05 22:03:48.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aa13ca185f04892898bcbbe62f9d82f : SUCCESS in 8m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/83214db0843d462b8d5bdc049d692051 : FAILURE in 1h 25m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/94181db2845d4d3ca382bdbfeef80b91 : SUCCESS in 55m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/83f549d112be4b7fa1f94d2d2f93eb78 : SUCCESS in 9m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e31036f2d1c1469a9442875c4c499935 : SUCCESS in 9m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9db6ae319c5746fdbfa3b14215918d47 : SUCCESS in 15m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d6a21e5a91ed43429417b6b5a6c367a5 : SUCCESS in 21m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/abc27398affe4d84a095972ec2c38a73 : SUCCESS in 15m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2088a7fac0194d7b92c9b576b406aaae : SUCCESS in 12m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/993a24b0f56e498e8828068d96121ebf : SUCCESS in 12m 08s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/df7b2d9bc50b4f2294259835eed84c51 : SUCCESS in 12m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3fb3f04dd38d4b42ac51ecdc6a59c72e : SUCCESS in 20m 41s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/fcf274a28561487d803c17358635b0ab : SUCCESS in 15m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8e96625895d249ec83eaea4522565953 : SUCCESS in 12m 41s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/22b649cd7d2f4aeb8275013a7dd142c5 : SUCCESS in 20m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a4573d3549fb467fb58809b042d5020a : SUCCESS in 12m 59s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0efabca62ab84bd196caba37927a5ad8 : SUCCESS in 16m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/237d7fc927184db398987e434fcfd0fa : FAILURE in 1h 00m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/65384db171144d318298a0f20f1e96a9 : SUCCESS in 28m 48s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/724a6d7a278047c7b39462db6959569f : SUCCESS in 31m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/beae8e92cf8d4c6aa6c1b43a8d90396b : SUCCESS in 30m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34d86fa47a8d4e22836304cb89ea27e5 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872f3ade82a44268b14ad101f6978b8d : SUCCESS in 7m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a2b52eb3dc649fc83928a2c328e1322 : SUCCESS in 1h 07m 04s","accounts_in_message":[],"_revision_number":6},{"id":"5a5b87239af9037ca2984dd0f00e5c226264cd01","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-01-06 07:25:23.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"244d299024a67b95995ee9954e4d5a507b036ae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 08:43:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4b8be553f244aa4ade216f1d4c7bc81 : SUCCESS in 13m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/295c3f5ba78540b280c85b900f28aa5c : FAILURE in 1h 15m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/06281f5fdd1b41ae83576ba38ec30c45 : SUCCESS in 54m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/db50e8452942455aa757149f4a1127fe : SUCCESS in 10m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d6879db21e7f453ab8d136c31de6303c : SUCCESS in 10m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cadd608c22f042388518db085d7dd762 : SUCCESS in 11m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9fbc272200544a8f9e0cedfa8c80f392 : SUCCESS in 13m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bcd331c4a8a944ff8d71d0d0166fac6d : SUCCESS in 12m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b97f1d7293554c28b8d25353230f3c1e : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/adb6c629244c4298bfca572fe636f6f8 : SUCCESS in 10m 57s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5a06b47192ee4dc39b24d9b4046fd587 : SUCCESS in 10m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a4f1436d25a43e69be49c8fec7ef22a : SUCCESS in 11m 15s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d7e2840b1d0e4e708ef33f31335bf14a : SUCCESS in 12m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8a61b84a38d444db983a0ad24389cf63 : SUCCESS in 11m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/05a6921e7fc74019afeef2c38df0d634 : SUCCESS in 13m 07s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/85cfb977c21247d08f018ecad308b459 : SUCCESS in 9m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a01cd3cdca6b4add84b308e23bd08edd : SUCCESS in 11m 36s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/18c225f91e8b4be58998d612935910b8 : SUCCESS in 13m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9949a1f7ba22471ba0c868aaedcf82f8 : SUCCESS in 19m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2309790a46554731bdc5739d9e4cda89 : SUCCESS in 43m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/828db9cef7aa445f8d2ea63c26d5bf8e : SUCCESS in 35m 34s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e2c60535182c43378a230de1c437a191 : TIMED_OUT in 51m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/66d940f1aa2c423da7dc2c6e9ae0a267 : SUCCESS in 30m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c785fd33ee64643878b098fb42261cd : SUCCESS in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eb3c351e72f4da8839309431f5e7b35 : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21bfd11a0b8b4b4fab9b70698f455064 : SUCCESS in 17m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f1da26f8ca840769199f4d3d4962d9e : SUCCESS in 1h 08m 23s","accounts_in_message":[],"_revision_number":6},{"id":"023dbb09bd46ce60ccdf7e16e31cb64f52421ea8","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-01-06 09:08:34.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ff18d7d00cbee58fb5e2674158f637cf3f7648c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 10:20:21.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c336a5b4e584bb1a883bd98139569b1 : SUCCESS in 9m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8d339c738dbb4adbb23ceee8dc9ffa36 : SUCCESS in 59m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db9bbc8fb3204870b60cfafb1b6e7a9e : SUCCESS in 59m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/926ae5d67add4f91b74dff0cf9e3ea86 : SUCCESS in 11m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/012503a541ba480a8ed59a8e6130d72b : SUCCESS in 10m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ce2732694448472f9775e88bec588c65 : SUCCESS in 14m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6ecf5f58c2714e2d9a44abad6d1c2451 : SUCCESS in 17m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/60c03697bb1f472c9e2151ad07f87c72 : SUCCESS in 15m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/39d5b1c477074ad6b9e7d11df000e5c5 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/44758fc24e7c421093f27f69d95f5e79 : SUCCESS in 13m 43s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/919def39f6414aa0b4d9e888035179bb : SUCCESS in 15m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6a57bc9fb3a64dfabd9da500f451ed20 : SUCCESS in 10m 24s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3c89e60682a547769f67547fa753d96b : SUCCESS in 16m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/44e25a2cea67448f9b124a70d3c5db19 : SUCCESS in 13m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1c2fefcbb37147ad9d7dac48ef3bbcfa : SUCCESS in 16m 45s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7d4495483e44450985fccc1faf104870 : SUCCESS in 11m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/77f038f801b2446dbe2b3d6b74993a5b : SUCCESS in 14m 31s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/87f65555d03a42d8933eef41994897fb : SUCCESS in 17m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/04e05f03ea644ad983036d71f9a6fa4d : FAILURE in 21m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dbfd285c204a426da6fcd61af58621e4 : FAILURE in 49m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2c38cab81b74c3182fbe5e9d20bd5ab : SUCCESS in 35m 21s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/92b4101c252d4442a43deeacfa441272 : SUCCESS in 37m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1e28f09bf30a4673b37d34d27a09661a : SUCCESS in 30m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b28389b7ad314e81884c3cc97a21331c : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf10aa071b7c4027836fdf9e8ac24260 : SUCCESS in 6m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ecaeae30805e47c1a65c3e51d648e2db : SUCCESS in 23m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/100131f397304f55babb8eff3e5e4567 : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":6},{"id":"1095f50c83807ddb3a11e884e7695d972f796221","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-06 15:47:25.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"dfa33fdb39d0f4889a70e1b4a670c8b3a0497c9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 17:57:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055a5aecf7fe4343a6440d2f9360ceb8 : SUCCESS in 11m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8cd3490ed6a54a2fa5df8e6c931757a6 : SUCCESS in 1h 52m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c8b0a8c579c94b198ccf9e05f1ac9efc : SUCCESS in 59m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/76ccb2c60806413ba5dcdd6e65b6d0d9 : SUCCESS in 13m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9f2c6bccc20d4d0ea2fbf6b8c256f9ee : SUCCESS in 10m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/98b64c23109f4bef907a178534b16da4 : SUCCESS in 17m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/25e05ab49cfc46f58e701348310fe912 : SUCCESS in 20m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3093e971818c4f40acce72c82cdba6ff : SUCCESS in 17m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/02a71684801e44cbab97ff382dfbb0c2 : SUCCESS in 12m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/295232c538de4e418ebb063593a70694 : SUCCESS in 14m 32s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f15bf3803e3a4ac695facfd51b3f2d89 : SUCCESS in 14m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/18150bc8f9954847b0a30c227ec00b98 : SUCCESS in 15m 36s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/58a0ed148541406baf746a6653533869 : SUCCESS in 16m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a3f16aec7312498387b0a9e9558c2f3b : SUCCESS in 12m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e47b1e2aec8c488cb87d4667c2ffb7b9 : SUCCESS in 16m 31s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6adbad7df17a4cd88d68525a9745af99 : SUCCESS in 12m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5c1b0d1b56a24957a35a13cea9eef67a : SUCCESS in 13m 38s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4a0ac8dcb0db4df1a91ff7d77333c422 : SUCCESS in 16m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/acca914fdff5469d9f2d7876aa27d212 : TIMED_OUT in 31m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ca1b30dfd0bc45e0a349a58c0dcfc168 : SUCCESS in 47m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae67db8b80544f4f85888b6386ad65fa : SUCCESS in 32m 56s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/517011c6337e4e7c81baff5ca6eea0ef : SUCCESS in 35m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d51547e398484b419f19c8f068e19a74 : SUCCESS in 31m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7faa5da5b1504ce9a29c4b61ca55206e : SUCCESS in 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c01af64167fb410b80257d7dd909535d : SUCCESS in 7m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/25ae5577206a49c5926705d4ae6caaf1 : SUCCESS in 24m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4acb422d9058448896e2da0a00461580 : SUCCESS in 1h 06m 57s","accounts_in_message":[],"_revision_number":6},{"id":"cfa00064ebf3e918aadaee648341269174cb1536","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 17:57:53.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c97b46bd9942862dd19c0dc0e1994f08edfccd3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 20:01:36.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd4b15ed5e9340719651d54fc6bf2eb9 : SUCCESS in 10m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/187019b685d0476397311a1600568938 : SUCCESS in 1h 18m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/81be6f0ae5e9414a81ae54d26e4f6a82 : SUCCESS in 1h 15m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/733dbb811a614862bf8bfdd55d32e3e4 : SUCCESS in 9m 40s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3a527c3beb964cfbbfefc05228789dee : SUCCESS in 9m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a757f0027adc40b3852882bc86ad5a41 : SUCCESS in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8477cf217513452ba330b9949a464022 : SUCCESS in 17m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/51ed8392c41c4c0880185a8fdcf72afa : SUCCESS in 15m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bba6eac453784116ab8a32e953f8bc0a : SUCCESS in 12m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0d9b685345c46b9b8d9a285db93e9c7 : SUCCESS in 14m 32s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1094faee51d447e199f4916ed160e603 : SUCCESS in 11m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5b790f83f99d46e49bc1b2fcd2c2397d : SUCCESS in 10m 57s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3f5609238e654bfbbe7fb3a9e93e9dd2 : SUCCESS in 14m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5d2e04c18b2e44099ea5ce987a44865f : SUCCESS in 16m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c96a5efc8a244d0085969a545cf64b64 : FAILURE in 18m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/14aabbbdabab4116be1e6d4f9db816d6 : SUCCESS in 20m 39s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/40aac9a9f97d4c26abd96429e0133ef8 : SUCCESS in 19m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d9e91a348b0f4bc9a633ca0f9d3d84a4 : SUCCESS in 43m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3cb4bbd521fa46d3848e1baad5c51a4a : SUCCESS in 32m 32s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/985dee339ba74d678b786a6ff8600db0 : SUCCESS in 32m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/312f031e95324a17bd6551ef42011dad : SUCCESS in 31m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57357c57f618435da742f53b9fa0a6cf : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a3a526225bb41a4a6a3ee39df113a24 : SUCCESS in 8m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5183a70209904f59a3cb6f0c7589fbaa : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":6},{"id":"59d2575f6e2ef60b8feb110cd5304c49b732ce1e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-06 20:46:39.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"2edfa4a2b92c6cb76090767d1817f7c173d18e0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 21:57:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/badf8ff5ef5142f69b7c6a9598a9d185 : SUCCESS in 7m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4bc689471350432db873b30db52ce19e : SUCCESS in 55m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fef8d614e971497dbcf87e874bb043f1 : SUCCESS in 48m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e2db6ee67da849798cf9f3a5f1b0d1ee : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/331198a730554b9cbd8d29e1f843cf10 : SUCCESS in 9m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4a13cf5f5a474cceb9886bd74cefb7cb : SUCCESS in 15m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4f49895276fb42a2947789da899ebd34 : SUCCESS in 14m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cef4aa47f34441a88dd129144819b985 : SUCCESS in 15m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b110408d3f3541558557daaa82683187 : SUCCESS in 12m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3f2977c2149f4aefa6733e5b52837594 : SUCCESS in 12m 00s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/2552c56544a24be7b176e1a744dc440d : SUCCESS in 12m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/49a4d8dc6182407bbb463ca21beeecf9 : SUCCESS in 10m 01s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/076462bb43814d9b9b3a88e3afa868b7 : SUCCESS in 16m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c7c2335b3795480ba765cd316f2e19f1 : SUCCESS in 12m 26s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9a6061e58d124be6b74fd94df4f8b305 : SUCCESS in 13m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/836bd19a3d254d19b21f2a258c83779c : SUCCESS in 13m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7d9fffaa7a7d48f6a3a138bf8d051cfe : SUCCESS in 13m 11s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/422b09ef634546128a34053ec21b2a8a : SUCCESS in 15m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/59c9c689d72d4bda959a4eae007c769d : SUCCESS in 15m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f8caa2e798c2460792476d86f6f68057 : SUCCESS in 46m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f58b9ec259b1486f97bc864318deb54f : SUCCESS in 31m 12s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/774831a4ae09462ebba0045d0f3befa0 : SUCCESS in 29m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/734bd76c575c4a7cb7d6edf1645f0b0b : SUCCESS in 29m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09b405582a8e422f912dffa7b424835e : SUCCESS in 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10aa9ea9a48d42f8b90c53ce5b132bb3 : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8f69178220dc4ba7965dec185c0ec183 : SUCCESS in 17m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a21893f6e6fa428e92394ec54921878e : SUCCESS in 1h 02m 03s","accounts_in_message":[],"_revision_number":6},{"id":"4b4b2a2575893d08fce2977b81f1c0411f2903cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 21:58:02.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ff4360535562e8230d026038aa24e7fb64c35a8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 01:30:20.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/931950c234bd4078bea2de91508559c4 : SUCCESS in 10m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/646e56cdb9ce476d99031bdc68ec56cd : SUCCESS in 53m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0b918921cd0748b89bd07cd3a58b42f4 : SUCCESS in 55m 45s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/09f2fe38df5b4073b2e05f84f1a7de54 : SUCCESS in 10m 46s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e9c2ec05e52449129058817fced71321 : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/402ccce06749487cbbc09ba0e642e271 : SUCCESS in 14m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b3bb6e0f5790490d8a3e273035fab7e9 : SUCCESS in 15m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e2af220823cf48cfb9b747e3db33aca9 : SUCCESS in 16m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d563a03feddf4e459626bcd5d30b9854 : SUCCESS in 9m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69ac35249a9a4a13b6aedc14fdbfcf7e : SUCCESS in 9m 50s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/64231efe6ac24b3baa7e61fcba7e182c : SUCCESS in 12m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6aaf252ec8f3452daf7cd33a0902c407 : SUCCESS in 10m 22s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ad0b98f03e2e41389bbb899875c38c90 : SUCCESS in 12m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a0651cede7fe488cb42ff63ae5000399 : SUCCESS in 9m 56s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2eb87f8b0d9745dc9ba394d8f1a98e91 : SUCCESS in 9m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0c77184ba2fa4270a489bdf6c6e2ffd7 : SUCCESS in 13m 10s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/9225061d5a5f413eab5bb3999c9e4677 : SUCCESS in 13m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/354953b0401c499fbe2b30d97a1ecd8a : SUCCESS in 51m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e1066f01bc145658ad918e274ddf76e : SUCCESS in 30m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/6bc85b79e473481bb485ed3c86c95e66 : SUCCESS in 29m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/166091b0560a4e24a87d59d1dc62f4ea : SUCCESS in 31m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d074e26322a46788a4789cd05127782 : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c9f05adecc54030b42a7aa5723118bf : SUCCESS in 6m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49e7c7847af2462bae10466c5f6ee4f8 : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":6},{"id":"b08c22825af94e248ec1c6cc02d923fb4b8fbbff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 01:30:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"b5ff3d668f25a66df65de3f4faafd12adb5fd7b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 01:32:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1b7b11b6552486297aa5ea8aa109536 : SUCCESS in 1m 22s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/53d353897fdf4ee9a8c8ea0af7852e30 : SUCCESS in 1m 03s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f725d915bfaf449bac210eddaf9a934d : SUCCESS in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4601548dabdec0a4dc89cefba11e963217255be3","revisions":{"732a49983ea2e8662a4439abacdbdc491d726484":{"kind":"REWORK","_number":1,"created":"2019-12-31 19:51:39.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/18/700818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/700818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/700818/1"}}},"commit":{"parents":[{"commit":"7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc","subject":"Merge \"sharder: quote() more Swift paths when logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-12-31 19:48:13.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-12-31 19:48:13.000000000","tz":-360},"subject":"Deprecate per-service auto_create_account_prefix","message":"Deprecate per-service auto_create_account_prefix\n\nIf we move it to constraints it\u0027s more globally accessible in our code,\nbut more importantly it\u0027s more obvious to ops that everything breaks if\nyou try to mis-configure different values per-service.\n\nChange-Id: Ib8f7d08bc48da12be5671abe91a17ae2b49ecfee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/732a49983ea2e8662a4439abacdbdc491d726484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/732a49983ea2e8662a4439abacdbdc491d726484"}]},"branch":"refs/heads/master"},"fc1a7aaf33359f1c509699b270c3cb3c7de61d6e":{"kind":"REWORK","_number":2,"created":"2020-01-02 19:12:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/18/700818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/700818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/700818/2"}}},"commit":{"parents":[{"commit":"7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc","subject":"Merge \"sharder: quote() more Swift paths when logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-12-31 19:48:13.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-01-02 19:12:17.000000000","tz":-360},"subject":"Deprecate per-service auto_create_account_prefix","message":"Deprecate per-service auto_create_account_prefix\n\nIf we move it to constraints it\u0027s more globally accessible in our code,\nbut more importantly it\u0027s more obvious to ops that everything breaks if\nyou try to mis-configure different values per-service.\n\nChange-Id: Ib8f7d08bc48da12be5671abe91a17ae2b49ecfee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc1a7aaf33359f1c509699b270c3cb3c7de61d6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc1a7aaf33359f1c509699b270c3cb3c7de61d6e"}]},"branch":"refs/heads/master"},"19915311be039954c440c40a867d21988f7cf59b":{"kind":"REWORK","_number":3,"created":"2020-01-03 05:28:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/700818/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/700818/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/700818/3"}}},"commit":{"parents":[{"commit":"7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc","subject":"Merge \"sharder: quote() more Swift paths when logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-12-31 19:48:13.000000000","tz":-360},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-03 05:26:02.000000000","tz":-480},"subject":"Deprecate per-service auto_create_account_prefix","message":"Deprecate per-service auto_create_account_prefix\n\nIf we move it to constraints it\u0027s more globally accessible in our code,\nbut more importantly it\u0027s more obvious to ops that everything breaks if\nyou try to mis-configure different values per-service.\n\nChange-Id: Ib8f7d08bc48da12be5671abe91a17ae2b49ecfee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19915311be039954c440c40a867d21988f7cf59b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19915311be039954c440c40a867d21988f7cf59b"}]},"branch":"refs/heads/master"},"a29d778c4764a7f7855eaba8acd9add94c1b2761":{"kind":"REWORK","_number":4,"created":"2020-01-03 18:45:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/700818/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/700818/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/700818/4"}}},"commit":{"parents":[{"commit":"7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc","subject":"Merge \"sharder: quote() more Swift paths when logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-12-31 19:48:13.000000000","tz":-360},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-03 18:45:06.000000000","tz":-480},"subject":"Deprecate per-service auto_create_account_prefix","message":"Deprecate per-service auto_create_account_prefix\n\nIf we move it to constraints it\u0027s more globally accessible in our code,\nbut more importantly it\u0027s more obvious to ops that everything breaks if\nyou try to mis-configure different values per-service.\n\nChange-Id: Ib8f7d08bc48da12be5671abe91a17ae2b49ecfee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a29d778c4764a7f7855eaba8acd9add94c1b2761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a29d778c4764a7f7855eaba8acd9add94c1b2761"}]},"branch":"refs/heads/master"},"183a48e27bea12e1ce64c3f46132a731b51c90e6":{"kind":"REWORK","_number":5,"created":"2020-01-03 23:16:16.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/18/700818/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/700818/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/700818/5"}}},"commit":{"parents":[{"commit":"7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc","subject":"Merge \"sharder: quote() more Swift paths when logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-12-31 19:48:13.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-01-03 22:48:01.000000000","tz":-360},"subject":"Deprecate per-service auto_create_account_prefix","message":"Deprecate per-service auto_create_account_prefix\n\nIf we move it to constraints it\u0027s more globally accessible in our code,\nbut more importantly it\u0027s more obvious to ops that everything breaks if\nyou try to mis-configure different values per-service.\n\nChange-Id: Ib8f7d08bc48da12be5671abe91a17ae2b49ecfee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/183a48e27bea12e1ce64c3f46132a731b51c90e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/183a48e27bea12e1ce64c3f46132a731b51c90e6"}]},"branch":"refs/heads/master"},"4601548dabdec0a4dc89cefba11e963217255be3":{"kind":"REWORK","_number":6,"created":"2020-01-05 15:53:38.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/18/700818/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/700818/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/700818/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/700818/6"}}},"commit":{"parents":[{"commit":"7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc","subject":"Merge \"sharder: quote() more Swift paths when logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-12-31 19:48:13.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-01-05 15:53:30.000000000","tz":-360},"subject":"Deprecate per-service auto_create_account_prefix","message":"Deprecate per-service auto_create_account_prefix\n\nIf we move it to constraints it\u0027s more globally accessible in our code,\nbut more importantly it\u0027s more obvious to ops that everything breaks if\nyou try to mis-configure different values per-service.\n\nChange-Id: Ib8f7d08bc48da12be5671abe91a17ae2b49ecfee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4601548dabdec0a4dc89cefba11e963217255be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4601548dabdec0a4dc89cefba11e963217255be3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
