)]}'
{"id":"openstack%2Fswift~700965","triplet_id":"openstack%2Fswift~stable%2Ftrain~I0c850b8ae40d1ab477d0f5d18f92579c457da54e","project":"openstack/swift","branch":"stable/train","topic":"bug/1475018","hashtags":[],"change_id":"I0c850b8ae40d1ab477d0f5d18f92579c457da54e","subject":"account-server: Correctly handle containers starting with delimiter","status":"MERGED","created":"2020-01-03 00:27:12.000000000","updated":"2020-01-22 22:20:43.000000000","submitted":"2020-01-22 22:18:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700965-1579731509563-46a14a57","meta_rev_id":"cf16f6ac27e90822f0c70be35dc3248aa830d33f","_number":700965,"virtual_id_number":700965,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 22:18:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-22 20:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-22 20:28:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-22 22:18:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5102606d56b0a33da8cd247131a09d1f0ceaeb6a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-03 00:27:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd4e6b313db5f7e4a83c8230d7c3805beeb1d306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 02:04:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/243506d759e148a481a9dcb5ae30f62b : SUCCESS in 5m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/055de21b700c4b408deb14886b0bf446 : SUCCESS in 55m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c03bf16f15fd433a9d648c8d77066461 : SUCCESS in 51m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/322fe9c4aecb4f3c952b9d29b7fa18bd : SUCCESS in 9m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cf726a8dcc3e4d839d58d36ef69ec9f2 : SUCCESS in 15m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b986c0bea66343b992f2076b1a4c41f7 : SUCCESS in 15m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/31e900f70d274541980406ec8f9006c2 : SUCCESS in 15m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5a3f12fdd9454443ab4a454b1aed92d7 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2e26c27964ef4b7abe076cda43e328fd : SUCCESS in 13m 19s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b2154af2595045afa08691198349d9c1 : SUCCESS in 15m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5e9c124426564d17b9236777a37cc8e3 : SUCCESS in 13m 57s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/8a114f13545a4c56afdd5e73737215e0 : SUCCESS in 15m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fd0cd56943dc480aaeb79b3fce74f9b8 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c5ed9f70a9854c63ab13b34a0c0d6c2d : SUCCESS in 9m 50s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5d7be0acb8f94c95ac86fe57860e610c : SUCCESS in 16m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a3e24339f90841dab1afb0690fd344fc : SUCCESS in 10m 56s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5e208552aade4701b83c5a26491e9e66 : SUCCESS in 12m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a4d18f78749d4f9f8b8529e5e71ecbe1 : SUCCESS in 25m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8e5e1d78ad594714951607e661cfb4b1 : SUCCESS in 32m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4b3a895331e347059f28a03194ed5da3 : SUCCESS in 32m 57s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/282c761e703f4c8d8d449f86316619eb : SUCCESS in 30m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5d01d2693ebe4acda41720f9cbf67223 : SUCCESS in 30m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c945490bc58744039176e43e449d9a42 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6209b70a198f48109ebab0f78e8ff381 : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d6e03bd03524da4a6010734bc118371 : SUCCESS in 22m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/deed14a4cad747c6923776e1a09910c7 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2ae615d9b0afa5ab7f663cb733849895cf8a75c0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-22 20:28:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bcb2bbf3d2180aa589da158990e49843b251a845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 20:29:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97ef810d97a5e1d9f8e4de2dac1663ae3a61dbbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 22:18:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e0cef84fc74f4f8aa20b1149b4d8ea : SUCCESS in 8m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/534b7e4915b042fbaf9ac75f9720414b : SUCCESS in 1h 11m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7516580ae03147ef9c1f92a4f5538738 : SUCCESS in 58m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0677a226eea04f4196b8a568c764819e : SUCCESS in 9m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1c2f64f0647e478db46c8f88694617e2 : SUCCESS in 13m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2db54eff2dfd4661a55295bcd9093f16 : SUCCESS in 14m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/56a59ae1f1d34593aea11ec20f2df710 : SUCCESS in 13m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/068896ad1d8549abbe517a1382c4fb17 : SUCCESS in 11m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b5beeba559c44b7ba621703e48b5817e : SUCCESS in 13m 21s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/68bf218fdba54ed9a7a2a3d708cb903e : SUCCESS in 12m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/786b8bd592dd4e3f8022ad9009ddf1a8 : SUCCESS in 14m 31s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7db2848fef4d4a5cb32bb3db092b66d2 : SUCCESS in 15m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/72f8b31c644747c28cbf197798d3eb67 : SUCCESS in 12m 10s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/edf7c8a343a64a2c9b207713791e759e : SUCCESS in 13m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9b5e8e8ea0f44cdc81efac46385fc463 : SUCCESS in 13m 39s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b40fb4275fa1499d9f3deda06c6e132f : SUCCESS in 14m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/de43f683bf024776b3ef435bb3f25960 : SUCCESS in 35m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62cd761b3d324902a82af1bcb31dccce : SUCCESS in 30m 58s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/776094431f884ea395e396e8571b3081 : SUCCESS in 33m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89adaad3e6da406589c1d3ec12ef9b5c : SUCCESS in 35m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87c7a4548fd24c64a6da640e2799bd38 : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12a429499ef740f2bc27afafbfd2ac26 : SUCCESS in 7m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb1505ae5b844a7880219992dc5fc82b : SUCCESS in 1h 08m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4f0ba8ebd67b9caced0ce1b9173ed05c53a22ae6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 22:18:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cf85744235cf3d8f184a035353114a3af9cd5427","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 22:20:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b838145e4c384d3d93e06dee6fba2e85 : SUCCESS in 1m 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/852b987040294a638d7c3ae4a1e6dcb2 : SUCCESS in 1m 04s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1d0b706e4ac57b1fd4bbbce386e5d8dadf171b8","revisions":{"d1d0b706e4ac57b1fd4bbbce386e5d8dadf171b8":{"kind":"REWORK","_number":1,"created":"2020-01-03 00:27:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/65/700965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/700965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/700965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/700965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/700965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/700965/1"}}},"commit":{"parents":[{"commit":"8fded7ad2521f03a101bc5376cd968742dc4c1b9","subject":"Merge \"Fix BadResponseLength error when creating symlink\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8fded7ad2521f03a101bc5376cd968742dc4c1b9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-03 00:21:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-03 00:22:53.000000000","tz":-480},"subject":"account-server: Correctly handle containers starting with delimiter","message":"account-server: Correctly handle containers starting with delimiter\n\nRelated-Change: I27701a31bfa22842c272b7781738e8c546b82cbc\nRelated-Change: If912f71d8b0d03369680374e8233da85d8d38f85\nChange-Id: I0c850b8ae40d1ab477d0f5d18f92579c457da54e\nRelated-Bug: 1475018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1d0b706e4ac57b1fd4bbbce386e5d8dadf171b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1d0b706e4ac57b1fd4bbbce386e5d8dadf171b8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
