)]}'
{"id":"openstack%2Fswift~701069","triplet_id":"openstack%2Fswift~master~I7373bbb0a50d090d6becf141e5832f8ae98381a4","project":"openstack/swift","branch":"master","topic":"ring/hardloop","hashtags":[],"change_id":"I7373bbb0a50d090d6becf141e5832f8ae98381a4","subject":"Import directly from the inside of the ring module","status":"MERGED","created":"2020-01-04 01:14:54.000000000","updated":"2020-01-04 05:57:17.000000000","submitted":"2020-01-04 05:55:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701069-1578117338968-69f755c7","meta_rev_id":"ebe1e21009642befdd63d6198e897bb158a4c3bb","_number":701069,"virtual_id_number":701069,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-04 03:47:18.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-04 05:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-04 01:31:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-04 01:31:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-04 01:31:04.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-01-04 05:55:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97cc350aa5992464f35f53131207f22d13eb406c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-01-04 01:14:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99616ef26a8731e995f7e51aa64f123f22764e36","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-04 01:31:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI agree, this seems perfectly harmless. I wish I could repro your trouble though -- for me, it\u0027s slow, but it imports:\n\n (py37) $ time python -c \u0027from swift.common.ring import Ring\u0027\n \n real\t0m0.532s\n user\t0m0.471s\n sys\t0m0.059s\n\nDrilling down doesn\u0027t really help any:\n\n (py37) $ time python -c \u0027from swift.common.ring.ring import Ring\u0027\n \n real\t0m0.534s\n user\t0m0.480s\n sys\t0m0.052s\n\nIn fact, that slowness seems to affect swift code generally:\n\n (py37) $ time python -c \u0027from swift.common import utils\u0027\n \n real\t0m0.535s\n user\t0m0.470s\n sys\t0m0.062s\n\n...if a bit unevenly:\n\n (py37) $ time python -c \u0027from swift import common\u0027\n \n real\t0m0.296s\n user\t0m0.262s\n sys\t0m0.031s\n\nstdlib\u0027s *way* better, though, even when we gotta drill down:\n\n (py37) $ time python -c \u0027from urllib import parse\u0027\n \n real\t0m0.070s\n user\t0m0.061s\n sys\t0m0.008s\n\n*shrug*","accounts_in_message":[],"_revision_number":1},{"id":"f0c906104e8c94c34d524221848b64f67ec47a7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 02:24:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0c70e1455f14785a4771161105b264b : SUCCESS in 7m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/25f72d1b529c40b899b3e1ab02a08b77 : SUCCESS in 54m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1232b744b3cc416d9c06cf2f3b66a535 : SUCCESS in 58m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e02263835bc04120ab0d86c1c556bfc6 : SUCCESS in 8m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/09879ab517d14e4fbf7450b8b93b8750 : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/039016c0a15247dcbfe15ae83173639f : SUCCESS in 11m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29223dac3c6b4d979c431ffec4e7ad21 : SUCCESS in 12m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/be4f71d9b75948aeb3d01acc1a051300 : SUCCESS in 11m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/03ac000eb420405d878955d16de6cb5e : SUCCESS in 9m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5d4ca1d567034c80b1830df7c6c858cb : SUCCESS in 9m 56s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/413e826d32c846888e6ef0d43c867ea9 : SUCCESS in 9m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/93330bec15644ded9190490419aba531 : SUCCESS in 9m 41s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/9c06b8570da0473a9e5673293efbacd4 : SUCCESS in 12m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/094c018c174e4ed2a37c5569d7635072 : SUCCESS in 9m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b655359c1105458482bfc3b8cd2953ce : SUCCESS in 9m 38s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fe689f55b86845fb91bcfa342c2c7cfb : SUCCESS in 10m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3ae2b8b0219f4939a678644d352f008f : SUCCESS in 10m 17s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1f4be912e56d4b2099d507b633fc514a : SUCCESS in 21m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd26e95ef0c04a18a92ef2a9c6d66b48 : SUCCESS in 15m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cfee8f3c57134cc695e56339a4df87a9 : FAILURE in 30m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4503c0b7d35440d7ba6214eb9c2dd943 : SUCCESS in 33m 14s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/457f9d384edb4e66916df53f5dbb17a5 : SUCCESS in 29m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/778c04ec0ebd46139271597e34fe1090 : SUCCESS in 30m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40e2a658083f48b783ab094530ab9b50 : SUCCESS in 14m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db88e488e2b3461595782e1f5ec6466a : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aec8d09edc1a4a46b1a53d7d76791920 : SUCCESS in 27m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46c4d4e0fb124bbbbb9e566b0f375f09 : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":1},{"id":"cec512809ea62999eb838fde95968b990309a074","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-01-04 03:47:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dbb4a76d62dff70ead632421f148b30b6fe09db5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 04:47:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71afc19c1e494ca69c4e030b4212de37 : SUCCESS in 9m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f1cdf8b7129d4fd1ba4d8440c7c00ba8 : SUCCESS in 52m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c1b0d5dbe29b4dc48ba615edf5392737 : SUCCESS in 59m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f05ebfefd1d6469d8df6fcd77a2c924f : SUCCESS in 10m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/26a04773739d4c2397b1095576a8a22d : SUCCESS in 10m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/66864d0c149a4b7ba427b87a1f6e7cea : SUCCESS in 16m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/030069adca88487fb9e5bafec813198e : SUCCESS in 16m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b91c448fc3de431e8c40228132d0dabd : SUCCESS in 10m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3090dd40206c4ebf91b4306a4d3e766a : SUCCESS in 11m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4fe62300f6244959925390890da13ea8 : SUCCESS in 15m 39s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/333677cb767c4a3f9a8d50c4b0868a37 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f4de3a6f39e54a608849ab76c5394bf1 : SUCCESS in 11m 26s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/9fbe9f6964c24b8ca7b40c8ab9ddc825 : SUCCESS in 14m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bd9e8be4018c4f2899668ddf12127bc6 : SUCCESS in 9m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0a92a8f7b4ac47a9a78a078269d54fa3 : SUCCESS in 11m 44s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/935a3fd51bf2440790636c6d11432b41 : SUCCESS in 9m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8a341ed944bd4687b74affd152fd94e2 : SUCCESS in 10m 19s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/24b59a2b09554821b6d5c17a596f741c : SUCCESS in 12m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d4a14f3815ec4109bda77a2e5b69122c : SUCCESS in 26m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f21b32b7386849bc8058956f7edaedd5 : SUCCESS in 31m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0578afd19b94474eb57a9781a27f9b59 : SUCCESS in 27m 59s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/015eeefe9d5f44b7b2fd1cf604c97556 : SUCCESS in 30m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/59fb5f699f244ab186da9a3cd30d3049 : SUCCESS in 30m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f21e2b7b2b44f43a21777083b5e6be7 : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/281326815ff7497ca94b46253478b23e : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/802a1de37f3b4b89b12a7239e553a6f9 : SUCCESS in 17m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca8964f3ba18406c8ce1117616f79cf9 : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":1},{"id":"525be52e2e1ab4da526a0d17c74f64044b969e77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 04:47:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39b722d2bf0a2f96b1f4f821b92b0ef08a27ab39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 05:55:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28638fbbb3254deeba506a790c20e045 : SUCCESS in 6m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fd9dc1d7555c46dba3062c6636c3d639 : SUCCESS in 58m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b1dca69df2ca42b8ab30fae8bbe9c4c8 : SUCCESS in 49m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/094dbc6c5b774b46b87e53b5415e956e : POST_FAILURE in 12m 09s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ab8442fb222e46638d51499a0eac82b1 : SUCCESS in 9m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fb535d11adb84d4298d760e9111eb095 : SUCCESS in 11m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/633e41101f134804b68caa2f8b3a75e2 : SUCCESS in 10m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5ebd1d79c83c4fcf8840e83698bb72ff : SUCCESS in 11m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/230a3997ca2a4d4cad4f1a8d1a923472 : SUCCESS in 8m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/489c298580964a39853727900a70abbf : SUCCESS in 9m 46s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/9f4b32aaac28433f81ac5236f820409e : SUCCESS in 9m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b720f8424b8d425db60b7d19579cc551 : SUCCESS in 9m 48s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d48cb7ec71fc4c23b664067a1236fe00 : SUCCESS in 12m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/77c74efaf29b47a280da718f24527687 : SUCCESS in 10m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3ca9a03f10da4ba6ba7e456107f44a79 : SUCCESS in 9m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/318edd65f97842b882487da7569265fa : SUCCESS in 10m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b82c30c9f7294667a1d3e599e78bc1c4 : SUCCESS in 13m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ced0a9e830c94d21876e4fd0de4bb800 : SUCCESS in 31m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f13b97b6b373473ea67f62371b404756 : SUCCESS in 28m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/795c2776aa68472ca80b0b21b979b3c2 : SUCCESS in 27m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4868e56d5bcd460088cea4045d5ecf88 : SUCCESS in 28m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b32913c7c1a4e5d971dbb2d37ab8d54 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ac24e7f328a4c35afe45d328d8795b8 : SUCCESS in 5m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/078a53e9750a44c2b9755be8a836fd38 : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0ad3a7f2becc7471858db9f68f4f4c5142d1d098","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 05:55:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"de6cc536f15737f62cf32ba96e672a37e775535e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 05:57:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82d19234be804383a88afc161cae0ac1 : SUCCESS in 1m 17s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5fcc961037534e3d8446becea0c096f5 : FAILURE in 46s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0c11bd611b7544d3909fd5d7bf3cab68 : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f13db30b6425b9c973b4add984e22ae56ce84c2","revisions":{"0f13db30b6425b9c973b4add984e22ae56ce84c2":{"kind":"REWORK","_number":1,"created":"2020-01-04 01:14:54.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/69/701069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/701069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/701069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/701069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/701069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/701069/1"}}},"commit":{"parents":[{"commit":"f73a1908378ecb667f152c3d58f14ac986af9b52","subject":"Merge \"Use less responses from handoffs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f73a1908378ecb667f152c3d58f14ac986af9b52"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2020-01-04 00:34:11.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2020-01-04 00:34:11.000000000","tz":-360},"subject":"Import directly from the inside of the ring module","message":"Import directly from the inside of the ring module\n\nAn odd thing happened: when my code did only\n\"from swift.common.ring import Ring\", and nothing else,\nthe interpreter looped on the CPU.\n\nThe import from the top of the ring module is our standard\nprotocol for Ring. It causes no problem in places like\ncontainer updater or list_endpoints.py. It is a big\nmystery why it causes Python 3.7.3 to loop, and only\nin specific circumstances.\n\nBut we noticed that the recursive import is unnecesasry\nin this case, so an obviously harmless fix exists.\n\nChange-Id: I7373bbb0a50d090d6becf141e5832f8ae98381a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f13db30b6425b9c973b4add984e22ae56ce84c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f13db30b6425b9c973b4add984e22ae56ce84c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
