)]}'
{"id":"openstack%2Fswift~701127","triplet_id":"openstack%2Fswift~master~Id3d84281f15815b4185c76874575e91a3589981b","project":"openstack/swift","branch":"master","topic":"duplicated","hashtags":[],"change_id":"Id3d84281f15815b4185c76874575e91a3589981b","subject":"Fix  the duplicated words issue like \"the the \"","status":"MERGED","created":"2020-01-06 03:02:21.000000000","updated":"2020-01-06 19:52:21.000000000","submitted":"2020-01-06 19:50:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701127-1578340235030-bbef5f2b","meta_rev_id":"5ba099884144d1b30adc274ff6a135d02dddcc8c","_number":701127,"virtual_id_number":701127,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 19:50:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-06 08:41:35.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":2,"date":"2020-01-06 15:51:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":1,"date":"2020-01-06 15:51:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 15:51:59.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-01-06 19:50:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3756920b486b4850661856147341fc1aa1a24795","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-01-06 03:02:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c005513ba693c8e4d6d12ac8b5e74b4e5863fffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 04:10:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af3c0984ada84bf989c03280b731f1e2 : SUCCESS in 7m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/da855b848ffb46ac85762e10b76965f0 : SUCCESS in 55m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a1a0df3df2814cd0b36c49c5bd756cf6 : SUCCESS in 52m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b27eef85bef5422fa11a8b3e79b808a6 : SUCCESS in 8m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b29c08d35559428d900aaf17fe396a7c : SUCCESS in 8m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e375c79f4da4605a31a129813f57b9c : SUCCESS in 13m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3011b2e41bc34e2aa6c046cfab3b12be : SUCCESS in 11m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c4b78dfb8339466580101e56910a29ba : SUCCESS in 12m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0b86bd5d305340bbbd8ebcc1a342bf19 : SUCCESS in 9m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ff879ba6984a434783e8cc1039dff112 : SUCCESS in 9m 28s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d6a77b7582464146873bcb2f5ff28db5 : SUCCESS in 9m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/414322cd5e5d4351a0aaa8b0e23d8108 : SUCCESS in 9m 45s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6d8103477e0843ea957edcb5f42331b9 : SUCCESS in 14m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0217819637704b039d9f4a58df0b5f8a : SUCCESS in 9m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f9087dc6b39c4c9294ab5e1ca649ebab : SUCCESS in 9m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1c074c484272447188567278d3c35eed : SUCCESS in 11m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5560e6cb19eb48ba827f0d2655bb1c36 : SUCCESS in 9m 53s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c24140d9121946ceac239db07bc959af : SUCCESS in 22m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/065eb1ea5fda4bab8a100def7591ea9f : SUCCESS in 18m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b3472cd66e2c43238ab854ee67eeced6 : TIMED_OUT in 1h 01m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9ba77f425af460e97b716ce17e7e6b0 : SUCCESS in 28m 27s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/af9b037bb1f44faa9733f9ac870633b2 : SUCCESS in 34m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9aefb8642be04ffc97eee3a80eb14ec3 : SUCCESS in 28m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/552ec3df84324878803d9a7e2a6db247 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1375346db53a4fd880362e221d05b3eb : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b9fd3e5322d049db954cc267d8d3937f : SUCCESS in 22m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a9edfe96ddb4d3ca1f0f9ba1de533b4 : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7acd4a0de59ee094dd3dd37dfbd95996c3ba7cc5","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-01-06 05:59:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7852d38cea9d9916320c757e6fe7966f4cff1840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 07:24:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce6d7921c4de4727b9bef5c8fc93f89f : SUCCESS in 10m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4c29e8e46e8242869cea288ba66ab17e : SUCCESS in 1h 01m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c19cf6bf17a34c7b9e2c787d0f033722 : FAILURE in 52m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2f8672b364554d5c9eb1543f4e404551 : SUCCESS in 11m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0149a5f22cf446a487068a85b87a9987 : SUCCESS in 11m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f704916411c64aef823c303b2bc18b0f : SUCCESS in 18m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a8947c67be3040129a49f06d44706353 : SUCCESS in 17m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/39220209603b4fcd96219e4b28cead02 : SUCCESS in 18m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fecb22ca32884be39ebe3899e1e9dd30 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a41cd2d7a0b740f789bc77d59b9dc272 : SUCCESS in 18m 11s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ef89e188b29d4efe87a60c4ad01218ea : SUCCESS in 14m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f7f0f1db2c7144d0aaa21173023ed920 : SUCCESS in 16m 10s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/391d5bf9f2de460f9a4f112263035c2b : FAILURE in 20m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/06aa0dc4c4fc455890ccd532708f2f51 : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e8a1329948894d4ebadd317d3f4ebb00 : SUCCESS in 15m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/aa21599539b246d9bb7399f086e33136 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d33f5bf0f5f94040af1aed0de338e42b : SUCCESS in 15m 43s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a497c18f7b714fe7acf9073a41b46332 : SUCCESS in 18m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fa4845d7151c40d1b04ff29a5fefa7e0 : SUCCESS in 25m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f57516712c294cd2acd0aecd6575a78a : SUCCESS in 45m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee0cbf1406fe482c9442aa51ff50ee7d : SUCCESS in 33m 07s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5fbe10fb9f6341a08ddd5819e344a925 : SUCCESS in 33m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/201d0b67c5f84eb1973789dd6d855e7c : SUCCESS in 33m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffb2e11548e843faa5d4e6c6fb9db89f : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02af8957bc244544be807bbd8e5db057 : SUCCESS in 7m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5113ff81f25840fbad3a161cc4456dd7 : SUCCESS in 24m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d77d456551aa47db9c30ccf24fc04ede : SUCCESS in 1h 05m 38s","accounts_in_message":[],"_revision_number":1},{"id":"397958e14ea8395bc57f2beb56636467da47a61d","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-01-06 08:41:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f7b1481b42aa213ef7ef1ff433260f26e5335f5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 09:49:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/616b7cf2524541858cf5a62144338e6e : SUCCESS in 7m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8c09b98d39b4432b914c569a00e0fadc : SUCCESS in 1h 06m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bec711c09c814e9b8a8ad20e90bc93d7 : SUCCESS in 1h 03m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c5846d6cef74bbba51a40f7136eedc9 : SUCCESS in 8m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/960d4decd1f347c3a474148b73bb3e39 : SUCCESS in 9m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/89a0a08ba33c40f2b5a584a117996a38 : SUCCESS in 11m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a9d073abaf940b08df5342e989a343f : SUCCESS in 11m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9f5a9d5beb0a4992b18752e3ae046461 : SUCCESS in 12m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b5ebdf15959147e9b6648df8b40d7d47 : SUCCESS in 8m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/71a2db4e86ee400e88448bdb96413aa5 : SUCCESS in 9m 54s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/70a6d74ecaa84bf4a23d0c49942bdb7e : SUCCESS in 9m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4edf55d8df134aa390a4bde054254fed : SUCCESS in 9m 34s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0412431f8fa94fb5a4862899540fc326 : SUCCESS in 13m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c1e3f69734574308959e234642d017e1 : SUCCESS in 9m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2e83faeb32674c27a8fd2f5cba26418e : SUCCESS in 9m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6b6576cbf1c44003b97bd51db2fe472a : SUCCESS in 9m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/91d517a48b30448cb2bb133320b0652a : SUCCESS in 10m 21s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3cda0684f849475db260f774b589478d : SUCCESS in 12m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9392606173064c92897bb9269b02b679 : SUCCESS in 16m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/69c0d15dde464c0faa65f6416f066d2d : SUCCESS in 41m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6bf6da908e574c8e85277dced123e833 : SUCCESS in 29m 50s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/00f9969e67ba4f5eb7cd053d7e412b02 : SUCCESS in 27m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/124da54e33404703be42fde619005e54 : SUCCESS in 28m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e23696296764bf0aca345074fada85d : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f435e6269534804b3893e57566154ff : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/82b005dbc91a4320b9ec51dd1a4f6b95 : SUCCESS in 17m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d55dc6930674def82797ab6b48fee32 : SUCCESS in 1h 00m 55s","accounts_in_message":[],"_revision_number":1},{"id":"31b597e640d8f61257f4ac438c3c0e303f48b5cf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-06 15:51:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThanks! Good catches.","accounts_in_message":[],"_revision_number":1},{"id":"b7d57a161a9ae0cedc853ceba602a5a3bc38a5f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 15:56:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c3a4139dc8b47a94df94e15d91832c5864bf571","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 19:50:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/756cf81633e74191b440c802bbc06185 : SUCCESS in 11m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5791db08dba2450998044081b26b3104 : SUCCESS in 1h 12m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/04835760b08a477ca21db4221853dc1c : SUCCESS in 1h 07m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4ecc5e778f2343f79e5b5655120bea7e : SUCCESS in 12m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ff0c1026fba84c0fa2cc2c6eb547bfba : POST_FAILURE in 16m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/45f940792c6249589120f5fd3571f835 : SUCCESS in 15m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2cf3355cbb72422cbb56b70778150886 : SUCCESS in 14m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ce1e85332fea49ddac33f96893d346d5 : SUCCESS in 15m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/07bcb877b5ee4faaa41ba063b4fce6ac : SUCCESS in 12m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2934558d019b4d10a5987349cb243b0f : SUCCESS in 13m 44s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ce7d051111c648b8ac93f0ba15bdd2f1 : SUCCESS in 13m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3c4b2067eb3e465f95cc7eee43841a4c : SUCCESS in 13m 45s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/8c4a1ef525f24a8494b116c2e33c9ab3 : SUCCESS in 16m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0acd1366f0a44e08931a8e74df921d64 : SUCCESS in 12m 12s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/33011827f3eb4722ab6fba9d0530bbe3 : SUCCESS in 13m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/39d638b168964ec787c6b0d972fb8321 : SUCCESS in 13m 33s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c348812e2d5a460082d9c612d92c25b7 : SUCCESS in 15m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/69ab25a6ef4447a18d0506401df0ca3d : SUCCESS in 42m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/51affd15b9f74a5ebaf2a2917aab30fa : SUCCESS in 32m 44s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d7850153528b4cc69355d502dde6bf1a : SUCCESS in 34m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fe85fb68461041a4a0281623e7b23f94 : SUCCESS in 32m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2001087cbaf4c68912435389b804aed : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e324e51261438a8b69785a68e6fba5 : SUCCESS in 8m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8f49db95dea4c229272042e56d6239f : SUCCESS in 1h 15m 33s","accounts_in_message":[],"_revision_number":1},{"id":"595194e715db3e54244ceb1c8fbe3f361381219f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 19:50:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4c5f0b2c1e4b561cda8a2852aa8b87a8a35ab0ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 19:52:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d21a0da70a234d7db6353dd21755e0fb : SUCCESS in 1m 18s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/472871eb62164b8397e1a40c4f102cb5 : SUCCESS in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d7f089b4f4a24a45a860cc751276fb2b : SUCCESS in 58s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47eb78897d60ee60a4e42953f5efc52a46d6cd18","revisions":{"47eb78897d60ee60a4e42953f5efc52a46d6cd18":{"kind":"REWORK","_number":1,"created":"2020-01-06 03:02:21.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/27/701127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/701127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/701127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/701127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/701127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/701127/1"}}},"commit":{"parents":[{"commit":"7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc","subject":"Merge \"sharder: quote() more Swift paths when logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7862ec7b8a7c56fffc32e0bfac3665f8b4f198dc"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-01-06 02:34:42.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-01-06 02:34:42.000000000","tz":480},"subject":"Fix  the duplicated words issue like \"the the \"","message":"Fix  the duplicated words issue like \"the the \"\n\nThis is to fix the duplicated words issue, like:\n  * \"both of which which must be stripped off\"\n  * \"In addition the the values set\"\n  and so on\n\nChange-Id: Id3d84281f15815b4185c76874575e91a3589981b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47eb78897d60ee60a4e42953f5efc52a46d6cd18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47eb78897d60ee60a4e42953f5efc52a46d6cd18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
