)]}'
{"id":"openstack%2Fswift~701139","triplet_id":"openstack%2Fswift~master~I33b21412287d1ca2d368393c89b5c910b9521dcc","project":"openstack/swift","branch":"master","topic":"dark_data_04","hashtags":[],"change_id":"I33b21412287d1ca2d368393c89b5c910b9521dcc","subject":"WIP: Add a reference implementation of audit watchers","status":"ABANDONED","created":"2020-01-06 06:03:07.000000000","updated":"2020-12-26 16:50:55.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"526ce277d720ccaacdddfbff4765944018e553dd","_number":701139,"virtual_id_number":701139,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-01-16 17:02:01.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-16 18:28:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 16:26:26.000000000","updated_by":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"reviewer":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"state":"REVIEWER"},{"updated":"2020-01-16 18:28:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa574431e5021c227db2a0d52ae91d5767b5ad4f","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-01-06 06:03:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4ac8c80d8d81e7320329dfad8dcc7d5b78e398e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 08:29:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1ab718cccf624c1192f82e7247c3135d : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d6ed88d76a4aeebb83fe349e97d705 : SUCCESS in 20m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/564afb0f5c8b47fc86cb68974b3e0600 : SUCCESS in 59m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a8f020fe177f4dffa2a799083fbb5c5e : SUCCESS in 55m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c26d1c6047594a22a0e255acd4969728 : SUCCESS in 9m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5e25edda552c438a9901878884a444f9 : SUCCESS in 16m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e64dfa44fa1e4ab99317e04af1daed23 : SUCCESS in 11m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fd1b69b11f8b481fa3b5994d0844317f : SUCCESS in 14m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/955249686cf3465894d7c10443e8692a : SUCCESS in 14m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25d90931813644d89c1522f3a78341dc : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/590cdd65f3ff4608aa6445daeea0dbcd : SUCCESS in 13m 15s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/9a5aaadcaf1f4dbb8eb84658812c0862 : SUCCESS in 13m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/83f21c27ff7f444aba67fcf604c9e672 : SUCCESS in 13m 10s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/875d8d20b7624e92a4d22a5f41ed4e54 : SUCCESS in 15m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b50c8caede8f484d93994ba78fbdb354 : SUCCESS in 19m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/55dca6f708ca4565a832630ac3452c4e : SUCCESS in 18m 55s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/bfe4f33180514de0b2666d0dc74926a9 : SUCCESS in 12m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3517fe4ea48049e4b12caec1a1656fb0 : SUCCESS in 13m 20s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f86dcea1593a48439a2c2881c6ed8c23 : SUCCESS in 20m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff36b8adda4f459bb980b7412df5076b : TIMED_OUT in 31m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/194baa115e3d454cbb59734956339172 : FAILURE in 59m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64cfb5e3ba7740548a1ced2c07c86dfe : SUCCESS in 39m 48s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/eb89b04b324f4af1805b52cedb91cf58 : SUCCESS in 39m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9cef287c216a4efaa0765b454528763f : SUCCESS in 38m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/416a048d0850401ab2dd9162c6d8d565 : SUCCESS in 19m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f614b21548854efb8cef78561adda99a : SUCCESS in 12m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d59b48730cb4259ab785106504ec96f : SUCCESS in 25m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1bf5778515a4e7e8d3064b39b3d2126 : SUCCESS in 1h 18m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5aa9d5429dbe1c25073921cacf3f8e1c8a1731a0","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-01-07 00:52:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e7b2d969e64119a7129ae1c8eb17ffc8ada04dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 02:00:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/185440c5fb15424e9cf5c174285c0ac5 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7861ef3354084790a00980c92d7a3b61 : SUCCESS in 8m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c1e23735b01f4349b866a74025e7c06c : SUCCESS in 1h 01m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c21f38709c04531ac6cbca46bc9c5e6 : SUCCESS in 55m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32ccb9f4c8f541a1b59e3e97dab74866 : SUCCESS in 9m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8f863c91566d45f79c13522b257b4b9d : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/30be3206ff50488a8548e5135b85994d : SUCCESS in 18m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/18712d75912b44d7b3f431e46d5fa5d7 : SUCCESS in 14m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/be1429f7dfa549f183b7dc7ac3579bc4 : SUCCESS in 12m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2e9791a8bc784ec58bd4c9a0f28259f2 : SUCCESS in 9m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/91e29c254066487fb777c8dd1623920b : SUCCESS in 13m 29s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/df546cb2d95f466182cb12e8607f99cb : SUCCESS in 10m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b57d64352cb41da93b20a2ba953c198 : SUCCESS in 12m 33s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/1bc84c615192431ca811d2813fe9249b : SUCCESS in 12m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/12fa84534e4e42c8831b34981d2353d2 : SUCCESS in 9m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2a283aca41d545abb3fe73e1c16d34ee : SUCCESS in 13m 49s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6b4f568c15f24861b10a249401a0244f : SUCCESS in 15m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/31021f57b91b4ba49ccd2fb6c5542859 : SUCCESS in 13m 44s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/e75e465aa4af4b01b07042f57aa12db1 : SUCCESS in 13m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e3024bd91ff1425093a2c3f34736b2e4 : SUCCESS in 20m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2ec4c28a880943ae87d6eccdfd7a0fbc : SUCCESS in 38m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8cf48f93a69e42299a0896b56028bf74 : SUCCESS in 36m 12s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0a73d378f28a425980ddb3f999a318ce : SUCCESS in 29m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb7c4dbd29e240e4a79f69a46f14a300 : SUCCESS in 37m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34443012b8fc49c3bd59e14cdea06636 : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93c126e17974402b8abb98a0d69caf95 : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dfdda90608b94051a2ff91d06e4085ad : RETRY_LIMIT in 7m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8cfc9a79d4ef4f00aff95c84b6478c39 : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":2},{"id":"42b57c0ba06b1fbda131d77219609dc4337d58db","author":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"date":"2020-01-09 16:26:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f9879b76d074a8bce487b0344345624f246f9d1","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-01-11 23:21:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"61aed7d6cff42ce09e8fe704830d1b4e7ebe75ab","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-01-11 23:28:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa392c268a3a448fddcad5ba339284f0a15b9c19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 01:06:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bef542b9f3834746a54f2e60174e1863 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f57fcbdd3bec47149a34c888b947c99e : SUCCESS in 7m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6551aa4cd5bd414fb5f35c7fd64367ae : SUCCESS in 50m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e342946030524b05a0fb946df1224bda : SUCCESS in 57m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d966a90b87474f9a9bea22017596fe5d : SUCCESS in 7m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eb24b371d35e4c768f256d116c896161 : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cdd60a661c0e449cacf0546735205a49 : SUCCESS in 14m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/430a1d5f21d74194b30838f850f151a3 : SUCCESS in 13m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1c98befb8fe440fcaf401c9e4e6b9b9a : SUCCESS in 13m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f4f9481314794f8e8b7e00f033ec5b91 : SUCCESS in 9m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3452f487177c49eb88b664a5e0aa6063 : SUCCESS in 9m 20s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8dad3f4910a54f0483170f5977e756bf : SUCCESS in 9m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7cba8ba24c954a09b09586e9c421c8e9 : SUCCESS in 17m 54s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/87f7bdbda81a4b5d9e3cf480026e618b : SUCCESS in 12m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/23df5e8c9b8446969cc037e15630d89d : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/44d6e3c2edbe4d9682b33849d8c82406 : FAILURE in 13m 39s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8859b74c60c24f69a61bac3ea078a77d : SUCCESS in 12m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dab4230868f848acbe4ab98d2679247b : SUCCESS in 12m 34s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3da9e502b01e4e50bc6a18c32cb87d61 : SUCCESS in 13m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/676b7048ba6b45c8be49d6732b744b39 : SUCCESS in 17m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/80cc17cd9c4542ad8e453f3368dbad79 : SUCCESS in 38m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d02a62c670bc42759e5a45134746ea8d : SUCCESS in 29m 08s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8ce847dbe8f84b24ad96895085f3f055 : SUCCESS in 27m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de7dbaf5cd6e46f69671af028fb33ede : SUCCESS in 28m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75f8db12859544a4aee7bde599427d94 : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/decc87441d1a4bce9a4588908f7fa336 : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aea4adb450c64ea0a99e7a1a5bd1dd4c : SUCCESS in 16m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7eb69ba802994773ba251b8ca66a2f50 : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":3},{"id":"1e39ae1b339320a1a1abcb277cafc0e1f05553d1","author":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"date":"2020-01-16 16:02:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"37085b0f9e0edb77b1af6440ce687801dc55c06b","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-01-16 17:02:01.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4e05c6249c63d993d2368b92e778c5329ecfacd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 18:28:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0d8d05e545b74c128e42e7a65139e1db : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8772a2a098d14a18ab900469102fbd0e : SUCCESS in 9m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9aaaab1282a7489b9581ccef9b8b8657 : SUCCESS in 1h 01m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/31cfa7fe14c74220bc56efad2218bc71 : SUCCESS in 59m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d3dbc1c8c3a5435696d9a2eeea5b4c55 : SUCCESS in 14m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2c5c155ce0a640e994807d8669fe089e : SUCCESS in 11m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e477867791514abfad7f00b1f9c268c2 : FAILURE in 12m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/acd51ab89f754c44bf51b800eb24942c : FAILURE in 24m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dc4a2f5e5f584a92a510cd5d44705114 : FAILURE in 12m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c349c18a1cf64c17b40552f748173d3d : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/28515557c9704e47a35ff655e30e6f41 : SUCCESS in 10m 45s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1c2b130ad6c64bf0837d7f8068c4f7a3 : SUCCESS in 12m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/806714f0b20c4dbf8fc8e5fda707bebe : SUCCESS in 15m 30s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7e290c31632f4bc4b9f20e2609134b6a : SUCCESS in 16m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cd4b36609b464f0d8637dec2650cbb75 : SUCCESS in 11m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/877663eb49e54edd966e1e7faf319287 : SUCCESS in 23m 44s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/02f6b7d86d8d42e7b47b54b7124f1cf0 : SUCCESS in 15m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1a1f9505c9c9450d926bc8873bf50a57 : SUCCESS in 26m 57s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/078544116f4e4c8ea8aafade8cdc57e4 : SUCCESS in 31m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/39405a25013b4f968abf75f5af333577 : TIMED_OUT in 32m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9bf350e17e4c418199d444854481b246 : FAILURE in 45m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0436cd9937644579e3db4e8836eb38a : TIMED_OUT in 51m 44s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/afa5ec5cc88c4055be74e4de5936aad7 : SUCCESS in 39m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/67d65083ad0645ccb1d87e9a1625c4a0 : SUCCESS in 44m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18c0ff756e784f2b88700071b7d4cacb : FAILURE in 14m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af121c10a74b4981a436688c1c06ded0 : FAILURE in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca67fded11f34a58b47b635071270208 : SUCCESS in 19m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d853acd3964b4fe2b7d6c33fe0ad1947 : SUCCESS in 1h 24m 15s","accounts_in_message":[],"_revision_number":4},{"id":"526ce277d720ccaacdddfbff4765944018e553dd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-12-26 16:50:55.000000000","message":"Abandoned\n\nSee I1be1faec53b2cdfaabf927598f1460e23c206b0a","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2f05664bc5dd706b2553a1b966fed31225228cd9","revisions":{"529165d6fe46944938ec1f7da897c72ee6b58d4d":{"kind":"REWORK","_number":1,"created":"2020-01-06 06:03:07.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/39/701139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/39/701139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/701139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/701139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/701139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/39/701139/1"}}},"commit":{"parents":[{"commit":"70385420121b2bf9a41e895c06a6989b533bbe56","subject":"Let developers/operators add watchers to object audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70385420121b2bf9a41e895c06a6989b533bbe56"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-12-11 17:20:06.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2020-01-06 06:02:36.000000000","tz":-360},"subject":"WIP: Add a reference implementation of audit watchers","message":"WIP: Add a reference implementation of audit watchers\n\nChange-Id: I33b21412287d1ca2d368393c89b5c910b9521dcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/529165d6fe46944938ec1f7da897c72ee6b58d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/529165d6fe46944938ec1f7da897c72ee6b58d4d"}]},"branch":"refs/heads/master"},"0ce368c9a5530991968ebd9eb83816188ce3b107":{"kind":"REWORK","_number":2,"created":"2020-01-07 00:52:44.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/39/701139/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/39/701139/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/701139/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/701139/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/701139/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/39/701139/2"}}},"commit":{"parents":[{"commit":"9b69116b5c552f1edd477c6afa838874fceedaae","subject":"Let developers/operators add watchers to object audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b69116b5c552f1edd477c6afa838874fceedaae"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-12-11 17:20:06.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2020-01-07 00:52:25.000000000","tz":-360},"subject":"WIP: Add a reference implementation of audit watchers","message":"WIP: Add a reference implementation of audit watchers\n\nChange-Id: I33b21412287d1ca2d368393c89b5c910b9521dcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ce368c9a5530991968ebd9eb83816188ce3b107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ce368c9a5530991968ebd9eb83816188ce3b107"}]},"branch":"refs/heads/master"},"3102034c69ab74bb07de6b7f1c0c13917b2a3f6a":{"kind":"REWORK","_number":3,"created":"2020-01-11 23:21:32.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/39/701139/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/39/701139/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/701139/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/701139/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/701139/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/39/701139/3"}}},"commit":{"parents":[{"commit":"9b69116b5c552f1edd477c6afa838874fceedaae","subject":"Let developers/operators add watchers to object audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b69116b5c552f1edd477c6afa838874fceedaae"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-12-11 17:20:06.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2020-01-11 23:08:54.000000000","tz":-360},"subject":"WIP: Add a reference implementation of audit watchers","message":"WIP: Add a reference implementation of audit watchers\n\nChange-Id: I33b21412287d1ca2d368393c89b5c910b9521dcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3102034c69ab74bb07de6b7f1c0c13917b2a3f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3102034c69ab74bb07de6b7f1c0c13917b2a3f6a"}]},"branch":"refs/heads/master"},"2f05664bc5dd706b2553a1b966fed31225228cd9":{"kind":"REWORK","_number":4,"created":"2020-01-16 16:02:35.000000000","uploader":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"ref":"refs/changes/39/701139/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/39/701139/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/701139/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/701139/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/701139/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/39/701139/4"}}},"commit":{"parents":[{"commit":"4a166b91a3d4ced48173de44c865dad48e118de5","subject":"Let developers/operators add watchers to object audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a166b91a3d4ced48173de44c865dad48e118de5"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-12-11 17:20:06.000000000","tz":-360},"committer":{"name":"dsariel","email":"dsariel@redhat.com","date":"2020-01-16 15:59:49.000000000","tz":120},"subject":"WIP: Add a reference implementation of audit watchers","message":"WIP: Add a reference implementation of audit watchers\n\nChange-Id: I33b21412287d1ca2d368393c89b5c910b9521dcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f05664bc5dd706b2553a1b966fed31225228cd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f05664bc5dd706b2553a1b966fed31225228cd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
