)]}'
{"id":"openstack%2Fswift~703654","triplet_id":"openstack%2Fswift~stable%2Ftrain~I30e3beb8707b88c36bd3cdc7a0887d069e943ba6","project":"openstack/swift","branch":"stable/train","topic":"701497-stable/train","hashtags":[],"change_id":"I30e3beb8707b88c36bd3cdc7a0887d069e943ba6","subject":"py3: Fix formpost unicode filename issues","status":"MERGED","created":"2020-01-21 17:03:11.000000000","updated":"2020-01-23 13:58:12.000000000","submitted":"2020-01-23 13:55:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703654-1579787749029-9115562a","meta_rev_id":"0f1d71ca74e674782ecdbe9bf3e2ff97c413bcd3","_number":703654,"virtual_id_number":703654,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-23 13:55:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-22 20:29:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-22 20:29:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-23 13:55:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c941d735691f13b4c264b1a50466d87aa550b339","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-21 17:03:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c82d832e474cc7b2badcd35abe9323b4b00d8d8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 20:20:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f531f932ee400196c99bc6f5182a37 : SUCCESS in 6m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5cf21304cf7e479b9c45b39fb989dc62 : SUCCESS in 56m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/40678f8114df4c6fa82e8adbd0ed0586 : SUCCESS in 1h 11m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/65a2ea06b7ae4690a09041bc5e39bcd7 : SUCCESS in 9m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d63db3e4afc94c079d4039f6d4585db5 : SUCCESS in 18m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6e0aadd9d32b403b8367e0d3617fd14f : SUCCESS in 14m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3ec98dc375264241a3d9e1dc66320671 : SUCCESS in 12m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cb5e9aa73162453cbf6dd4cfa452b516 : SUCCESS in 11m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/653ef9c9d41b43f98836bb960f1af82a : FAILURE in 15m 44s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/20921ea8a18a496e8e6a40401237d648 : SUCCESS in 14m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/590b38b5836049f49c9a455e39da2d1b : SUCCESS in 15m 46s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/fd29c07baead4411b2bc7c3c119f57b4 : SUCCESS in 17m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a158045b83da4b26a0d5dc5b26edd707 : SUCCESS in 11m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f955c9e759a040eab909b8bebb71e689 : SUCCESS in 10m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/df2b60bbdb8a443f86cedfb8958cbe68 : SUCCESS in 11m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/222e838728b246fb8c2301d26ccef3eb : SUCCESS in 17m 58s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ffed11b6ec1041eb9ca92026dc97af4d : SUCCESS in 14m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69df7b6f1fce4573aacfe3bdb2ddb7fb : TIMED_OUT in 31m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b44865ad495e477d85e391048d57bcf4 : SUCCESS in 36m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d2511a78ffd442b4b5735842869898ab : SUCCESS in 34m 12s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7e46e09f7bc6460ab2aa22a581488dc4 : SUCCESS in 35m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a80fde7bce64341823aca07b0d01270 : SUCCESS in 30m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/124121e102e943bdb90affb3fa971b3d : SUCCESS in 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b453aaecf3495ca908ad48cff102e0 : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/96b9f77a2ecd47169241eb9fe97f8237 : SUCCESS in 19m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb9ef231f5c54c4a8b05e57697067e35 : SUCCESS in 1h 08m 36s","accounts_in_message":[],"_revision_number":1},{"id":"bd2f4cff99b92330cb6159a7e94f3219e2307d47","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-21 20:50:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c8ada5206262196b697c06ada4df504e0dbb1e71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 21:52:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaa1ea478dd548b99a9ba47270b25f84 : SUCCESS in 6m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1e0933bed57d4ed9ae04b902c69d5c91 : SUCCESS in 56m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/632f62e25051406789668c3cfe6516fd : SUCCESS in 50m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4e885b7ec8fe49878533afd785a86857 : SUCCESS in 11m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9b5bd72f3ac241dc9938b1d4aed6d1da : SUCCESS in 14m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4acedbcc9558496aa8c28b94be190625 : SUCCESS in 14m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7c15bcc6346747748c00b4ef093a3fb5 : SUCCESS in 13m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f01a3b270c694dc39786e83bd7a13171 : SUCCESS in 12m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cce04f93257842e79a245de6b51148b7 : SUCCESS in 11m 18s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4a8e8709d1714a4882ce32addcf142b9 : SUCCESS in 10m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/35f1b73d5f564e60b11121c25d35b2d9 : SUCCESS in 12m 38s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/07e8e4ac978d4b6bb349af750fab0e80 : SUCCESS in 16m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cc5023baf5484281a98e2575988f118a : SUCCESS in 13m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3308bd17176643e78d38fbb5c34d5237 : SUCCESS in 12m 39s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/dcaea57966894eff92ebc7e177ae6d03 : SUCCESS in 13m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/018aaebe33f545cea036e20378eadb59 : SUCCESS in 14m 23s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/39ada98ba8bd45ccbe2b58c147688baa : SUCCESS in 17m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ec1169a8619b46248760222763372efd : SUCCESS in 22m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b03689d7f7904673a7293d2ed88d97f0 : SUCCESS in 35m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08b07ad8a1604eee9715c90cee3fa167 : SUCCESS in 29m 11s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/53a16feed871415cbdaa9b2a1262352a : SUCCESS in 29m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c1f54fa65975403993f4f2b540f7c2bb : SUCCESS in 28m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a03c6338e63480d8b0b26849622244b : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af6de73fc8c34c75b820c63b5f3a0be3 : SUCCESS in 7m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b0ecac369114cee9c3090c4614854f6 : SUCCESS in 21m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7374c765353a4f2891092c8c986d4a1c : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":1},{"id":"942c2ba195665bdf49c83887599a40a48ceb2578","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-22 20:29:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"355c6bc5000efeff514ad7e8f66a483bb0bc276a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 20:29:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8bb69d6fd737237122c5a61d7093535ace48b932","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 22:18:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b609ea308e884069bc380ebb65e50f6f : SUCCESS in 7m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/85291990e9cd45efbf585524a1c687a2 : FAILURE in 1h 05m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aeff609871904f50bcdf2fd80f3b12bc : SUCCESS in 55m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/32c68bd4d0884528a0f673ba3cba89d3 : SUCCESS in 10m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e36206776f4146a2b0e938ec3db8b132 : SUCCESS in 18m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fa856b9f74fb436aac2dd10502950ab7 : SUCCESS in 17m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6856b53474e74f11b33078b1df098fe7 : SUCCESS in 15m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/66aa0d8db1b642c1a4ffb0ec28e6176c : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0f151df8e5ef48fdb4950692cd40ca92 : SUCCESS in 12m 50s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/574e3b6cc23f4a099cc87827bab2fe83 : SUCCESS in 19m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b303c0bece834df5b8bdc8da51920d82 : SUCCESS in 13m 25s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e3653c024578433da6af7078d8cd9584 : SUCCESS in 19m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ef308a1043d74d4b9070f8f92125da6d : SUCCESS in 13m 12s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5df477959cef43948f737ec496a5fdfd : SUCCESS in 16m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4d69b5e5f0b6489b97d9fb586e729709 : SUCCESS in 14m 54s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4225a1dc6d3e4de5aa5843168b8ea6d6 : SUCCESS in 18m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e996cc7637db453089b9f3522c41403d : SUCCESS in 35m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/04d8534c921f489b8589a085dab92aab : SUCCESS in 33m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4911bd3655994999afb2814bda2b378a : SUCCESS in 35m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/241ddba219754e538dfb18c22729575d : SUCCESS in 32m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9548c621b7c44901ba5e80aecaf40dda : SUCCESS in 15m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8cc81684b714ae29ee417074f4c2192 : SUCCESS in 8m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5532c1dd5c02475fb8e4c3f815ef17d6 : SUCCESS in 1h 05m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d1cca78ad9ca056fa5d3ef8acb1c26ad4c023b60","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-23 00:26:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"82642a5ca8b11d90d4378b743f298367c5749c92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 01:49:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f425bbd4c9454da57fb41405b9c2f0 : SUCCESS in 5m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/03dff0d32734434099762088245ad41d : SUCCESS in 1h 09m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/079fdad7633746f08da4ae7fbbdd8bcb : SUCCESS in 1h 00m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5cc36cfa626946fb8928ae1188a309cc : SUCCESS in 9m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e7a1baed090a41b8ab03d9bbd1d1896a : SUCCESS in 12m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eb0c953e92dd40c19b2651113c16066e : SUCCESS in 11m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/20c284fa2162413494cfb9e680abd7cd : SUCCESS in 12m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c9e23484ffb0451392ad805e559a141a : SUCCESS in 9m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aac18a1809544781a4922998eac95c28 : SUCCESS in 10m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1b094b303b904b14b6f9480f8090610b : SUCCESS in 10m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/03177b1a3f154232bb5697c3ab9254b9 : SUCCESS in 10m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6fb1111f43be4c2aa900c0151a8dbe03 : SUCCESS in 13m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/96731797794242fc8dc4e515fa591f61 : SUCCESS in 9m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/52b3320a2cc048d2901ab99ae28949e9 : SUCCESS in 10m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/dcc57e9403bc47e59c087b84b4baab47 : SUCCESS in 10m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/645fb1c79406441999f6fd4b606e8eca : SUCCESS in 10m 13s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d89861dcbf9d4bafaa96d39ee08f741a : SUCCESS in 12m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7c1eea5200314448974f66aaf673fc3e : SUCCESS in 24m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/facf2d8a03e64bcdaa5b9cc7998706fd : SUCCESS in 43m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/168cf28a2f974fae9cb64289e447bde7 : SUCCESS in 36m 37s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7b74541a335643689cfe4505f4619e27 : SUCCESS in 48m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/73173dd466934599982a2ab6a7a94654 : SUCCESS in 38m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7916b15d948f4a21b1c58e3def00cc10 : SUCCESS in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/089f11bd425a418a8d1980c35e16ecd6 : SUCCESS in 7m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8d2797cfbac64e69b7dd6bfb45a77fd1 : FAILURE in 31m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f10bf0282714768b8559bcfb4a9542b : SUCCESS in 1h 20m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d39825fc0d52da359ab99840307c6bf185d1c114","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-23 02:24:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2f8fd07bdfe086a17f2b6f811c2c5e8f11f95d97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 03:40:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f64a054be52e4989a1f63a05950bf9a5 : SUCCESS in 6m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e03cc520d35946999e123d472e186034 : SUCCESS in 48m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5b905e14f0ba4f6a804c0c7c9a2be79c : SUCCESS in 1h 00m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9cb30142c28547b7b33352aa864ab09f : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f77036703ea443b682645b95886941f2 : SUCCESS in 14m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a55292181ef4d1ab995e1a08bf20e9e : SUCCESS in 17m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/79990d6f21814687970da03d004da1ca : SUCCESS in 17m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/03541c1eba2f4178985cd56b64064de2 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5c3fc7684fee44f5839328535564b610 : FAILURE in 16m 21s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/667c09e57e6a4d1aabba1185150eb2e1 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e484415236c54589b8f7325ead1e0b66 : SUCCESS in 16m 52s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c2c1989ce5de4fdebee8f5ee901049a3 : SUCCESS in 27m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3b8157c8f1354782aa1044ff4fcb0e0f : SUCCESS in 24m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/05f6bd63425044e6b8fa5f1bf61d00c0 : SUCCESS in 21m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/25dacd5571684ea39127e74f0e87e5a2 : SUCCESS in 21m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/35e46dc6d56240e5854d710ee6079fca : SUCCESS in 22m 22s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ed6fc4e3ffd44b458de6f65e9e7cbdd6 : SUCCESS in 25m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/02e0789fedba4daaaa4d3117b4cddf01 : SUCCESS in 21m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ad94e08757f741158d7f70df116e590d : SUCCESS in 32m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/83802992ec19477b8d7e193a3a5cf604 : SUCCESS in 42m 44s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/2ea0df6f9cb04bce91f6bcbe47a3aa15 : SUCCESS in 42m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c1199b2015d948b08194f272964e6d5d : SUCCESS in 42m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a55ff84b8fa244c3b8f74e510cd207b6 : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41dadcba72b43028c085ba83867db84 : SUCCESS in 7m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55fe8acfcb1b4c308ea7bf9cac17e97e : SUCCESS in 16m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e5483d7f4bb490d9d1987a777299f72 : SUCCESS in 1h 09m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b157b85a04a0487cdb907d8882be762bc136f494","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-23 05:14:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8644d58bda99e201ab31254a1c06ed1faa866b96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 06:57:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02ed6e96c336428d9cddfea86f1eb228 : SUCCESS in 4m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7ca08096f1b145d1a2670c2ed4afae58 : SUCCESS in 1h 12m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59cef07ec1f34a48a54882ce4176a01d : SUCCESS in 1h 00m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8fa4992021474ee0b2b041f6819b767e : SUCCESS in 8m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/26e382190a6d4a79a935d8f33bf29005 : SUCCESS in 14m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad179cd18af242208d48da5a62576d1e : SUCCESS in 14m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4c397dcd93f349bcbaa0f34330f87937 : SUCCESS in 12m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9c81c540dc10453087113f9350193f25 : SUCCESS in 9m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/732a885286cf4994884436c98ddee607 : SUCCESS in 10m 00s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/20dcbca3b1f446e5b8de3b08f9eb4b88 : SUCCESS in 9m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/07c620cdf70e43d4aa5909de19a6627a : SUCCESS in 10m 53s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d52829f6b91940ff8fab29bb2e58ea51 : SUCCESS in 16m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cdb029405e27494bb836244024f903e2 : SUCCESS in 10m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ba04b79850bf4ccd9c1a94821062c6ce : SUCCESS in 9m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5a01aa9dcbbc43d99fa472945b4a6b6f : SUCCESS in 10m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6b6dcee335674b69b3688ce153ebb3d0 : SUCCESS in 11m 17s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ace78169e284491ebfa112fd33e0ae81 : SUCCESS in 16m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6af1c8a76e824672a1da9bdbc3f27dc9 : SUCCESS in 16m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1f9e1e9ad9d64d28b0905fa22b33107e : SUCCESS in 33m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8ba52ead35264d309603fcc84ea31d00 : SUCCESS in 27m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ade59775a7d64139b89eb23740badd5b : SUCCESS in 29m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30920d61e1f64ff3bc31a2adf98ba527 : SUCCESS in 28m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75bd9333dc314f00a9820ea2fd543b33 : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/279d9299d16a4207b0ffb7a60d086b22 : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/572ac568f7eb435b92b4cfbfb565f492 : SUCCESS in 17m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cf1c457f2d7468aa62476676a7a9e7c : SUCCESS in 1h 09m 32s","accounts_in_message":[],"_revision_number":1},{"id":"215d9352f0acb97dc36bb14e4fb6afa01bfc047c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 06:57:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74f02a21d4163a4947cf6e207ed05484c7ddd6a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 13:55:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ec7801e1f44b4380f41f6b29452fac : SUCCESS in 7m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3b677ba679044353a13b822c25901146 : SUCCESS in 1h 07m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a51da620465843b98123130261417ebe : SUCCESS in 1h 02m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b00fd0b9cde4413f9ff192b4adba1b44 : SUCCESS in 13m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ebde5b8797514558b8e1a1dc90dea8e9 : SUCCESS in 15m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6bca18b9b9ee41f3a4c78d3fc65b1148 : SUCCESS in 14m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ff80011241c741a5b365712564fadbd8 : SUCCESS in 14m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/29360aa583fa4ae28fa5a45f6e5f7bbc : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5f8daaff3e9f4611aac54a7c8fffc644 : SUCCESS in 18m 48s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/087aece4075a4504999cb2b2e7d791a8 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f4dd37d7287140729d72673f0e9bb0e2 : SUCCESS in 16m 32s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e80dc93442be4ffcb31f206751f93da4 : SUCCESS in 17m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e5d19a21fa014cdfa2acf0d36a681b4f : SUCCESS in 14m 25s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1afe97f31ce94146bfd0f3913ecdb42a : SUCCESS in 14m 42s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6807bbe9c6014544afa13f6a0dbfa5f0 : SUCCESS in 19m 27s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/72e25eebdbd5438cb6b2da3af36fed21 : SUCCESS in 22m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2618e2f66b9a464da77c0aaaa47eb321 : SUCCESS in 39m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d670b32cd6941b09355e3c1bc42534e : SUCCESS in 30m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/af5098efd10e4c7dba58982887d02434 : SUCCESS in 34m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d7506f2e6ec41dbb726973df6f1b3ff : SUCCESS in 30m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7720d695b8c48428bca96e5381f75ad : SUCCESS in 15m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f7cda1b9a574ba29a7b1e0ed5a7916f : SUCCESS in 7m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb617b6a8142487db1a3fd643198284a : SUCCESS in 1h 24m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9f7e53b43f48d07228556020d91487f01b60cd09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 13:55:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4eaf374637abb1950cebcad796b818cfa54a8ae5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 13:58:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae087e0418e49c795a5282b859070ea : SUCCESS in 1m 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/26d39e5e8f694b7fa622d4acdd382fab : SUCCESS in 1m 35s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48099c9823934bbda9fca82a49fc990fac3a9928","revisions":{"48099c9823934bbda9fca82a49fc990fac3a9928":{"kind":"REWORK","_number":1,"created":"2020-01-21 17:03:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/703654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/703654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/703654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/703654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/703654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/703654/1"}}},"commit":{"parents":[{"commit":"8fded7ad2521f03a101bc5376cd968742dc4c1b9","subject":"Merge \"Fix BadResponseLength error when creating symlink\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8fded7ad2521f03a101bc5376cd968742dc4c1b9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-08 05:16:37.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim@swiftstack.com","date":"2020-01-21 17:03:11.000000000","tz":0},"subject":"py3: Fix formpost unicode filename issues","message":"py3: Fix formpost unicode filename issues\n\nPreviously, we took the native string filename attribute and put it\ndirectly in the (WSGI string) PATH_INFO field. Now, we convert it to\na WSGI string first.\n\nChange-Id: I30e3beb8707b88c36bd3cdc7a0887d069e943ba6\nCloses-Bug: #1858259\n(cherry picked from commit 9483630ae130c5b2616f8617da25bcb51b647208)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48099c9823934bbda9fca82a49fc990fac3a9928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48099c9823934bbda9fca82a49fc990fac3a9928"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
