)]}'
{"id":"openstack%2Fswift~703857","triplet_id":"openstack%2Fswift~master~I8f42127fd6eb930282161930c65e8614f3e3175f","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I8f42127fd6eb930282161930c65e8614f3e3175f","subject":"Lower the log level for falling back to mkstemp","status":"MERGED","created":"2020-01-22 19:05:08.000000000","updated":"2020-02-15 02:37:09.000000000","submitted":"2020-02-15 02:35:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703857-1581734129099-b9790512","meta_rev_id":"e0e46dc46b3e19294086c2a6058278dc1f0ffcfd","_number":703857,"virtual_id_number":703857,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-15 02:35:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-15 00:05:38.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2020-02-13 00:39:40.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2020-02-12 09:38:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2020-02-12 09:38:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-12 09:38:27.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2020-02-13 00:39:40.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-02-15 02:35:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c1100f42fc4dfa2deca19056f56a4ff63ab8e18","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-22 19:05:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3af007130d80c388891c1dcc6796aeeb90c53db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 20:45:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb0ff0af21249b59f63e1b9a953ac3a : SUCCESS in 8m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a907efd7563b4714ae212a0b0d3daa6d : SUCCESS in 55m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a0ad82318136400fb3727a67d269798a : SUCCESS in 54m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0da47918aa984a538e2049c461cd4168 : SUCCESS in 9m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6f395bd5cdf9447e8303d5cef5a6ddb6 : SUCCESS in 13m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4fefbdb37c6b4c699798c9b2f7b02db4 : SUCCESS in 12m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/198b537c727f4674bb90888ae8da62b4 : SUCCESS in 16m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/51a78866db824d5bbb1dffe4c5f46a80 : SUCCESS in 21m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f9d5129e99944c949bd0261f0cd6c252 : SUCCESS in 12m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a9d58d67c0944e5abd21d059983df698 : SUCCESS in 18m 50s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ba679a33b9734fb6b0b8b7b85fa1ebb0 : SUCCESS in 9m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/079a139a1d5c460f8e4c2b09f31b82c1 : SUCCESS in 14m 05s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d90c5846d5b74f7cb3c5a94e3f9e25fa : SUCCESS in 14m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5f7ce0e68e6e4b94be9c996135715083 : SUCCESS in 10m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f71a6b8c80ad46248ad9ea7bad81d980 : SUCCESS in 16m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9fd6fa68a8554f1281d0a7dc0f9e379f : SUCCESS in 11m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/55665f97b9ce42c286d58cc63314bedb : SUCCESS in 19m 19s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4222ee1bce62475b862f2b0585294a1b : SUCCESS in 14m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/54577bf1e0f14564b6d40e80a42c5b95 : SUCCESS in 19m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7d4309d423114b748466d514812eb68c : SUCCESS in 54m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/53e9cf5538d7419aab399334270c4738 : SUCCESS in 31m 50s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0c37e41ae5ec48aa9c3967ac62d61c89 : SUCCESS in 31m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a489603eeffe49969caa03d5390675b8 : SUCCESS in 33m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fab7e14546db49d1a726286c58ea87f7 : SUCCESS in 17m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c25a32bc86ac4498a3f2e5d53f459484 : SUCCESS in 8m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b62bff2f32547b28033e809b1f80b78 : FAILURE in 27m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/934a943c64b9443a8d4768df3e2952d1 : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8c6c562cedbe84c65591a4b5c7f1b65ada60b0ef","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-23 00:35:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3e8d49271290c765295078edc6a23c8136b8f944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 01:53:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b0e3aa55954c798b810b88e8f3ccd5 : SUCCESS in 10m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5ef0575609a9421d9a9a88722327b39a : SUCCESS in 1h 13m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c021d488fa044a199d6a40ba31f67445 : SUCCESS in 1h 13m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5619761311d64fdfb1e412eb4f85601e : FAILURE in 14m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2b336d07fc854adca48d8e298c2bc30b : FAILURE in 14m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca9f1d1ec7a24c41b91e2151572e3a43 : SUCCESS in 16m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a8219c5581d84f18abd7fb884e68722c : SUCCESS in 16m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7bc25ee4df1c40e7899006c3010d91c5 : SUCCESS in 13m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c3a37b2c34f14f7abe87bfc0ea3c5fdc : SUCCESS in 10m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0bc3e2b442574e4095866f7241fe22c2 : SUCCESS in 10m 21s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/dd0b5020426945b4801d4c4282170b37 : SUCCESS in 10m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4fa393a15b04c33b02d249f19516001 : SUCCESS in 10m 33s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e7ab52b6ae754c8a92090cb0beaa9d17 : SUCCESS in 13m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a0240ef3672249598225425929883164 : SUCCESS in 9m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/67c1d9f44fe84461997f9c6baa1cb8ec : SUCCESS in 11m 37s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c89bc538fef24b218d1e4ca3f34f6c7a : SUCCESS in 10m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/15be59b051b843df8fc84425d3ee03ae : SUCCESS in 11m 00s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/27e7284b7e1640c58f2ec9b60e92346f : SUCCESS in 12m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6fd089164bf49e392f235fc518019f4 : SUCCESS in 16m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2aa1266712864f8291cbada237f115f3 : SUCCESS in 43m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/adb13852566947218b02671256ecedb6 : SUCCESS in 29m 45s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/422a3cc678164de685b8b5910a963f3d : SUCCESS in 27m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/449f643739a2404fbfd7796a136a3d60 : SUCCESS in 29m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f524fd1f5f74459ac054be34984c33f : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/644b5183c9754fea9cb568444079586e : SUCCESS in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30ac240cba70499599b0e165609fce21 : SUCCESS in 19m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46a84397c8f242f9815cb5ab675b2d57 : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":1},{"id":"39471746c2f2e9b66bed4c9f2c8535f5e40b19cb","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-02-12 09:38:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"adf865c35c0548ce2efdbc6c2b73d774bee542f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 09:38:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce74abd6cf0aedfb7db9b1833a952099301723d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:16:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b37e221a71fb485e83597c38413ac415 : SUCCESS in 11m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f9ae378e64834af78f04191cf3a451ca : SUCCESS in 56m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/61c7b037168f4563953a7447ab5edef5 : SUCCESS in 1h 03m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a188a1cf5d83479d8d7f35e0f5dc3810 : SUCCESS in 11m 32s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2fb5aad2e6754aa792da8214554808b0 : SUCCESS in 12m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3e0a027e5a40413ea07087aa92f5eda8 : SUCCESS in 14m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/392ad3368eae44c2b85ba5f26a957150 : SUCCESS in 15m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/84ddd10119e84f039c7c2c844bfa166d : SUCCESS in 13m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cd070e0025d94cdf95426010179305fb : SUCCESS in 14m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bfcde9f5f15a4fe2b29670f3ca287444 : SUCCESS in 11m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3349d505368f45dca1081fee77c69dee : SUCCESS in 13m 48s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b20ce0770698460f873b05bcd203359f : SUCCESS in 11m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ce46062da64245cfa95e2fcbc989f159 : SUCCESS in 11m 55s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e2b2f17bad64449dbdb5367881353c3c : SUCCESS in 18m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f0fd49efa6a1446ba168c973b075efd4 : SUCCESS in 11m 29s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c2b490c9676d452fa22c74b130466599 : SUCCESS in 23m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7830c864f3bd4dc581a29b36c3f8d612 : SUCCESS in 14m 10s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/533ec981db0c4cf7a875c1af00b1899a : SUCCESS in 23m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a8cfdd4e75e94014a8f69510b0f19976 : FAILURE in 9m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cdba3b71ac2545149c3ef5623a3b1635 : SUCCESS in 35m 03s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/764ad23db7e641f2b8ef60ec85da6cdc : SUCCESS in 31m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a637d27e94d4fdd95e037713dd529fc : SUCCESS in 29m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cc1c6ae136d4c588103632f86176481 : SUCCESS in 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a8f3e6ece4044f9b0271fc864044b52 : SUCCESS in 7m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e943aee5b8a4961bea517136defa53f : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5cbe3c6effb5db0bb619b4a00655debaab7c2f40","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-02-12 12:42:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1496734e974bf219274921d4418e9d534242aee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 14:26:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882e03a9a2e44821affe1bc8bda2df51 : SUCCESS in 11m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8240398a9d7d46ca8a0ea37f41e76389 : SUCCESS in 59m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b48816ecac1b406ab7ec9bbf78c35200 : SUCCESS in 1h 14m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d8d3763dceeb4a1783ba3184c2879df4 : SUCCESS in 12m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/26f4820dc7ce4a27804b7eeddbf389c5 : SUCCESS in 10m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ff964632d75d4f27a00dfda075fe9fbb : SUCCESS in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8eca6700473943a783419fd7819f8ad9 : SUCCESS in 17m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8dd6cb59f5114c279a81398be3945656 : SUCCESS in 17m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d8ab2284245f4622a06bf68e3c2b265b : SUCCESS in 18m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6b2b5ddda917493fa58f8f3caacb760e : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3649c63671844407bd8eacb6e3d23c48 : SUCCESS in 16m 39s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/e3153a8cc7d146d3ac8516ad716ec840 : SUCCESS in 14m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aaeb067fc0c54f8f81933272bb101766 : SUCCESS in 24m 50s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/afe18b4f4df847a4ad60a82118913034 : SUCCESS in 20m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/45c0f62142484e51bb6f271dcb63639b : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/eed3dce91b4b41ee9e5066012953be39 : SUCCESS in 13m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/dd946b10454143f084f86885adabdb5d : SUCCESS in 11m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/481700fa9d794d0faef6f77b105bb6b1 : SUCCESS in 22m 39s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5e312e479bb14468bf1367da558daab1 : SUCCESS in 20m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a937acebeaa8477091e3e405f83bbee3 : FAILURE in 6m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6d6174882ef548509057243b5bc87887 : SUCCESS in 55m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f9ca849e3ca84642b3ed383362bd4ff1 : SUCCESS in 43m 11s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b137210b2ca0414ca769cedfc9a647bd : SUCCESS in 44m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d72951bc6fde457d8934dd695014779d : SUCCESS in 45m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/561226bd57504ca8a3475f3b55a24f86 : SUCCESS in 16m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac88d9faf6634285b0651eeeddbf4faa : SUCCESS in 7m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/feeab28f88984d2d821ff2087e88f98d : FAILURE in 14m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8643645e09640dd99cf76ff51321864 : SUCCESS in 1h 19m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ca6b7d6fffa44d7913af11494bdff5a5f2159d8c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-12 18:18:00.000000000","message":"Patch Set 1:\n\nrecheck\n\nLooks like virtualenv got fixed in https://github.com/pypa/virtualenv/commit/77f0dd80a12105864d8475abb6609dfd693ef5ac#diff-79b6454ee141f415b84cfbe2e037cc60","accounts_in_message":[],"_revision_number":1},{"id":"f46029f931fd8e18c180e962b300000004af2733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:28:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c3bd9902144f3f856a3d7208194ac5 : SUCCESS in 13m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/65ceb22085c74e29bcd80494bfbd274c : FAILURE in 1h 06m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/efadd3173a9d4bfabc52f4ae6062cf6d : SUCCESS in 56m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e81f0351c294ffebc34d7927cf17356 : SUCCESS in 15m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c082077d7a6d4406abbb2d7cc131a5f5 : SUCCESS in 10m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/516e1fd723684f85b9097f1fb0172412 : SUCCESS in 18m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cefe7049fa174547b87c34def00b5447 : POST_FAILURE in 15m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7dc44687130c4c7c9f7d208280feebe7 : SUCCESS in 15m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3242ed90f01e4c63a1978c4b99e13f88 : SUCCESS in 14m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6c4d37fdbac04bd789b62994900459e4 : SUCCESS in 18m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/74f4446757244b539826c54a5ad0b208 : SUCCESS in 15m 01s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/6e8ecaa892704257a31a7b94125fae4a : SUCCESS in 15m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/56520873d4c44216ba3e5ab24fde28dc : SUCCESS in 15m 19s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/55f3dd284f4743759dac5fc2e0091b35 : SUCCESS in 19m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/905f593089214dea99ea2f233080bb0b : SUCCESS in 12m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ea705f0716f24cb9b8774176bb1be6a0 : SUCCESS in 20m 37s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/74b0699b78054459b0d70bb24bbf25e9 : SUCCESS in 14m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f074e1f02dd948728b5993d02d32472d : SUCCESS in 11m 48s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c0d6ce390bcc4f8480c28f9ab5f48143 : SUCCESS in 23m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/24c5bc15bd0b4d48873b7c736cfcd6cc : FAILURE in 10m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/15c6b90403324e1da5a5be895d1a17aa : FAILURE in 44m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f0e551ac12c2406394c1e679073068a6 : SUCCESS in 37m 51s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/61c09037c3e04915b01b0bda4be2a9b6 : SUCCESS in 33m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22d646fe022b4690aeac131376db7dd0 : SUCCESS in 37m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb16454797384ccf95e095b0e38cf6cb : SUCCESS in 20m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70abcaf22031497c83ea5a07e4e3fe59 : SUCCESS in 9m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3eef852dfacf4a639500e09534090933 : FAILURE in 13m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e88725dd0e44b46bd6cd9c1366ba1ab : SUCCESS in 1h 01m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f86965bc6278b215c3cbf633c1ccdacc61ff9631","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-02-13 00:39:40.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ca9379325e358df356a3da3b8e27c1376c4ad3c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 01:53:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9af52b7bcba047c191b906ea6037003b : SUCCESS in 9m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4897c9870c54483a9123372b380b4807 : FAILURE in 10m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d897caa071ce4378af4d45e1848944fc : SUCCESS in 57m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/021a67d0d00943f8858960339b489cb4 : SUCCESS in 9m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ce08682bb2c64f419b3ad7255293d9a0 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/16eaba3f184a45f19f419bc789b0ab9a : SUCCESS in 16m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b8bc1e51df4e498b9fa6ce9314023a76 : SUCCESS in 12m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f1306513f66a432f8420f8d9c96e9238 : SUCCESS in 18m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dc89c04873864065a5d4ba49a5e78be0 : SUCCESS in 13m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/030cff6381d446af850792f0881a40e2 : SUCCESS in 12m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f3e70b1ccd4f461dbb2e980d7a2f8946 : SUCCESS in 14m 13s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/92a60734f168424ca662bdc7d4f94720 : SUCCESS in 10m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/beaee2ed84dc4b3f9e7f160a78d030f2 : SUCCESS in 11m 38s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/cdc39d795fba4ba584aa48488d0c62eb : SUCCESS in 19m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f2e1ea5907af40a293602ee7bed12f73 : SUCCESS in 12m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b3c07769d4c04f82a24584efbb42049a : SUCCESS in 12m 54s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/49dd845210dd4dbcbbd424ed1f6b4c1d : SUCCESS in 11m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/271b7cc7998d49d281bba3a736e6d71b : SUCCESS in 11m 37s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d384ff5ce08e407785c13a72fa5b5cba : SUCCESS in 15m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0e277dc966ee49f984fa0180c8cfeada : FAILURE in 5m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4c2b1b02b0434d478a4b6d3c98ee6cd0 : SUCCESS in 46m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/172249ae6ba240ad9e5c126973fc7498 : SUCCESS in 30m 20s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8c625239e389495cba4bb93f5dc0ff22 : SUCCESS in 31m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8fee86ee05f54d61bac3d7636c8a89f7 : SUCCESS in 31m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4c21ae2552249f9bca3d1523c613586 : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92db5473f8f34207bb6af29130df590e : SUCCESS in 6m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/85000a7ac7d34edd9b40840da75b443a : FAILURE in 13m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97364042a2da43efa19d2f8ec1d3b636 : SUCCESS in 1h 08m 22s","accounts_in_message":[],"_revision_number":1},{"id":"77c9db4b7d014a6604142e4880361ea376b2d272","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-15 00:05:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"79f7bf5f30c57733c9686e65e0a16e4e865b6a69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 01:24:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90f43118b0a44a95a769784c18029d1f : SUCCESS in 9m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c115a7f3942241728b878a8c276e1527 : SUCCESS in 58m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/141d1ca570284c93bc59d57d4fe90353 : SUCCESS in 49m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/93358061c56641f89d95fd2169cd3fa0 : SUCCESS in 10m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6d4a710e2c5640f79b4952f56c9fc411 : SUCCESS in 11m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7032b40b6d8f452a858023bd1d13fd5a : SUCCESS in 15m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/21c3ed9a2f704147a3a35f3bf8d71594 : SUCCESS in 15m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/99fb64f92b7c46e9a4c144f779e83b43 : SUCCESS in 13m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1419d2ccffc14c8ebf68e04234a064d6 : SUCCESS in 23m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e3ac05535848496a84beacaa01b41c42 : SUCCESS in 10m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/21fd9a3e364b4d0e9f0a79d3ec259c3b : SUCCESS in 24m 45s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/36c3dd00d4ae47ea9684b3740af31294 : SUCCESS in 19m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6fb0c6e196af4cd8830bf973edf17093 : SUCCESS in 21m 01s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d9293f0b0ae04fdf944125a9e8864c38 : SUCCESS in 26m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0f1f3393eb6e4d6db40dcb02d962f73d : SUCCESS in 10m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/694b09f591784b6aade489a18a092ffb : SUCCESS in 13m 28s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/46612223481d4567aa0a165ab461b3c2 : SUCCESS in 22m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/aa2334d199bc4f66bcfb8bee6af0cc19 : SUCCESS in 11m 52s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/775af787f6734d819aa205c742ca207e : SUCCESS in 16m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ffa4a1a4d6c946b2a2dd81e17696e4cc : FAILURE in 5m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f422a3992d9e4025a4853eaba87d8486 : SUCCESS in 45m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf5ecbfeac514ac088555ff5b7f881fc : SUCCESS in 34m 57s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f8f18dcc17734dc9abe8cd1db8325fcd : SUCCESS in 28m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/467f79c9ab01476d9e15d3e0fc3d5a15 : SUCCESS in 47m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f50f7460dd64a438e655f5a987b0452 : SUCCESS in 22m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d9c5b346cd24d9c90643c4be5f4a4b5 : SUCCESS in 6m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d28c89ea407e478e9b9135d0213f7eb7 : FAILURE in 11m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f325ab7effb4690bb09d84a98b941c2 : SUCCESS in 1h 15m 43s","accounts_in_message":[],"_revision_number":1},{"id":"31866550f560ee6cfccb6257fa3ef987436d8e68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 01:24:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"369f36455203fcbdbe213e646497298b156c0786","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 02:35:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67a5dea2715645698aa7be5f8fad1b23 : SUCCESS in 9m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6867f703df924dd0b599b122a3b6348d : SUCCESS in 56m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bfd4501fcd9546c380b340f0bb71a245 : SUCCESS in 59m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8edbb02aca35435f852fb5a2547659b5 : SUCCESS in 10m 01s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/0401bc727680465197f16ad4d83958f0 : SUCCESS in 11m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6bfd5dcea5fd496f82fe2318c532a56a : SUCCESS in 12m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/71fede7054d84277b186b0f017f96d54 : SUCCESS in 13m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/861a21c88df24e75b554ef5cf8f14576 : SUCCESS in 13m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3e7d3c750d594590984b193137f99eff : SUCCESS in 13m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3c71058cffc44158b979d28ba95fb4b1 : SUCCESS in 11m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/96129cf4f5a14944ad376bc6a294f9ea : SUCCESS in 12m 06s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/0e83ec6c082d45bf83318d3a5dc27bb3 : SUCCESS in 10m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6d79ce84694447ec9ec05c550cdb1610 : SUCCESS in 10m 59s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/be366239a774462bb6acce3840112624 : SUCCESS in 15m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c71ae29ab7954503aab8a1d4c333c5c2 : SUCCESS in 10m 45s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f645f6570c1c4a08b0f0465dded792e9 : SUCCESS in 11m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9957afd65a4c464294ef4040bdbdc0a6 : SUCCESS in 11m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ad39a6c69a3944979e747f4095e71be7 : SUCCESS in 15m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/899bed4f57e943f5a3b3ba15b9404c16 : SUCCESS in 47m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3f84c6a54874b0e95abb2e7f5e3be47 : SUCCESS in 29m 04s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e60b4ec78af6480b945a9652258327e1 : SUCCESS in 31m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/131047c466df4f35be6df877d89dda4c : SUCCESS in 32m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/261a8842d74a4527ab256a9a7153a23a : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f2106b252f640f5ac5d255268fc872b : SUCCESS in 6m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7fd723c266949dfb1a4bfbdb5c5f27b : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":1},{"id":"735d8049d5c9a0bb65f1cb8ebccb132f6b9cbe5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 02:35:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e0e46dc46b3e19294086c2a6058278dc1f0ffcfd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 02:37:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dc5ee246338473094a832a31f8f6f1e : SUCCESS in 1m 23s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/94b21c1a4e4d44298a7c69e210342505 : SUCCESS in 46s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/36b956a0237643629988e3eebfb9be55 : SUCCESS in 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55cdb55075657c283ce786ca634f8a42ec786490","revisions":{"55cdb55075657c283ce786ca634f8a42ec786490":{"kind":"REWORK","_number":1,"created":"2020-01-22 19:05:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/703857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/703857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/703857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/703857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/703857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/703857/1"}}},"commit":{"parents":[{"commit":"02aea34c46249fb315ff1fc0884c2a91ac663783","subject":"Merge \"swift-account-audit: work with encryption\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02aea34c46249fb315ff1fc0884c2a91ac663783"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-22 19:00:18.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-22 19:00:18.000000000","tz":-480},"subject":"Lower the log level for falling back to mkstemp","message":"Lower the log level for falling back to mkstemp\n\nWe recently started trying to use O_TMPFILE without the kernel version\ncheck, we gracefully degrade when it isn\u0027t available, and we don\u0027t give\noperators any way to avoid the O_TMPFILE attempt -- logging at warning\nseems way too noisy.\n\nChange-Id: I8f42127fd6eb930282161930c65e8614f3e3175f\nRelated-Change: I3599e2ab257bcd99467aee83b747939afac639d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55cdb55075657c283ce786ca634f8a42ec786490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55cdb55075657c283ce786ca634f8a42ec786490"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
