)]}'
{"id":"openstack%2Fswift~704108","triplet_id":"openstack%2Fswift~stable%2Ftrain~If5fb164693e395f89d57899fb8ab355f1e3f817c","project":"openstack/swift","branch":"stable/train","topic":"fixBulkDeleteAt-stable/train","hashtags":[],"change_id":"If5fb164693e395f89d57899fb8ab355f1e3f817c","subject":"Allow bulk to fwd some headers at tar extraction","status":"MERGED","created":"2020-01-24 06:45:43.000000000","updated":"2020-01-25 19:23:17.000000000","submitted":"2020-01-25 19:21:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704108-1579980108859-8f9c0b7e","meta_rev_id":"36d0afe856bc527ca7e2f6a8484eef0d5dc5856e","_number":704108,"virtual_id_number":704108,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-25 19:21:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-25 01:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"date":"2020-01-24 06:45:43.000000000","_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-25 01:11:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-24 06:45:43.000000000","updated_by":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"reviewer":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"state":"REVIEWER"},{"updated":"2020-01-25 19:21:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"827772bd37975d0b583999e80028e97042f59cbb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-24 06:45:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"38c94b5bcb6f53cea6918f13a5371002a6548f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 08:32:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd6a224b6c3b49679992e791631f7bf3 : SUCCESS in 5m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0e218e75e6ac49e58cf7b0a8ec1a9dec : SUCCESS in 1h 02m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/285f247633bf49e4b2fe97fd1be06c9d : SUCCESS in 54m 36s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/74bdb5c2e9454814b814fa90a5105ce4 : SUCCESS in 11m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/318c111509f54719a2f1cba3a059aa41 : SUCCESS in 13m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9869a11557294d1fa0f43b5d02cbe317 : SUCCESS in 17m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b291d9dfdcd24a2f879ba026036c2c4c : SUCCESS in 12m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ce3267ebbc73442c8e1f8143ea054980 : SUCCESS in 10m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/10f29f3551394114934765a516deff90 : SUCCESS in 12m 47s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1512d4014d8341a6afd334c435682161 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/169ecf28cc3c4859a1d2f06cb2e85038 : SUCCESS in 15m 08s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/fe546f6f7ac84f3d94e757bd10430638 : SUCCESS in 14m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b8b6e2b0bbe44724b4d3e8b41f7ffc7a : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8020b5888bb54c30bd35b10cd0d0e8bd : SUCCESS in 19m 14s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b8c3a441edf047c4810d9b7f5e9f2a1c : SUCCESS in 18m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/95d6e3bc9abb4af19d4e79826a08bbf5 : SUCCESS in 17m 03s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/e4ca299c1133413797353e105e0e2446 : SUCCESS in 12m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e35121c331804ca3be12b6c160fb8d4f : SUCCESS in 16m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dda960821a844f8998fe3d26fe865615 : SUCCESS in 33m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8884c16b6edc4eeb941f2ac85450cb47 : SUCCESS in 36m 43s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f6cdf9c47ce24f0e9833b22c492888bb : SUCCESS in 29m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a9d9057df5e4056b5ea5b886c090cc8 : SUCCESS in 30m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c7c2530af214e55b1320ecf9bcd8e20 : SUCCESS in 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/398556ce4257430dab1148588f96b8bc : SUCCESS in 7m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/442dbfff145a42728fe210107a8fa486 : SUCCESS in 22m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d20a93b835c54a148f68bb1593ce73fd : SUCCESS in 1h 21m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d4f85a6ec6cb53ded634fbc2d5e432cd98891597","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-25 01:11:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a19ce1c57d80ae3c04dadcfe9b980b779d133569","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 01:12:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c21cf2a8f28b4627b1664f0e624cd5daa7d3675","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 04:31:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edec1e5c87ac4e4b8a3866897301ed23 : SUCCESS in 5m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/95e171f359c142cc9a47a88c9dc1e0db : SUCCESS in 55m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9ffd8eb26d864036993dcb0deaf1360e : SUCCESS in 56m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/66a2875edfee42b98fe27abdd29fa12d : SUCCESS in 9m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/80cb53884b80434cb2b20bd5ee35ae03 : SUCCESS in 15m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/291baa3e131d4b44b661dc77999250b7 : SUCCESS in 15m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bd7a21eed2804994b8775155bf8b5741 : SUCCESS in 30m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/628c40f9f96548b395e4ceb6d16f7039 : SUCCESS in 11m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a7796be4b46a416cb6008550911bf665 : SUCCESS in 31m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/7f05b4ff8ed4430d83f9647efe201a34 : SUCCESS in 11m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f501014c406c4a47b8accaab44f4a543 : SUCCESS in 18m 35s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a1ec745688e04da5ac552a10c615763e : SUCCESS in 36m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f003d0a1e5b54b18b6a2e95c103e77b8 : SUCCESS in 29m 14s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/acbbbf5b6f5d4c01a54483989a9d61b3 : FAILURE in 29m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f0a7f672590147e4ab0fe6b89e3a31b1 : SUCCESS in 21m 05s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/55129e8ddea34a91a0a8888d3845e68c : SUCCESS in 40m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/568b10f368dd414d9857f4aec8a6b02d : FAILURE in 52m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fe5356201e03445a8e631d9a3293bc1a : SUCCESS in 45m 43s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8c4515537a4f4341b8c96a5d59c12138 : SUCCESS in 32m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de0313d92236432db5b508fd6c2dec0c : TIMED_OUT in 51m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/632cf235ce0e42638609bac9c94d996e : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f35a2966ab114a7589edb5b69c6c8015 : SUCCESS in 8m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f6f52cc0e334abdbc1b59d75a90ba04 : SUCCESS in 1h 18m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d4b8c20cceb51f17f6039d8d400f953eb6d9b59e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-25 15:47:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0d14164963ad648bdebbee7472f9eb2012587a00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 16:56:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06037a2da13d4dc6b98b548881b30480 : SUCCESS in 4m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/52c609b1e670404d81e093edde8ebdf7 : SUCCESS in 1h 04m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/adda91d12c5d4352aaaa031e07338edd : SUCCESS in 1h 00m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/099f826536454078955f4f4f53096273 : SUCCESS in 8m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/77c40f3936c94f84a786165b946e3ece : SUCCESS in 11m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ee0f9563e77484ba1a14ea19790246f : SUCCESS in 11m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2a36e57afd2744e496cfa8ad639556d2 : SUCCESS in 11m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/307957076d5d4ea0aa8af638ca9f1d63 : SUCCESS in 9m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6877141bde3444f49bd44596b2b4648e : SUCCESS in 9m 53s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/0363405354bc49018cd0bfb44b1a4ebf : SUCCESS in 9m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3d0fb88e54014c36ae3c5a5967371dad : SUCCESS in 10m 24s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d4609aeed1ee46478fcc69bf8e6fffdd : SUCCESS in 12m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bb79f4f7b168490ca809280a3669cd8b : SUCCESS in 10m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/829c8aefc67243cb9addb0063c09e838 : SUCCESS in 9m 59s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3918a52ad5b245d599ebb4059330ce45 : SUCCESS in 9m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8914c4dee2f4498a94c100f99ebd2cdb : SUCCESS in 10m 20s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/82884d35474d4f76ab7877d42b2e9e3e : SUCCESS in 13m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f8b56deddb2e4041855f3f0abc737eaa : SUCCESS in 16m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a47e991e2be048cabac1ab0966fed11f : SUCCESS in 31m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52f643c9bc164cd6954dfe543f3d296a : SUCCESS in 29m 07s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/03ccd3b83600421ea897085156f35c02 : SUCCESS in 31m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/35a3804cfc0b452f924b1e46ff5c9e79 : SUCCESS in 31m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04dc4e3d3bd14f26b1f093fc38237257 : SUCCESS in 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c9958123e8e4f93bc255f5d553568d2 : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ea49f285cc746e0b0bee54cf1c6ba65 : SUCCESS in 16m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4a63c62044a407e9fb76df298e9649f : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2406ae1ad146b2a13c98abe04af905a12b1d93d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 16:57:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5915a9b257c3c7857f9eae07193ce6c607a7b0b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 19:21:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4c8508bced27136af6b2bf5113d655a1aebfbb5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 19:21:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b56e81bb2184d76840b94eff6666827 : SUCCESS in 4m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1db0caeb222b4243b0f9b981c1a0fb18 : SUCCESS in 57m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/57d932a08eb944389560a9ca46077c87 : SUCCESS in 54m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8f87aaf8b96b43f79df756ce90a5a153 : SUCCESS in 9m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94bede107b494d30b0e3c7dcdf7b6ed7 : SUCCESS in 11m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b31be437e6bb4d4e9b50883b28e9e171 : SUCCESS in 11m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b15a9363f8ea4a9d81e4446aa522be25 : SUCCESS in 12m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fd6cfc659ca045b3b31f1092bcf65d3f : SUCCESS in 10m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/40eb96a12b874ddfac0c73d29ec1d2b8 : SUCCESS in 9m 38s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5e66228a3d1d4149a0ea6bde43725cd9 : SUCCESS in 11m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bea84d3ee0c24601b476cc62e6195869 : SUCCESS in 15m 09s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/dffd4f3db2ff4bdd853f96616c6621df : SUCCESS in 18m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fff5d1f852074355a2160f46d685b380 : SUCCESS in 9m 34s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/24698358022243b6bebc4a63b87ff9f7 : SUCCESS in 11m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c8131a8aba2a42c9856141af7edcfe46 : SUCCESS in 11m 42s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/57e9f7750d21455c8ce107a0b30285ae : SUCCESS in 15m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dbb760369d9d4358b42609c4093c6184 : SUCCESS in 47m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0636b52c27124758b6eab24ff25b9318 : SUCCESS in 24m 45s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ca195bd5a3134b6fa06d76f54e16c5e9 : SUCCESS in 30m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0da6cd62768940b9b560e8e3dd1024c8 : SUCCESS in 35m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aaaf1322acbf42b29d1cbba7b573ebe4 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8fa18df36a44a4fafa97c9a61baf95d : SUCCESS in 7m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5129d7b689094e2ea9759df52d7f1a8d : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4732de29c479d53ea49b64f34ce6e441dc29cdbd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 19:23:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f29b4677d3554dfdaca0b643bb2e805f : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/de1c8a52405944bca9392baa742aa639 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e413c650e385ba44523e99e023ada0a44aee8c7c","revisions":{"e413c650e385ba44523e99e023ada0a44aee8c7c":{"kind":"REWORK","_number":1,"created":"2020-01-24 06:45:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/08/704108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/704108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/704108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/704108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/704108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/704108/1"}}},"commit":{"parents":[{"commit":"21f81ed8f63a0da40b7dcfefac54c535cc17ed2b","subject":"Merge \"py3: Fix formpost unicode filename issues\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21f81ed8f63a0da40b7dcfefac54c535cc17ed2b"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-01-09 14:08:25.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim@swiftstack.com","date":"2020-01-24 06:45:43.000000000","tz":0},"subject":"Allow bulk to fwd some headers at tar extraction","message":"Allow bulk to fwd some headers at tar extraction\n\nWhitelisted headers include X-Delete-At/X-Delete-After and all Object\nmetadata headers (X-Object-Meta-*)\n\nCloses-Bug: 1857546\nChange-Id: If5fb164693e395f89d57899fb8ab355f1e3f817c\n(cherry picked from commit ff0753fe19be3d7442aef3972e08adc48d25fe93)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e413c650e385ba44523e99e023ada0a44aee8c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e413c650e385ba44523e99e023ada0a44aee8c7c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
