)]}'
{"id":"openstack%2Fswift~704356","triplet_id":"openstack%2Fswift~master~I09301bf81d68357ec006a21dd37e72c56cefffe3","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I09301bf81d68357ec006a21dd37e72c56cefffe3","subject":"Add unittest2 back to lower-constraints.txt","status":"MERGED","created":"2020-01-27 16:50:02.000000000","updated":"2020-01-28 01:32:37.000000000","submitted":"2020-01-28 01:30:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704356-1580175054780-fa858c1e","meta_rev_id":"21b8c6bd159c9cf02c7a4c82dc4d057003c7cc4c","_number":704356,"virtual_id_number":704356,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 01:30:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-27 16:50:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-27 16:50:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 01:30:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bea67436e4111d78bf60b21d58f27b965dad6c6f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-27 16:50:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4308b7fe31b3fa398513847ae63a2c89fbc50ae9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-27 16:50:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6480d60b45c19a7f085f07621118e0bfa10ddb10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 17:58:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/689f338f6f2c4e049e69aebfe06cceba : SUCCESS in 12m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/366f5317fa754dd58c4d61dbad3b1791 : SUCCESS in 4m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2f6c105c930c4955a7d72cbdf0cc97f5 : FAILURE in 55m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6b525c3568a548e1b35222f8fba63454 : SUCCESS in 1h 05m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/caa2d8642120452c930eeee4ef0b9f97 : SUCCESS in 10m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d5b7cc53576f4095a4263d5836790113 : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/020bf7f179734cd89f059da9ad7572be : SUCCESS in 15m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e2b7ff3186554a36abe7bd2af23aafe8 : SUCCESS in 16m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5fa373b56079413f99673db1c1ffa58f : SUCCESS in 17m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/858b4f58c21b41b8810d642bbc383a76 : SUCCESS in 13m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d6f0dd68e7124f2f8516a5ad6bab2e8a : SUCCESS in 23m 09s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/575357c31de74ab59e212469dbb34912 : SUCCESS in 13m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a9f3f026058c48e9b3aa725cfbb0ff90 : SUCCESS in 10m 12s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/407723eca4754c8c940fdda6c8686ac1 : SUCCESS in 20m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dc443a506e80424792d1efc6a8a210f4 : SUCCESS in 11m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5accd3d6728a41b0b98528524da876bc : SUCCESS in 15m 46s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f9f8a5b6d86f4e35adf7896aa940eb44 : SUCCESS in 11m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a7c0e95240434aaa8fd68c2709a0a222 : SUCCESS in 13m 53s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/cf08e8409aa64282ac5a62904a5e8f7f : SUCCESS in 20m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/adb1b89d9337467484939edb908cebbe : SUCCESS in 21m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c22561f95608428fbb652988103dd0bb : SUCCESS in 49m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f62ef897d5f4138bf069958c3778de1 : SUCCESS in 31m 23s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1da6cbda13a14143897f8cf5788ea717 : SUCCESS in 48m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/49460bbb86b5427ba5d1c3f8035165a0 : SUCCESS in 31m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d92cc672f504b46a5a8682b282df1ba : FAILURE in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0961d226247b4178be5bc1a4ee7162d3 : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f9d2930b3d34dfe9d5a2b5a4507842a : SUCCESS in 19m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55098c4f644f492ba720c2c16e4c6843 : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1bfd5205b794a4f37c20d988293ef6ae4decba94","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-27 19:24:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"715728cf648a28b0c04fce0ef4d5a168a806a73a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 20:43:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01b1c0fe59e440a5a68086092d218639 : SUCCESS in 9m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/efda23d0ea504cb58ca216085f1dbd91 : SUCCESS in 3m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c961203f27714c678b3944f853637d1f : SUCCESS in 1h 12m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/838db675dfba4921833b8d5d09fc2400 : SUCCESS in 58m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/37c970bdaa87430892269a369f668f90 : SUCCESS in 13m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/831898b31a5b4b2fb10465fdd44a2fdf : SUCCESS in 13m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ae3435ce90e4aaf91f1715279a12a25 : SUCCESS in 13m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f564d1dd8aab4a0b90ee6bea359a27b3 : SUCCESS in 13m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/68cf3035901245508a20c5c3afdbf077 : SUCCESS in 13m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f0d6004a1a89421fa037a9881d0f4adf : SUCCESS in 10m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3000514a0fad4d698494988f3b1f8530 : SUCCESS in 13m 49s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8599692386104cb0a0b3185057d25601 : SUCCESS in 11m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3e3ca0f1bb81439a85e602176ff36e68 : SUCCESS in 19m 49s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ca58aff7389c4b57a522b9690addf0ef : SUCCESS in 29m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ec70c70bdced49699f1109cb548e8df2 : SUCCESS in 11m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f83765868e884ad0bb17f326a3786504 : SUCCESS in 16m 07s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4f3ccc02e5634c6591467efa4998c8c5 : FAILURE in 20m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bcc59bb755194861af21beadc203692b : SUCCESS in 13m 02s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ea8d615616014a87bbe04ebc7bdaea0e : SUCCESS in 16m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1d57a574d1a64e5d94272d3b7dfe7708 : SUCCESS in 22m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/49f0090755124926b801a354822fd6f8 : SUCCESS in 48m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8b5e52ed42504f59b15ba872366c3754 : SUCCESS in 31m 31s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a1fa2367d3524afdb2a3b4ace6edb8fd : SUCCESS in 33m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/00d66082a70e4ab18ea9c325151994f1 : SUCCESS in 32m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f27932007b704f54b66c10dfbb1bdcff : SUCCESS in 18m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d844fbc775e40bd950e685fa733e012 : SUCCESS in 7m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f67789e2f92d4c10a96163caaefaed27 : SUCCESS in 24m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5e0390f00184872a1539fba51bbac43 : SUCCESS in 1h 06m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3b91a0c1e37ee9ca527333c1c911bbe7092cfe3c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-27 20:49:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"93ba9b02b251e123c3c3123fb014a2288d1818b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 22:44:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/267702427de749e984683be2eeefbe71 : SUCCESS in 10m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ce1fb6450f2b410e8e8a75554b8c23c1 : SUCCESS in 4m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e11fbccb1e4a48f6a1f639c4bde9e9fe : SUCCESS in 1h 09m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f7dfc33d71648d6b7a3e3b8309a7f4d : SUCCESS in 52m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3d12d1d5ebc64ed1a384e7ae1b3700cc : SUCCESS in 8m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/27d37d966edc4e458f4e4d50a2dacccd : SUCCESS in 10m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d2b89d8ace2741ad8319c8efc2509189 : SUCCESS in 17m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/61b6bb58d9b0449e9ab41c9eb86f7093 : SUCCESS in 14m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9b4adba54f424d02a2930bed0d92a44f : SUCCESS in 13m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5d426652870d41a39e58015599d889a1 : SUCCESS in 11m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/861c1d15c3694c41bea949ebe89035ab : SUCCESS in 14m 34s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/700d34dec95146e196ffe4a067ca08aa : SUCCESS in 11m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3de5ac0877a74833abb4855af8adf048 : SUCCESS in 11m 53s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/2a72a8cdc9e44e09bd05ac1831463da4 : SUCCESS in 29m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f86a902dfcb840db856e4d3b0b2f6f7d : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c1544e1cebd04477835ee25c4882d293 : SUCCESS in 15m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2cc1436edaa24a9994a4ace92baee0e5 : SUCCESS in 12m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a1781c52328b4c42ba9e1f2ed3b9a292 : SUCCESS in 15m 20s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/75779162be1543c697fcaffcf1b37fb8 : SUCCESS in 17m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/35d4307f1d024f659cd36584f0a43498 : SUCCESS in 17m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d9ee007482104f8381257e7715fd8a22 : FAILURE in 50m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5db07c2326384e11b608ce0162bb84cd : SUCCESS in 49m 56s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/cb3cac8c581f44c3bc2a6ec93bb43f5b : SUCCESS in 44m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4be74fca45304e21b982a04125fe6450 : SUCCESS in 33m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cab30984acf479bb9bfb08052e53861 : SUCCESS in 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3235b74e20fe45969afa131bdb013bc5 : SUCCESS in 9m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06ad57dab4ac4d789690e66eb8e9e7ce : SUCCESS in 22m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/659439d87314454ba58d60abca56f637 : SUCCESS in 1h 34m 24s","accounts_in_message":[],"_revision_number":1},{"id":"37365731652ddca0ccca248da021f0b376d735d1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-27 22:55:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"183504e8b146aaf218abcee0dfcce4be99b1e7e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 00:15:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/418f0ed3c0194f4eb01505c2553f3eee : SUCCESS in 8m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/02d26daf9c784678af5040293e49fbbb : SUCCESS in 4m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cac3998960df4e7c85acda63bc0ce7cc : SUCCESS in 1h 11m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5783bf02b41640ce8410988b9ea5b7da : SUCCESS in 59m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1e26026b79a546ab93f40f7cab451286 : FAILURE in 11m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dc4f33c7a05c4942a991cd626b872692 : SUCCESS in 16m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/69302c158897469cad126d54c73125c2 : SUCCESS in 19m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4759706e538641019fe079ff5ee39633 : SUCCESS in 17m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0a5b6b67a1574c4c802cdd03a34fddb6 : SUCCESS in 13m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/08b56b376c414d97a9370197dd8468d1 : SUCCESS in 11m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/95b1b69f07264a44b9fdce460c0ced2a : SUCCESS in 19m 27s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/14605da6702944b785ed1fcf85c08b16 : SUCCESS in 13m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4f191e2ff64a4b5badec075b21feac2a : SUCCESS in 15m 21s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a2e669fdfca24988a4cba443fbf76067 : SUCCESS in 16m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6b04721571b74927bf35fcb35f44e8c0 : SUCCESS in 23m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8b644329c1a14c1681210cdfa3822ba9 : SUCCESS in 13m 25s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4cfd88b045ab49469786a91491b8f1ba : SUCCESS in 17m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/baac763c353f41559d363a5b417e6c20 : SUCCESS in 10m 52s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/7700f8c3b62a44018af6d43d3cc25c73 : SUCCESS in 19m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/95587d0b699e4e52ad9bc12b5588eed6 : SUCCESS in 17m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7251909403ac43ee94c9b2f5a6a66e86 : SUCCESS in 51m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4bacbfdc3cf46b9a265437f804837df : SUCCESS in 30m 55s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/cd47d19af7ca47a3b85ce2f638fc7a3d : SUCCESS in 29m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0da209e085d5486fa86149519cf3e79c : SUCCESS in 42m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a3081ebc78840d0ac86f98b9da069f7 : SUCCESS in 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4cb1295808943f19afa4f8e06d9b358 : SUCCESS in 6m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/97569c52188c4499bfd3a29d57a82ab5 : SUCCESS in 17m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae1556c157a3474199edacc15d514c6d : SUCCESS in 1h 13m 03s","accounts_in_message":[],"_revision_number":1},{"id":"915c6b9a63c6362b6a854051aed202711f8dd01a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 00:16:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d08086a2962ba4020cd400aead6496c900c5251c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 01:30:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2336e54a5a2245ef99947ed54e507017 : SUCCESS in 8m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/67d6932dc9ea4e85b6c565505e2c4ab6 : SUCCESS in 4m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f6b9349e91124a0d9e20e0a7de656dc2 : SUCCESS in 58m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9e959fd0d7c42dc8e4ed18cc7c6f3e8 : SUCCESS in 48m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/39533239de0541db9d0355ca0d060db9 : SUCCESS in 9m 45s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/097efd1c1d204e3194f55e5bddd03725 : SUCCESS in 11m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f70905fd517149bd9b4e9576a725b8a7 : SUCCESS in 11m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4731cc74dd924801b26a7de9fff8c4e0 : SUCCESS in 15m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/28dc6af5d9a94eb89cf795f8ae134706 : SUCCESS in 22m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8205bb3e67f749b2a926d62e951aaa2e : SUCCESS in 13m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/13b86ed4ff8345a1ae16326a39ad9cc3 : SUCCESS in 12m 10s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a8d2a0eb960d4480b9406f9ab905ce93 : SUCCESS in 13m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/de6b768e036c440bb67ebc10b56ffe90 : SUCCESS in 19m 44s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/102a7bec202b4c39a6e265dd58701cd5 : SUCCESS in 19m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/12d655c6302f4e6a9da5110cbc4cb7e4 : SUCCESS in 13m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b55e4f30cf7a42c0bb17b614e3c95af2 : SUCCESS in 14m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c5fe5cadda7a4663b0591afe70369c3f : SUCCESS in 11m 33s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0dbd818a60c9418abf4f4a5cfe4b5a92 : SUCCESS in 19m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b5954c662f6e4cc3821b80f9c354518e : SUCCESS in 47m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6821df9dd1b24818b23a01cc3f25f17c : SUCCESS in 29m 06s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8ee8a483de8f4da79d935c4d58ef40b4 : SUCCESS in 31m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/72e4055021514d1ba49d3855e0418b11 : SUCCESS in 29m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55e14048af9443eea88e1554f5afb11c : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/077d34d0267b4e1c8c3e6d90f1cdfa50 : SUCCESS in 5m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b7b76433622470e8a4916847c43cc55 : SUCCESS in 1h 11m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6b92f1ffc79b5348280d752c72db5adac25e9f0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 01:30:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"21b8c6bd159c9cf02c7a4c82dc4d057003c7cc4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 01:32:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20504b1721444f3baaa3cf4c66a6360b : SUCCESS in 1m 20s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/366238cd26a94508954d7bdca0b11bb9 : SUCCESS in 56s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/85d4c1c2521b4d049a886ce358d65a1a : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51c9b13082b70bde91e553c5803bb2f403657b34","revisions":{"51c9b13082b70bde91e553c5803bb2f403657b34":{"kind":"REWORK","_number":1,"created":"2020-01-27 16:50:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/704356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/704356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/704356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/704356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/704356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/704356/1"}}},"commit":{"parents":[{"commit":"742835a6ec12944f606d6cd65ca7f18fdf78a09c","subject":"Merge \"New Object Versioning mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/742835a6ec12944f606d6cd65ca7f18fdf78a09c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-27 16:47:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-27 16:47:28.000000000","tz":-480},"subject":"Add unittest2 back to lower-constraints.txt","message":"Add unittest2 back to lower-constraints.txt\n\ntesttools (required by os-testr) still requires unittest2; until\n\n  * testtools no longer requires unittest2 and\n  * we bump up our minimum-supported version of testtools to a version\n    that doesn\u0027t reference unittest2,\n\nwe should continue listing unittest2 in lower-constraints.txt (just in\ncase they ever have a new release).\n\nAdmittedly, that seems unlikely at this point -- unittest2\u0027s latest\nrelease was back in 2015.\n\nChange-Id: I09301bf81d68357ec006a21dd37e72c56cefffe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51c9b13082b70bde91e553c5803bb2f403657b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51c9b13082b70bde91e553c5803bb2f403657b34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
