)]}'
{"id":"openstack%2Fswift~704481","triplet_id":"openstack%2Fswift~master~I60d4ad62c7d14ed15e114ed95102a2368c48f9f4","project":"openstack/swift","branch":"master","topic":"704481","hashtags":[],"change_id":"I60d4ad62c7d14ed15e114ed95102a2368c48f9f4","subject":"Authors/changelog for swift 2.24.0","status":"MERGED","created":"2020-01-28 01:10:32.000000000","updated":"2020-01-31 05:59:02.000000000","submitted":"2020-01-31 05:57:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704481-1580450244599-798fc2db","meta_rev_id":"bcfa8d8f90590d0c1f5cfe5a7be2eefc30982209","_number":704481,"virtual_id_number":704481,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-31 05:57:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-31 05:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-31 05:18:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-29 06:09:31.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2020-01-30 03:09:45.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2020-01-31 05:57:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76d1d2f7c4ac67e82a4a3a4e8b18ae73a2eca73b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-28 01:10:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f284d35f19758a7bccb98940a190bf854fad4b3e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-28 01:11:50.000000000","message":"Patch Set 1: Workflow-1\n\nI\u0027m going to rebase as things land, but may as well get my draft out there.","accounts_in_message":[],"_revision_number":1},{"id":"836963cabcd27c0814912061b0146867ac2b9b6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 01:27:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a757513570748949ea100505318ef4d : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aec0811ade344705bb91f3aa63de20ee : SUCCESS in 8m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66317d5ec5f748369e5a838ffb7ac42a : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fd2adab2e8a542229b6fa4565cb02fa3 : SUCCESS in 9m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f65ac79433c142ebb8e465d332319c70 : SUCCESS in 14m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fa28fbb17b194d03a1427736cf9db2a7 : SUCCESS in 14m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/21adcba88ef1468d9490583d82b7e4bf : SUCCESS in 15m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc1be0186d7c4010998f64ccffad8ed0 : SUCCESS in 14m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e3cece0e05465081a763f5e6276dd6 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1095000e0be14edb17792cfce70468eb472240b9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-29 04:56:57.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"34a076514b160a51f89ca6fb31202033b61c9040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 05:20:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f032d2264e4e8ab5955f2c46c4fd0c : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d0ef894c790411aaf89f131bbd22a56 : SUCCESS in 9m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/29be80cc9d984f68bcf471fdec7a0831 : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ac82af3211f54dbda73d4673c4645d14 : SUCCESS in 8m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1459a33865b04963a74ea71d137e07c1 : SUCCESS in 12m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d3246ba87f8447bcb0126092e4292cc4 : SUCCESS in 23m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/726cbfac960f46b1879259b449709aa0 : SUCCESS in 22m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f53acdb2836407795481825ec9cba42 : SUCCESS in 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a54c15e2fbcc479e848d4fded841db3a : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":2},{"id":"5d06a8acd67ee9ad8b83fd4e66a135c799b2cb67","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-01-29 06:09:31.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)\n\nLooks good! A simple question in-line. And will look at some of the merged patches just to confirm. Gonna be a great release!","accounts_in_message":[],"_revision_number":2},{"id":"4f051d05b5de2f4002611b61750e4c86a38ed8c5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-29 17:54:07.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1be272399c8440e3b3b3a2e87cad95675f79a593","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-29 20:13:46.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"114b56ac54d6a53a2fd386ec0ad35f803b11084e","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 20:59:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/984f874551ec44a9a9343b5a8217ae7c : SUCCESS in 18m 20s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/670b09a0ac2d4b7fa5d7c09f8de8fd9e : SUCCESS in 23m 22s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/39c3424e4d8b4cc4a081def9e808d8e9 : SUCCESS in 20m 44s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/ab9c9a3a1e5343d0a5cb18c6f6c09871 : SUCCESS in 20m 17s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/62cf32434e304b5b80bd92eca0a2fc1c : SUCCESS in 23m 14s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/87cb85fecba84929943c41dc5813bc7f : SUCCESS in 19m 18s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/6ef5804421614aaa87d294d1f6112fb9 : SUCCESS in 19m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3ee65a6feeaf9ca3e26b549ab0d51ef6ed5cbfd3","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-01-30 03:09:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3602b1fc64280785fc2b2559aedb2fb47abd632a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 00:17:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3d79b53c197d931c1bf46476d957c0564d1b43ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 00:36:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8785839d849f46b7a0f1c3e941a6bf9c : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b85429a97084062af0c1adf03e23e14 : SUCCESS in 7m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/274e3a3ea59f4e22b7ef9b36d4b7fb82 : SUCCESS in 9m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/95cc0d9f47d042a794299ed05797a600 : SUCCESS in 12m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/49b181eb5d704bc6a89b98e8da15553b : SUCCESS in 15m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/63eda059c5004f77bb54041b5a927aed : SUCCESS in 15m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b5fbcaf0a17f4ee1b07d8af4ca7fa7ff : SUCCESS in 16m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c33452588a8848bc872d159ff31c7485 : SUCCESS in 15m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7345ce40739b4f86b0d3f4efd6f95e2e : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a6ce6dd06b243fea1aaad23b2b4b0bf : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":3},{"id":"17e782bb4afa67be41580999cf470d2b10ea110f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 05:18:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"43e96fc76658566be223dc6acccb9fede7d63b3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 05:18:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fdf91a4f059d3fba0fff1e051c5698b323011240","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 05:57:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1e45e559112bdfc1ff4ed346d7794639d057c99a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 05:57:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a7edcf0bce4574b420e6879b62c4da : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74808e65d0404ca8a793236e3fdf9f79 : SUCCESS in 7m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8ba73a5f47d646bc81c397615f92187b : SUCCESS in 9m 44s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d587eb1d054e4394ae7ac3ff61fc260d : SUCCESS in 9m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d6c162227083413ba18a5f29fbc2f357 : SUCCESS in 12m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/99757a6ee9a1459d80063839e3238c33 : SUCCESS in 12m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f090b213f3ef411493c4c4ba3e85c0ed : SUCCESS in 11m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3b9a2ab067f0448f86125d5eafba304c : SUCCESS in 11m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/975495aee3e14d0287853619401a9d29 : SUCCESS in 9m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dc4daaf974224a098ce73fc2cbf2dbbc : SUCCESS in 11m 59s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8c6601b96e234b73aa3d654834acc088 : SUCCESS in 9m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ce547c803cc64eedb7f7dfb92cbfa29d : SUCCESS in 10m 29s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/054f91316c7543d590972d947ee5a6c3 : SUCCESS in 14m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a1189a9860d34546b38af022834ec4fb : SUCCESS in 10m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7ad1919992734755a66c915aa5631333 : SUCCESS in 9m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/42b1bda2b4884a0a9e80f22a91d73dd9 : SUCCESS in 10m 42s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/eacc20b8af6f485aab479cc1ebacc804 : SUCCESS in 15m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30c84b2119474020b8d3aded70c63093 : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f45b505cf6d040c1ab60978797d8193f : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":3},{"id":"1fa3055e3796126952f84cabfcb81f1c19da91de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 05:59:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9491a9e9c1bc421890faeb404d731c2e : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fbe939b37384dd18f696c23ff1a7de5 : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4844a036204440fdba0b4ca1860a4631 : SUCCESS in 46s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/92d77e90370d4452a414e23f9cceee5b : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5cd854b783ce80356ec62e709f0817cbe699c8cd","revisions":{"2d589bc44f6afb5fc94135ef70f92fa5f116d36f":{"kind":"REWORK","_number":1,"created":"2020-01-28 01:10:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/81/704481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/81/704481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/704481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/704481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/704481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/81/704481/1"}}},"commit":{"parents":[{"commit":"742835a6ec12944f606d6cd65ca7f18fdf78a09c","subject":"Merge \"New Object Versioning mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/742835a6ec12944f606d6cd65ca7f18fdf78a09c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-28 00:11:55.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-28 00:39:43.000000000","tz":-480},"subject":"Authors/changelog for swift 2.24.0","message":"Authors/changelog for swift 2.24.0\n\nChange-Id: I60d4ad62c7d14ed15e114ed95102a2368c48f9f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d589bc44f6afb5fc94135ef70f92fa5f116d36f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d589bc44f6afb5fc94135ef70f92fa5f116d36f"}]},"branch":"refs/heads/master"},"5ca0ed031094f5947e4e4555b358da46962e066a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-29 04:56:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/81/704481/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/81/704481/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/704481/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/704481/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/704481/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/81/704481/2"}}},"commit":{"parents":[{"commit":"3010d318043e8a2fbd1c6bc8595e185b2201e247","subject":"Merge \"s3api: Implement object versioning API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3010d318043e8a2fbd1c6bc8595e185b2201e247"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-28 00:11:55.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-29 04:56:49.000000000","tz":-480},"subject":"Authors/changelog for swift 2.24.0","message":"Authors/changelog for swift 2.24.0\n\nChange-Id: I60d4ad62c7d14ed15e114ed95102a2368c48f9f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ca0ed031094f5947e4e4555b358da46962e066a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ca0ed031094f5947e4e4555b358da46962e066a"}]},"branch":"refs/heads/master"},"5cd854b783ce80356ec62e709f0817cbe699c8cd":{"kind":"REWORK","_number":3,"created":"2020-01-31 00:17:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/81/704481/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/81/704481/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/704481/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/704481/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/704481/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/81/704481/3"}}},"commit":{"parents":[{"commit":"f9cdb7b6fb3bb39f4e6733c0f1f373ab6eca1c56","subject":"Merge \"Add py38 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9cdb7b6fb3bb39f4e6733c0f1f373ab6eca1c56"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-28 00:11:55.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-31 00:17:05.000000000","tz":-480},"subject":"Authors/changelog for swift 2.24.0","message":"Authors/changelog for swift 2.24.0\n\nChange-Id: I60d4ad62c7d14ed15e114ed95102a2368c48f9f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cd854b783ce80356ec62e709f0817cbe699c8cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cd854b783ce80356ec62e709f0817cbe699c8cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
