)]}'
{"id":"openstack%2Fswift~704646","triplet_id":"openstack%2Fswift~stable%2Ftrain~I36f0954fd9949d7d1404a0c381b917d1cfb17ec5","project":"openstack/swift","branch":"stable/train","topic":"bug/1669888","hashtags":[],"change_id":"I36f0954fd9949d7d1404a0c381b917d1cfb17ec5","subject":"s3api: Better handle 498/429 responses","status":"MERGED","created":"2020-01-28 17:57:40.000000000","updated":"2020-01-29 22:11:34.000000000","submitted":"2020-01-29 22:09:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704646-1580335751531-a46f390a","meta_rev_id":"a59c079c45f038c7fcdaf82b9631f8e574d9f40c","_number":704646,"virtual_id_number":704646,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 22:09:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-29 17:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-29 17:56:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-29 22:09:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5dc28d21cca4d05d570878958d67b037d4c4e435","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-28 17:57:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d7dd032178c7ede6a35db5705f864c14fe3d00b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 21:52:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88e54b19a3604d70be21536ad4092b2a : SUCCESS in 7m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/44606e269274433e920e0f0afc3108e8 : SUCCESS in 1h 11m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5d18a20248c1441ba61d9dc7b2dfe6ff : SUCCESS in 54m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e8d27a4f1f1441d2a8cdc53551c16fbe : SUCCESS in 9m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/deb7f99b19fb40aa84448840a808ac4c : SUCCESS in 16m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/777b94eee8964da8b641231bd9df60e8 : SUCCESS in 14m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/59f2fed9f0e14c33a0239c7719e542a4 : SUCCESS in 17m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b16a590e09b0421a89b8cacb7b530830 : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/da376bb21639456b923b37be8a582fbb : SUCCESS in 11m 41s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/e57fa18e33c64fe49bb9af8e6d15c222 : SUCCESS in 11m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf832380bd8f403bb6d93dafe5f51384 : SUCCESS in 12m 47s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/59ac61ebf13049cca2dcfcfe430416a3 : SUCCESS in 13m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f6fe2502baea4e5b84da90eefded68ef : SUCCESS in 12m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/21598b17c13f47cf97e98cd406c22e19 : SUCCESS in 11m 59s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/dc43904890f543a88340c56fcd2f59e7 : SUCCESS in 12m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7158012fcf64451892794856888c2ff9 : SUCCESS in 12m 12s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/07bb10e47b8f412b81d5c120196287ca : SUCCESS in 15m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4c3cceacc83240a9ae5d8537a30661a1 : SUCCESS in 19m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0d53d130c8fa4e3fa5c95b1a7dc676c7 : SUCCESS in 32m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5daf4d7bb1ec4ed8bd6eeadec08a7b64 : SUCCESS in 29m 35s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c9f065eec2394296a5b408c9768bfba8 : SUCCESS in 30m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2d6b82e30a624dc5910aefca07fdd5bc : SUCCESS in 32m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14a1087bb9c840ae922a77c7626eddc2 : SUCCESS in 19m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c84194cdd604f339d93c26c19c95af7 : SUCCESS in 8m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ed4e96b4e645418fa3081f80b7e22340 : SUCCESS in 21m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0badb67e248246a497a1e7c73703ed3a : SUCCESS in 1h 21m 11s","accounts_in_message":[],"_revision_number":1},{"id":"fcfb5f000c78d351b8209c3fcf365326b73ecf4f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-29 17:56:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d6731f9276f09acc9bede26c38d31125873f2314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 17:56:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1a28a84f28e9be3e1804120fca32db51a58a094","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 22:09:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9105c07904ed4c478797195fac1d2eec : SUCCESS in 7m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/400d7ed77b7e4e1c83a1f06eb468c24f : SUCCESS in 1h 10m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bc69356f18984f399f484baaba06d2d4 : SUCCESS in 1h 00m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9e45768e09664748a816d176b269a4d0 : SUCCESS in 10m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f5e0b23ca9d04fd788e7e71d9e69ad05 : SUCCESS in 15m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7778fd2d25df42339d15fed2ff97d7c9 : SUCCESS in 14m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a81d69252371423aab4ff8e4190882b4 : SUCCESS in 12m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/461bdf37ab2740c5ac23104f23ac30f1 : SUCCESS in 10m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2a0439c6e7c94b34a2a5e4b12549c99d : SUCCESS in 14m 33s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/695789f7c9e14961b09ec8935182b52b : SUCCESS in 11m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/89c57b5fbf5d47b6897be2195c277f95 : SUCCESS in 12m 56s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/57299e7d4d1a4f018b467c800ef05032 : SUCCESS in 15m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d6b368fe6a4b405eb60349fde99bebd4 : SUCCESS in 21m 54s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/80df53d2f0754cf9b58e31d6d0ccff02 : SUCCESS in 11m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d8a7cf5b84994ea69fb280b2dece0a06 : SUCCESS in 12m 21s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/707b3f39a8334d46bfedd83772eb72bb : SUCCESS in 16m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/952311ba70d347feba6296cd447c5960 : SUCCESS in 29m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25d3c97e770c4ca7a8787a4484ec62aa : SUCCESS in 45m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/3aea12b5d76e46458f74f3bd6cc7ec8b : SUCCESS in 37m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/41c06a0a830b43c0ab0f4690dbd9fcbc : SUCCESS in 37m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25118ae73f0640d3b0379c691a8636ff : SUCCESS in 18m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/298b8c9f4b624e8e8f221feb702be88f : SUCCESS in 7m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3beb0446a8ec4ebb82c4830ac3de4d6c : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":1},{"id":"108dbe3447d50b49b570414922ab811fe481ddf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 22:09:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a5f172e01e045143cf0282fcf28bcd3295686ceb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 22:11:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a38840e8dd084fe78570bdf45ca5e947 : SUCCESS in 1m 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f06f6d67f8774df699978bf9c5cb82bb : SUCCESS in 1m 26s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"092054da632037cd3632d8a2c85b7d2015998fff","revisions":{"092054da632037cd3632d8a2c85b7d2015998fff":{"kind":"REWORK","_number":1,"created":"2020-01-28 17:57:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/46/704646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/704646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/704646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/704646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/704646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/704646/1"}}},"commit":{"parents":[{"commit":"e413c650e385ba44523e99e023ada0a44aee8c7c","subject":"Allow bulk to fwd some headers at tar extraction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e413c650e385ba44523e99e023ada0a44aee8c7c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-05 17:42:33.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-28 17:57:24.000000000","tz":-480},"subject":"s3api: Better handle 498/429 responses","message":"s3api: Better handle 498/429 responses\n\nCurrently, they just 500 as an unexpected response status. Much better\nwould be S3\u0027s \u0027503 Slow Down\u0027 response.\n\nOf course, that\u0027s all dependent on where you place ratelimit in your\npipeline -- and we haven\u0027t really given clear guidance on that. I\u0027m not\nactually sure you *want* ratelimit to be after s3api and auth... but if\nyou *do*, let\u0027s at least handle it gracefully.\n\nChange-Id: I36f0954fd9949d7d1404a0c381b917d1cfb17ec5\nRelated-Bug: 1669888\n(cherry picked from commit f33c061ae92f33ea467c58749d3f15d2b1cc942c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092054da632037cd3632d8a2c85b7d2015998fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092054da632037cd3632d8a2c85b7d2015998fff"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
