)]}'
{"id":"openstack%2Fswift~704662","triplet_id":"openstack%2Fswift~master~If23592855db5f5bb0ec1e7c679de15769fd86871","project":"openstack/swift","branch":"master","topic":"docRelinker","hashtags":[],"change_id":"If23592855db5f5bb0ec1e7c679de15769fd86871","subject":"Add a note to the part power increase documentation","status":"MERGED","created":"2020-01-28 18:52:37.000000000","updated":"2020-01-29 03:54:27.000000000","submitted":"2020-01-29 03:52:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704662-1580269960494-bbef5f2b","meta_rev_id":"26dca8467e3f268b9bde02d62dae611c6179f386","_number":704662,"virtual_id_number":704662,"owner":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 03:52:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-28 18:54:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-28 18:54:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 18:54:27.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-01-29 03:52:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32213be92fc78d760f2288e6c094184e85451e55","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-01-28 18:52:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd7808841811607da18851a0f460c272784a0bd8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-28 18:54:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"672b1155ea3c2e7af98444fac90d6d839f7df33e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 21:32:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b1cf52b5c2a4394bced2029a0cbcfec : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4688d70b93c983e75dff0afc05189f02e6270638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 21:32:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"058d82c6805f272ca3d5df917a411e57eea9a5ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 03:52:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d19020b5d584cafbcd29ae3b1f205e6 : SUCCESS in 9m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7a7b53c1ec5d413d9637376c7d8d6852 : SUCCESS in 10m 56s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/258121198b3140f0a8e3f31174ed14b9 : SUCCESS in 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"155b9481151b6b38196051a5be972efda622fa5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 03:52:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"02dbf6ecbc061180645de295cda6cd1e90c58c45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 03:54:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7912e3b4486048f08dafa569bcc9d6d4 : SUCCESS in 1m 08s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4b68fad4f41b4b1ba3f567e62ec23504 : SUCCESS in 57s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6ee04aa0af7d4721a62deba52eaa8ce0 : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebaf154fc4d7f91ff704ca2367556c137ed52137","revisions":{"ebaf154fc4d7f91ff704ca2367556c137ed52137":{"kind":"REWORK","_number":1,"created":"2020-01-28 18:52:37.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/62/704662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/704662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/704662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/704662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/704662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/704662/1"}}},"commit":{"parents":[{"commit":"742835a6ec12944f606d6cd65ca7f18fdf78a09c","subject":"Merge \"New Object Versioning mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/742835a6ec12944f606d6cd65ca7f18fdf78a09c"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-01-28 18:51:05.000000000","tz":-300},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-01-28 18:51:05.000000000","tz":-300},"subject":"Add a note to the part power increase documentation","message":"Add a note to the part power increase documentation\n\nSpecify that the relink command must be run as the same user than the\ndaemon processes to avoid creating files that are not manipulable by the\nserver/replicator/...\n\nChange-Id: If23592855db5f5bb0ec1e7c679de15769fd86871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebaf154fc4d7f91ff704ca2367556c137ed52137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebaf154fc4d7f91ff704ca2367556c137ed52137"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
