)]}'
{"id":"openstack%2Fswift~704923","triplet_id":"openstack%2Fswift~master~Ifcb5013a5728d93cf5491fbff81b2677450698e6","project":"openstack/swift","branch":"master","topic":"dockerhub_description","hashtags":[],"change_id":"Ifcb5013a5728d93cf5491fbff81b2677450698e6","subject":"Dockerhub description of saio image","status":"MERGED","created":"2020-01-30 04:38:13.000000000","updated":"2020-01-31 18:49:29.000000000","submitted":"2020-01-31 18:47:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704923-1580496454730-f323dc7b","meta_rev_id":"f1bff65f893861ba5447f2281cb2e3bc5311be22","_number":704923,"virtual_id_number":704923,"owner":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-31 18:47:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-31 17:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-31 17:32:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-31 17:32:00.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-01-31 18:47:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1186f5dcc0aced6e2b319710a1698a5c6736285a","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-01-30 04:38:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce28e462c66bddc5dc31bda2f51afabcc8846a90","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-01-30 04:41:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17f33fc236afb807e410513294ea2579518c3811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 06:08:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de46bf623ac40568d8d0ed50ba87e5a : SUCCESS in 8m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/35492727f5444bdf909b91b3306d7030 : SUCCESS in 57m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/96a14d3f27f64c04b265d6a0d6a3db97 : SUCCESS in 52m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/34ed890699384b50a9445456824f89f4 : FAILURE in 10m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/08e4a3b417474e9891b48dcaa76eba04 : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aa1931282042478b8f4d59f79c68487f : SUCCESS in 11m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ec0eb4f5b524a9a8dd45accfcd52169 : SUCCESS in 14m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ccebef435bde40b3aa6e5ceb25ffea39 : SUCCESS in 11m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e18ae5f9921248d0ab05127e66e939ad : SUCCESS in 9m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a86c6f8e2024975bcb6390206af73fa : SUCCESS in 12m 14s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ab70ae99222e4139b1c8c7be8634bc5d : SUCCESS in 9m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c4b0040950184878b22f15d39757ed23 : SUCCESS in 10m 26s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6629b5b2ed9b4e29b8f37a2448580cb6 : SUCCESS in 15m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4e7d7f5c511a42f7a22518a16d30260d : SUCCESS in 10m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/71a423c25f9c4bc398d8e5f920051534 : SUCCESS in 13m 50s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/630e3a6491294de2aceef1b1acb2c979 : SUCCESS in 9m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/75bfc5ae53de459bad370dfa7683d734 : SUCCESS in 10m 43s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6691e064ac564b25be6f179479ededfc : SUCCESS in 15m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0c749d9c342a441f96d077cdb812173a : SUCCESS in 19m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8f51b7d137b34d83bd67804fc266ae36 : SUCCESS in 43m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ca74d60f361a41d189e2f3683f782222 : SUCCESS in 28m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7e23edcded124abdb6e043bfef875e10 : SUCCESS in 28m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ceafff5101b74526bdbdd82e4cd75526 : SUCCESS in 31m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1902055d8b0c43929afed6296b5fb559 : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a0dd1acc2594413a2952fec5a886deb : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0bb6d570fe8f4713971339bd835c064b : SUCCESS in 17m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37841e1650814e2494bb49b481823f6e : SUCCESS in 1h 18m 33s","accounts_in_message":[],"_revision_number":2},{"id":"e8b53edf47ab8211d487393dfdc6658c091fa537","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-30 18:03:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"efa9450680c8e8ff55c3cc0a43b39bce315f4d41","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-01-30 20:17:19.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"614d64dd944ff351b46cafb64efd0674fa690259","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-01-30 20:17:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"881ad03b08f83156438bd8cfaf0019d62a72dcdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 21:27:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7177c38c152049c1831e695d07154496 : SUCCESS in 11m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9202c966ab2d4d1488b7018e70fbbee7 : SUCCESS in 59m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/264ff46d66754200ae12c1eba7fcc75d : SUCCESS in 54m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a4d6e8ff10d24a7296d0a045f248f7ae : SUCCESS in 12m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/14c429cf816043ab9edcca7b27854727 : SUCCESS in 9m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/81895a3e874948fba9100cab287b3f1c : SUCCESS in 15m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8efa854c7754de38dc7727b5eabf3b0 : SUCCESS in 14m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7048f62689994bfebc703bec6f6e4ae1 : SUCCESS in 13m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4a6514d434e6472a9c8163fe094d972a : FAILURE in 16m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c4dacbcd86a64332a1fc183958281169 : SUCCESS in 11m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/87ddab185a29487790b740f2385eb185 : SUCCESS in 12m 37s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/920d380698874086b33e20e925b8945d : SUCCESS in 10m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b30aae408e541fd9b38ab7fd063c11b : SUCCESS in 13m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/409f4f55584845cfa78e1e5644b03fb6 : SUCCESS in 16m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d4caff57ca624fc88167a67c2df2d136 : SUCCESS in 13m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0637ffa7b534454c8426e898c08d7604 : SUCCESS in 15m 23s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3e221f3a5fa344e0ae52b13c6deaaaa2 : SUCCESS in 12m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2e54a55c481444feb30b76abffc195a9 : SUCCESS in 14m 44s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5b20345dc145499483129f92053786c9 : SUCCESS in 16m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/40ffd246c2bf4624aafed2f165fb4405 : SUCCESS in 24m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7b6e1238cdc3443dabd7febdd160af84 : SUCCESS in 46m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4863ea55fbda4162a679ca4a13d7f6fe : SUCCESS in 33m 36s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1099ec897c9f4d0da2d791251142fe9b : SUCCESS in 36m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6e3b61b3dc3a4ae0a5c19fa1a3fc7615 : SUCCESS in 31m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a7c2f0b85b9492a9aadfd348615e13c : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e5a2f4201244166b8cb9c207b1a07ab : SUCCESS in 9m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2cc8b54bfbe94e7aa8a7dc9015465b97 : SUCCESS in 20m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8da8a31a06f54d9fbaaff66e4154fc7c : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":3},{"id":"21f9be52d3a5e497f92459951f77fd11dd6979cc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 00:34:15.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"681850beb4b32f5bfe4d515d97b185f9b3a2be31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 01:46:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d38a26c975ac49cabae3c26810f032dd : SUCCESS in 8m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e799ca7aeb95404888360f9e3d8c60a4 : SUCCESS in 1h 05m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d76c27cada1e4b6d95380b47caf1df0f : SUCCESS in 52m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b49b05be41c47c8898f371ad270ddfb : SUCCESS in 8m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f709d7249aa64651ab63341e48c3e47e : SUCCESS in 9m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f960029a28934a048c848e2b3520e8f6 : SUCCESS in 11m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1d2cbdace750493a90630264a54d7d17 : SUCCESS in 13m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9533cbae14c94f5d84b8899b689f5dd7 : SUCCESS in 12m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ef07fe44fb6c471a95e161795a9b13b2 : SUCCESS in 12m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/62ac7d2304af4c28bc849644c5029245 : SUCCESS in 9m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f885ffb89c6c4c379fd7c21e733dc345 : SUCCESS in 11m 52s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a8bbcf5321e94429aebbcf95a1b60e1a : SUCCESS in 9m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/199d82ca2859494b8aa2d46e187395d0 : SUCCESS in 11m 34s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e06cab9eb9dd4250aa034a95bf9c60dc : SUCCESS in 15m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2624f54a8d5f469b89a4588d188e5a25 : SUCCESS in 9m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5df93d66e9f24786a19de74257069f76 : SUCCESS in 12m 56s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1c47dc7ccf9d478abdaa55e16f7232f7 : SUCCESS in 10m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3c357faf769846589daaf904d30d678a : SUCCESS in 12m 39s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/25b249ffca354410a752d5a620a9a1d5 : SUCCESS in 14m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/98cb825ef3f94c07a55bba6dafa60582 : SUCCESS in 24m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7e46c28f1440422894471f30f4d3a928 : SUCCESS in 46m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/775f4894bc5245a7aaa3fa711348d81c : SUCCESS in 33m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4654d04b8fc241a9ad1f4048dc677340 : SUCCESS in 27m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95129566a9c142b8938973e819f9a058 : SUCCESS in 32m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fd717a4bae64d1db1ba0ee400fb8940 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5b77daeb7c4e6db23ef2d1b11219ea : SUCCESS in 7m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd58a29d23fa41269c1a4e68e12969fb : SUCCESS in 19m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30f1217abf3a447a8e8c89ed81562e10 : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":3},{"id":"ea945c5ee772e8ec94c768cbe554f1c229c9d993","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 17:32:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0d5bdf3fe963f39eaee67de40caff3cecaa52c6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 17:32:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ce10e80d2cc468097a2f109a0a88d217eca99fc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 18:47:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4eea6df006a4c55a9c11ed7061e3f0e : SUCCESS in 9m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/083e83282ae848319df536c0417b4360 : SUCCESS in 1h 07m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3199bc74b924637a52a7d71b65f01bd : SUCCESS in 55m 18s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ce67129bf9754afbb810e48ec6edac7d : SUCCESS in 12m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/84813349da3e494aa7f7a057d55364d0 : SUCCESS in 14m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2f004bfdd5aa409187d94e6961d4b3b4 : SUCCESS in 14m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f5d4262d89044968827bd15ef1905675 : SUCCESS in 12m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9709d38765314f07b5dc3e88a378f6bc : SUCCESS in 21m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/be91c740030c47839291d9d60f4e1185 : SUCCESS in 18m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2ace9861985a43b38316909e1f451b22 : SUCCESS in 10m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/710635243334470881295c17a5920146 : SUCCESS in 15m 52s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b5934c44563e44a396f2dcb8241f253d : SUCCESS in 11m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5fa5600105ff443f92e92d00e8459066 : SUCCESS in 14m 17s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/418af4a14ea4449aa60faa1e5dcc7e11 : SUCCESS in 19m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b97db4ab5a314900939fc635c7c9b370 : SUCCESS in 12m 11s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e9059488435b468ba9517cea360426cc : SUCCESS in 12m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fe7ec534622048aebf512fb502818632 : SUCCESS in 16m 40s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6f45ee9fdf564e7e99e6502ecb2b1a90 : SUCCESS in 17m 09s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5a6f318b14fd4575953c4c41f0d327b0 : SUCCESS in 48m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/96e0bed63ee748648e38a576f4756add : SUCCESS in 42m 21s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/bf9835820c914442b14660aece05381d : SUCCESS in 29m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9ad5255d6a5c4f34806763ddafd24622 : SUCCESS in 33m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bf6fdf207724eb39d42ff215d4bf1cf : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c998c7ec728a468d9745b567615ca368 : SUCCESS in 6m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/078c52670e114a62b833dc18911998ce : SUCCESS in 1h 00m 33s","accounts_in_message":[],"_revision_number":3},{"id":"581dd7a426c4ee165a5603b6a7cecbdf468a48ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 18:47:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4cfe53b4b3caef98329bbe6969c87b771bb84d50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 18:49:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd3a37b313940ce86d85abea32ebbac : SUCCESS in 1m 27s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7bd0666d14854c43906af239d8d46956 : SUCCESS in 1m 07s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/108f81305dc54160bcca901177807c82 : SUCCESS in 1m 04s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c7cb34ad61feb5b830ccb79437b32965916220ef","revisions":{"c880cdc62400de153894d864b109fe73fbf9f35f":{"kind":"REWORK","_number":1,"created":"2020-01-30 04:38:13.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/23/704923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/704923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/704923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/704923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/704923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/704923/1"}}},"commit":{"parents":[{"commit":"5e70039df5672150ec971504921f2e8255348c4d","subject":"Merge \"Bump up ceph tests timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e70039df5672150ec971504921f2e8255348c4d"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2020-01-30 04:33:41.000000000","tz":660},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2020-01-30 04:33:41.000000000","tz":660},"subject":"Dockerhub description of saio image","message":"Dockerhub description of saio image\n\nChange-Id: Ifcb5013a5728d93cf5491fbff81b2677450698e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c880cdc62400de153894d864b109fe73fbf9f35f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c880cdc62400de153894d864b109fe73fbf9f35f"}]},"branch":"refs/heads/master"},"b5f90fcc8d7f82c2b4dfe6c2f8a77f5f13a2ca46":{"kind":"REWORK","_number":2,"created":"2020-01-30 04:41:30.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/23/704923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/704923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/704923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/704923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/704923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/704923/2"}}},"commit":{"parents":[{"commit":"5e70039df5672150ec971504921f2e8255348c4d","subject":"Merge \"Bump up ceph tests timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e70039df5672150ec971504921f2e8255348c4d"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2020-01-30 04:33:41.000000000","tz":660},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2020-01-30 04:41:11.000000000","tz":660},"subject":"Dockerhub description of saio image","message":"Dockerhub description of saio image\n\nChange-Id: Ifcb5013a5728d93cf5491fbff81b2677450698e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5f90fcc8d7f82c2b4dfe6c2f8a77f5f13a2ca46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5f90fcc8d7f82c2b4dfe6c2f8a77f5f13a2ca46"}]},"branch":"refs/heads/master"},"c7cb34ad61feb5b830ccb79437b32965916220ef":{"kind":"REWORK","_number":3,"created":"2020-01-30 20:17:19.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/23/704923/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/704923/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/704923/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/704923/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/704923/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/704923/3"}}},"commit":{"parents":[{"commit":"5e70039df5672150ec971504921f2e8255348c4d","subject":"Merge \"Bump up ceph tests timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e70039df5672150ec971504921f2e8255348c4d"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2020-01-30 04:33:41.000000000","tz":660},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2020-01-30 20:17:07.000000000","tz":0},"subject":"Dockerhub description of saio image","message":"Dockerhub description of saio image\n\nChange-Id: Ifcb5013a5728d93cf5491fbff81b2677450698e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7cb34ad61feb5b830ccb79437b32965916220ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7cb34ad61feb5b830ccb79437b32965916220ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
