)]}'
{"id":"openstack%2Fswift~704924","triplet_id":"openstack%2Fswift~master~If6a25d6573502eebff3e28aa503721eda73af556","project":"openstack/swift","branch":"master","topic":"versioning_flag","hashtags":[],"change_id":"If6a25d6573502eebff3e28aa503721eda73af556","subject":"Add new versioning flag to docker image","status":"MERGED","created":"2020-01-30 04:57:10.000000000","updated":"2020-01-31 20:43:22.000000000","submitted":"2020-01-31 20:41:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704924-1580503285464-9115562a","meta_rev_id":"90401fc75bd2f4894ce72f0d5106330a986ff3d0","_number":704924,"virtual_id_number":704924,"owner":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-30 08:03:12.000000000","_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-31 20:41:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-30 18:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-30 18:00:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-30 18:00:28.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-01-31 20:41:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01855252e5311e9dc288b6ab1e4d948526eeb6bb","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-01-30 04:57:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb2cfc1ac327a65919c43cfda86f2d436e4a14ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 06:03:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b361da34edfe4f3f9a1eb358627f1ed0 : SUCCESS in 9m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/43a953cf7c7545b4a94c30e5beee325c : SUCCESS in 57m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f0e4d465e37b4a9c9482e6d43a881d62 : SUCCESS in 59m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fdb5712253bf4102bbd1a4398d82f598 : SUCCESS in 9m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e98748bcb36c40d29ba16beb51e76739 : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f70ff9553e704d2ba276ab78e7137c7e : SUCCESS in 12m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe333c6256334beb8db6504981740beb : SUCCESS in 17m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/453eee9160224e668efbec19dbf7a9b2 : SUCCESS in 15m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a7ab8db5ad41429098b85197180d3712 : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/595fff4fe3e94d6c964a46c475435739 : SUCCESS in 17m 16s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/574dd6b9d53946d29d50638b1798cc88 : SUCCESS in 13m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0d4bbcfcc7814017ade0f84f69653d06 : SUCCESS in 10m 43s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f24fd0630fc64f44aaf08e9730aab610 : SUCCESS in 24m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1dd018455a904f8d879ba906aea7e6a8 : SUCCESS in 9m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cfa8f7cb6f0a4393ad467d8741ca5b3a : SUCCESS in 17m 56s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b52cd560581e44feb8d0769377ec2b00 : SUCCESS in 10m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/24309366d5a84d11b709ada00dad9bd1 : SUCCESS in 14m 03s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0ad28698fc134b48803b272e41982c5b : SUCCESS in 14m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eae7817756ef4a748a1770d19a04cd3e : SUCCESS in 33m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c06e43b31a574e22ac335887c2628581 : TIMED_OUT in 1h 01m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c5394483a0c34052acfd14b6f4ae9124 : SUCCESS in 29m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7de5d38ba46b4340acf2cca6950d45fe : SUCCESS in 27m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bbeb11f435f243f6b7a23cc2948fa12e : SUCCESS in 28m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e40abd3202d46feaccf41cb92ae3a40 : SUCCESS in 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/649645f05c5643548b663542061e8073 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d57e435f8b44e529e5e6731a4338255 : SUCCESS in 21m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30667391df364967ab85bc2245a3f728 : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a9768f37c4cf21b9b2c879e5f9e5c6a0829389a4","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-01-30 08:03:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2aec0bf65bbee145c092d17c6cfb19b0fb988bf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 09:14:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec39489aeed44339f1689b1fc98a722 : SUCCESS in 12m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/db0ca89c17cc4a67a0ba790b3da1332d : SUCCESS in 1h 08m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/142285e6af1e4984a7a7497f326b4345 : SUCCESS in 1h 04m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/08ecd5b011ef46b4851a717fd77c705c : SUCCESS in 17m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/facd6d42d1c5419b94ade014196deda0 : SUCCESS in 18m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2a4d647b2e6f4ddfb69736a36c463f1d : SUCCESS in 17m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee4d8dd5aa5b4421bf47b14b03ef994f : SUCCESS in 17m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f3ce12e16f914f4e9c8dfa3ae6595f4b : SUCCESS in 13m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3e69a6bb173845c692757d68e4d49e66 : SUCCESS in 10m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb1ffbb221b34354b26b5eab55b08b2d : SUCCESS in 14m 39s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/dc84b03ed25544ccaeee5b2806aab81c : SUCCESS in 10m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aee244a1dbfd44bbb119cf52b7d1a8ec : SUCCESS in 11m 09s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e67370cec30641a3b4c132de68c4aa6c : SUCCESS in 16m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d662e89d07d44c96be2f573aecc236e1 : SUCCESS in 11m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ae12fb6d88c0407181ac1ad8aae185c7 : SUCCESS in 13m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c44bd4e235404dbab610eda708bbc043 : SUCCESS in 11m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b95cc07456214104b033acef02788926 : SUCCESS in 12m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/07b8347a9dba481db969459de966dd1c : SUCCESS in 18m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/72bfea678e024ecf884b909ea953c597 : SUCCESS in 20m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e1fae241e3c3433eb76205227a23f2de : SUCCESS in 45m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/33cb33439a9849b6a099916893fdcaad : SUCCESS in 30m 42s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a5db015e29d84cf3b01c8cafada0ff12 : SUCCESS in 28m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b590dd5bfe542c6a34bb18352bfafb9 : SUCCESS in 32m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1b8c1de0d8649f3bcae298923032be9 : SUCCESS in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7012a4528074a21907d81ce200b0e8a : SUCCESS in 7m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8032de4abcf24934af49adb3509c0bab : SUCCESS in 18m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69e49b79aa9447af8c71a97ea0610118 : SUCCESS in 1h 05m 21s","accounts_in_message":[],"_revision_number":1},{"id":"cb9441752e1f56ad22fcd63ff8f0c1f1c6db0c7b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-30 18:00:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7ad7916e77ece89be6c3417e5877cf332c2c381c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 18:00:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"323868475d1e8678bf3e8490768e26d319522e3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 20:00:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb0f0adc283435a962b7fe464c760e5 : SUCCESS in 8m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d8cad92b9cbf4fc6818b9d71af0f8c9b : SUCCESS in 1h 04m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/30eef5d1a52942f6bc4d30424985a595 : SUCCESS in 1h 27m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f6609af25e2746d0a92c91296bc7b1d6 : SUCCESS in 10m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/98be3e2e1c1e487a877b2665389a904e : SUCCESS in 11m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9685f10e1f664026ad33badaa72c0d1c : SUCCESS in 14m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff44689faf19401d822eb22d0aacf6a7 : SUCCESS in 16m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c2b4832902d6429ea955c6a053c89c45 : SUCCESS in 17m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/538c7544864f4ab4bb196751d31a51d4 : SUCCESS in 10m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6f26abf7177e40afb0f141f1b0468937 : SUCCESS in 17m 40s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/158cc249d294416ab6f1a9b74a9440ce : SUCCESS in 10m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5bead2c9c5e045eaabf5307ede8fefae : SUCCESS in 12m 03s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/dc49763d25f44965b6c2b42c5a0b049b : SUCCESS in 19m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/196ff44752b346b68b8ee14fe8618c01 : SUCCESS in 14m 57s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b130e28299bf4e99b8f0e610c6b3fc6d : SUCCESS in 12m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d21359a164af435b9fc831364011501a : SUCCESS in 11m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ee747724fc224925b957d54b2b40d2ab : FAILURE in 23m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e0942f6bfd014638bf7a789df0f0c635 : SUCCESS in 48m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/23451acb4d4546f28331da5400a20c39 : SUCCESS in 30m 52s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4a8cf947e68f495b88e14ed04151fd8a : SUCCESS in 36m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/55181edf4f094b4bacd32a7e5cb8bce2 : SUCCESS in 41m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfbf1f4d69334caa89e08572e4914498 : SUCCESS in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db21d49001994148ab909da2d8d1cea7 : SUCCESS in 6m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2260817b2cce4cc7927124ffcffedebc : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7f4963d04c8ee75c24a039239b64eaac3cadd8cc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-30 23:25:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"50c7cda5e63eaa4c72b0169d9cfeab3e20eea164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 00:54:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512e52019ecf41d4bcdb80ce1fa6f4aa : SUCCESS in 7m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/459c36e7393940558b83945a958f9b7d : SUCCESS in 1h 17m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bb04951d0ccc4239bf38db008e96f1b5 : SUCCESS in 1h 08m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/674bd1b57f6e439582baf2ad9042baf2 : SUCCESS in 10m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a9360b2f94ac4e228b6074f1e7431c83 : SUCCESS in 12m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca3c002157c84653a98ad2990914b748 : SUCCESS in 24m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/58487a20e3944b7093d701e44f604bbc : SUCCESS in 23m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ce48b711ac034096b1df8ac7a9d961c8 : SUCCESS in 24m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eaa362eb7ea24d8b9c9250b5a01a9458 : SUCCESS in 23m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/62266811159c4ff8b46d68a3c2d7fa75 : SUCCESS in 21m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c2d4b5e01b0740498bcfbdf4c8f9b6fc : SUCCESS in 11m 36s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b211b230755d4fcdaf0e6c7bc27ea099 : SUCCESS in 21m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d7c23ed21e004874848256fa0f919b93 : SUCCESS in 13m 36s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f713c8af2f0f42618b6ee89bab2d7081 : SUCCESS in 14m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a1d8456f91f74cb3aeabf1846ef7d126 : SUCCESS in 22m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/163f5267f95d409aa9ab76c896ae3bb4 : SUCCESS in 13m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/29cadbf7f6c44239986f89dc381fcb2e : SUCCESS in 9m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/62d6da3a44f74285b0975596ba98d798 : SUCCESS in 15m 30s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2b4345a9c4184c11b1076821093ebcba : SUCCESS in 19m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0fc28172925d4f8a82e917b69a3807dc : SUCCESS in 27m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8d298ec8334a4bd3ab8830f10a8daebc : SUCCESS in 57m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a61b435992b24450bbb303d8ed1b9213 : SUCCESS in 42m 12s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/632d783cb424479bbf8768a8cf6e55cc : SUCCESS in 33m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e4a30d13dff4192b2981e5d60be9291 : SUCCESS in 42m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a7218fa75414721a532a87f45af4215 : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/120b09cddb2a43bdb35b24303f89e3d5 : SUCCESS in 7m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e43a9ce30f1a446abec5f6d8bb86beeb : SUCCESS in 25m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd716db542af4d8cb3219144b1d4e89b : SUCCESS in 1h 19m 00s","accounts_in_message":[],"_revision_number":1},{"id":"608c4e164ea9811fddfe2e4925b260d29e063cd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 00:55:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7084502c5c062ffa29d90590a2069c4a73eb83bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 05:09:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6ff451805c24ed88a3b1b079673ead0 : SUCCESS in 8m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b9198ca81cc542d8b01a0ae22287f6ee : SUCCESS in 1h 07m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa2b0fa351c14c69a7c4b2283c61f181 : SUCCESS in 54m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5c923d03b5914c2d81b574c75dd0dee6 : SUCCESS in 10m 25s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/0a4680c3545a4537a79838ba5186ba83 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4840220e5500455caef8c50fbb9cb940 : SUCCESS in 12m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/43756b2f35e04eb78ac57321f714d543 : SUCCESS in 13m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/615542d69b484055a13a536165aabd10 : SUCCESS in 13m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7866714027c84f6790ba802a35a75972 : SUCCESS in 13m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f92a923cda6f4c44b71d26eb7d431f87 : SUCCESS in 9m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/36abb41b6dff420abafe0a4e0ac1b86d : SUCCESS in 12m 00s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4a620a9cf3b5494593b8922db7145237 : SUCCESS in 9m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3a726234d79b46c0b6e703f17d4cacc8 : SUCCESS in 9m 52s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/37862da5435142cc8f4466d5886b4fdf : SUCCESS in 22m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b5b31361288147c9a384b71cadb29351 : SUCCESS in 10m 31s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d060eef1917f47a4a695ab69ba34adea : SUCCESS in 13m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b081d1483baf46dda3f13ebcf6865bb2 : SUCCESS in 10m 56s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/47349119b1d34f6c8bd965ffcb6ba8d1 : FAILURE in 20m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9c2b571276284b13a4f28e3f02d1c13e : SUCCESS in 52m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4bcf5b0eb8234c03892d297bcce75898 : SUCCESS in 30m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0d05d529d34d4283bf7bef601ab69dbd : SUCCESS in 26m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/00a82ddce8394048be3e70d4462865f7 : SUCCESS in 34m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abdb665d6731418aa505cdab072f5034 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63955966d56342128d62d8a8068e31c1 : SUCCESS in 8m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f425ff4120834345a889f87f421d3d8e : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":1},{"id":"95a7e77b49d82e4f157cfe39571dbd47ccde1165","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 17:31:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc9804d1789a421f05e62e73b3c158225ba2a388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 18:48:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c3536d7186147bcb3936e5b302280a7 : SUCCESS in 7m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0d48d8cb1db64afb9e4581fc825baee4 : SUCCESS in 1h 02m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4bf6f4b17c044e83a5306d393a018b63 : SUCCESS in 55m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d87c76b5b1684e37b63e4dbe4522ca69 : SUCCESS in 9m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/06598a600be042f1bdea678c165f9b0a : SUCCESS in 11m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ffe0a11df0d54adbb33849b50cb5f034 : SUCCESS in 13m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8a038c60ef9f4fe59b043041177f69d8 : SUCCESS in 15m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/79ae5855f0af468fabab446a2a821979 : SUCCESS in 13m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2c4f50939adb4457b876f41a7075e5d2 : SUCCESS in 14m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e429fb333a4949999bd1dbb14727fea4 : SUCCESS in 12m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8fcde9212c88479b94be8d148926f1a0 : SUCCESS in 14m 59s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b360e899c6be464e800b1848feada681 : SUCCESS in 20m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab8c51e509cc4654937bf9eb7e15fdfb : SUCCESS in 20m 27s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a90382448f364a73b826818e81375aaf : SUCCESS in 18m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/68adc5aa92504aa7a3f23f0e9b59bfb2 : SUCCESS in 11m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1ebb2d08c74e443cb94950bbdd46ca25 : SUCCESS in 14m 23s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6edcc02265524a8bbe53accba3439bed : SUCCESS in 11m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/872ddfa5cfd24193b6a88103260ea016 : SUCCESS in 11m 57s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c31453c869214a09b46f90ef25fd573f : SUCCESS in 16m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6ffbda43f6840c294b1899b23279c9d : SUCCESS in 22m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/68d0d579b2f946da801397fd76d8b10e : SUCCESS in 49m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/925193438aae4850a63e1b5f43784792 : SUCCESS in 32m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/202bd3cb163a498f87a974b1c9740fa8 : SUCCESS in 40m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/819a045cf62c462eb8f0026cfccfc7b9 : SUCCESS in 32m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10cadec20f5d4215ab2826c8535a6762 : SUCCESS in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e91563dfd10545eb82530eed2e043b62 : SUCCESS in 7m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05c001ea0c1b42beaceb309577c3b1d4 : SUCCESS in 18m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b474528f4fbf4b23bce41017bf769a6f : SUCCESS in 1h 10m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0c9f0f1e9b748da28ac029fdab6e6b369bb616b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 18:48:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e759ac41d06c4275c00e9bbe580fae9a0e42ba88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 20:41:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08bce49db9148bab9df197e76edcef5 : SUCCESS in 12m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/82cefca0b742415592e6d5e610ff3030 : SUCCESS in 58m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75f1b475afe24ccbb471aef2ac5c3d0e : SUCCESS in 53m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3130cc91221847deaf6dcbaee94fcef5 : POST_FAILURE in 15m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c85adfc4d46f43e59dbecf7b43147e5f : POST_FAILURE in 16m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fd46a6b1b3fc4872bd9c4bec1203adf5 : SUCCESS in 11m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe554822f5b74cc2a3f103cf142f0b63 : SUCCESS in 15m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9f1a16bfaaf94283bf486a73a05b1371 : SUCCESS in 13m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d68e71d9d8c04ae2803a77904bcadad2 : SUCCESS in 13m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/74d35e81e3264dc8a845fd571bd9a890 : SUCCESS in 18m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aa7c3a9cdafa42c4a6b11cfc1e5f2f64 : SUCCESS in 20m 16s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/2e5c0285081c42a08cc9268be47a8881 : SUCCESS in 11m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f2f5dbbcc5724c97bb6b740ddf3910ff : SUCCESS in 14m 02s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d41e973fb1fb45108dc90eb7c5b217c6 : SUCCESS in 15m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dafb1557d03045faad046a49efbc7c67 : SUCCESS in 14m 18s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/53e9c143ea4a4870a05f91471e13cda1 : SUCCESS in 12m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7dd87e38bd714685850209e80c9b0a7f : SUCCESS in 11m 57s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b7ce7211ddf74efc98422ed89e13d3a1 : SUCCESS in 17m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4cd76811946b4d5bad4f3e59da555911 : SUCCESS in 47m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/40feed9edc35467eae097ec90b8353cd : SUCCESS in 36m 42s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/6eac70d1cba742d79f330fb3b4b00682 : SUCCESS in 36m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/839d338a2c944627afa129a6de3f30c3 : SUCCESS in 35m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a2f7b6046d44a6fb881919ee46b1633 : SUCCESS in 14m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30c896e09ec342cf8af2e76268a37119 : SUCCESS in 7m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b585afa72c642faa4303b77676b94fd : SUCCESS in 1h 06m 03s","accounts_in_message":[],"_revision_number":1},{"id":"bd94f1d264ee210dd0e3146dfaccee7b1201df88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 20:41:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7a498e7e2c29d8e5529569bbb29c3526db87b85f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 20:43:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc8d6d26b2b478eb9ec73a1376913c4 : SUCCESS in 1m 34s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/32a94ad63a294dfa9213cba969b0a48d : SUCCESS in 55s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ba1cd7add2fd4d15b3edbe10fb8d021d : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a776c4fbd69271b6197b2c20beaaae8a2b2e980","revisions":{"4a776c4fbd69271b6197b2c20beaaae8a2b2e980":{"kind":"REWORK","_number":1,"created":"2020-01-30 04:57:10.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/24/704924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/704924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/704924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/704924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/704924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/704924/1"}}},"commit":{"parents":[{"commit":"5e70039df5672150ec971504921f2e8255348c4d","subject":"Merge \"Bump up ceph tests timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e70039df5672150ec971504921f2e8255348c4d"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2020-01-30 04:56:36.000000000","tz":660},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2020-01-30 04:56:36.000000000","tz":660},"subject":"Add new versioning flag to docker image","message":"Add new versioning flag to docker image\n\nChange-Id: If6a25d6573502eebff3e28aa503721eda73af556\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a776c4fbd69271b6197b2c20beaaae8a2b2e980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a776c4fbd69271b6197b2c20beaaae8a2b2e980"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
