)]}'
{"id":"openstack%2Fswift~705578","triplet_id":"openstack%2Fswift~master~Ic0f54f393002e2170e7f1459625ee5a2b37df900","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ic0f54f393002e2170e7f1459625ee5a2b37df900","subject":"py3: Fix up probe tests","status":"MERGED","created":"2020-02-03 23:15:46.000000000","updated":"2020-02-14 21:40:44.000000000","submitted":"2020-02-14 21:39:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705578-1581716341115-b4850ae3","meta_rev_id":"6872b4a2e96c65ab13f4fd217080066605666a74","_number":705578,"virtual_id_number":705578,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-14 21:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2020-02-14 20:37:32.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-14 20:39:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-14 20:37:32.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-02-14 21:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d0ce7442212705ab2b486a5d2ad407ae74a2efa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-03 23:15:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ec32f72c035eeed5f600be921be286670947543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 00:17:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/021af7c9dfce441bbb45b576535fd8cc : SUCCESS in 7m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d686421d294c48a8b58866fef534451e : FAILURE in 58m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7827f2a0bb9451bb84bc03631cae869 : FAILURE in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5f136ed86ab8c9cb34ca51fdde070ca2c3cb7694","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-04 03:26:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"76e66b23c78242b8ae0e159f038e30640b761cf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 04:16:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/035165fb24dc4bdcb998318e897a063a : SUCCESS in 8m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6afd1b752329440189c51b51fa79014b : SUCCESS in 45m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d84f5e8446bd4d3fa3fde1f186152c1f : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"64d1e70028de45ff613d710ceaef4827563d71c3","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-02-14 20:37:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ba8e9c940b535391fbc6a00257925274b3e00205","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-14 20:39:08.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"567d48bbe2496c0d10dea1a1474c26706966c34d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 20:39:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9695a928b5213ebb034104850b093e0e13c00c9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 21:39:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011ac9347ad142679514d5b330ac61b3 : SUCCESS in 10m 41s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4bfa2db553924069b779ed6d3587d2e2 : SUCCESS in 10m 55s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/69bb9cd0a70a4fb596e71a10fb8184db : SUCCESS in 10m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f815edc256a5499ea39eb90c27bc6226 : SUCCESS in 20m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d648729a85ce4927b573f2065b5c4aeb : SUCCESS in 14m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3749ae621de643ad9eb8f747a4d0f4d9 : SUCCESS in 14m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fa934d2a053042a1935164e2f6ed5089 : SUCCESS in 16m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2591fb0c5f66423591d1e71e584d4a21 : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6ded1d26949741c1ac72a8be06f9d053 : SUCCESS in 16m 43s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8564876685ce49e29aecfcf6a6d74d5b : SUCCESS in 14m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a90ead73de6141bdb24b228e180a8e62 : SUCCESS in 15m 00s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ca7b710cbd2147ebb4a4a6e3008beba8 : SUCCESS in 19m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/568da8a99d9640809f9fdbd38b370e9c : SUCCESS in 13m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8c1adc4296c541929ba2800e8b07db73 : SUCCESS in 11m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/759eec5e8efa446ea40dc03e70decf61 : SUCCESS in 17m 33s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c6b712ac0c524a698b23ec74b77867f2 : SUCCESS in 18m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/debf72f4064348b0b344502c6d5a5645 : SUCCESS in 47m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fddeaf21ba01443cbe7206e16e4a3b4e : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":2},{"id":"dc1ba6ede6cadcfba4a87fcae2dbc76d2cc4d9f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 21:39:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6872b4a2e96c65ab13f4fd217080066605666a74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 21:40:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde3bcb956434c0480f392db7893e96a : SUCCESS in 1m 16s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6335b5b068e444dbb43c989da6de2cf2 : SUCCESS in 1m 02s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1835cd5c737d4692b3dacb4ede4b2b46 : SUCCESS in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff885d30e46f65042843480f58a6f07ce66211cc","revisions":{"382e9a7ab681bf6da91e6e207b70937e726f4387":{"kind":"REWORK","_number":1,"created":"2020-02-03 23:15:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/705578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/705578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/705578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/705578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/705578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/705578/1"}}},"commit":{"parents":[{"commit":"522a072e726e2b5f07a0f6a216ec1b4fb499ac63","subject":"Merge \"clarify s3api ACL test HEADs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/522a072e726e2b5f07a0f6a216ec1b4fb499ac63"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-02-03 21:01:51.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-02-03 21:01:51.000000000","tz":-480},"subject":"py3: Fix up probe tests","message":"py3: Fix up probe tests\n\nChange-Id: Ic0f54f393002e2170e7f1459625ee5a2b37df900\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/382e9a7ab681bf6da91e6e207b70937e726f4387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/382e9a7ab681bf6da91e6e207b70937e726f4387"}]},"branch":"refs/heads/master"},"ff885d30e46f65042843480f58a6f07ce66211cc":{"kind":"REWORK","_number":2,"created":"2020-02-04 03:26:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/705578/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/705578/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/705578/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/705578/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/705578/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/705578/2"}}},"commit":{"parents":[{"commit":"01b1d60cd9d55089d3d704d979bd1804f37bcaf2","subject":"Merge \"enforce account quotas on empty accounts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01b1d60cd9d55089d3d704d979bd1804f37bcaf2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-02-03 21:01:51.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-02-04 03:26:17.000000000","tz":-480},"subject":"py3: Fix up probe tests","message":"py3: Fix up probe tests\n\nChange-Id: Ic0f54f393002e2170e7f1459625ee5a2b37df900\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff885d30e46f65042843480f58a6f07ce66211cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff885d30e46f65042843480f58a6f07ce66211cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
