)]}'
{"id":"openstack%2Fswift~707762","triplet_id":"openstack%2Fswift~master~I4c8a3756afccb2b42acbd9aeac5908f3d4febcae","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I4c8a3756afccb2b42acbd9aeac5908f3d4febcae","subject":"Make rolling-upgrade job non-voting until virtualenv problem is solved","status":"MERGED","created":"2020-02-14 06:02:58.000000000","updated":"2020-02-14 22:55:10.000000000","submitted":"2020-02-14 22:52:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707762-1581720748167-798fc2db","meta_rev_id":"c193faaaf477daa24e9cd5d6e8996d0190cae3da","_number":707762,"virtual_id_number":707762,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-14 22:52:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2020-02-14 20:23:22.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-14 17:52:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-14 17:52:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-14 20:23:22.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-02-14 22:52:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bbbbf5d846aa65fbc827cf27fe24ac72fe313a0a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-14 06:02:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73d3082a1ea322960fb96bf100127078a7f0c621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 08:22:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aba3ec9353143a6ab993de2b8256b1f : SUCCESS in 9m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fff6c249a64b400a803b565a83baaa0c : SUCCESS in 1h 00m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c0681983c2874f0abe106bc65ed921ee : SUCCESS in 56m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3332a869ac7d4b7c8382d43d280e26a7 : SUCCESS in 9m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/106032e981a34b26b090b7666d166da3 : SUCCESS in 10m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/365a636ef31f489cb8d39dd1c875d3e9 : SUCCESS in 12m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a3a8ae6001204f11a0aa98ab7ccd4094 : SUCCESS in 12m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d4a4eb6ac96b49738f954e2bfa1ce778 : SUCCESS in 28m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d7a02974e1a84d769679f6bc718a345b : SUCCESS in 28m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/712a521bff37489bbcbef588ff38f4b1 : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1b0f7877ee79444fb1a059ea33dde443 : SUCCESS in 20m 11s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a4a0a849b23745a68488feb3505c3043 : SUCCESS in 11m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef9baf13ddea4552b91f3e88d8736c0e : SUCCESS in 11m 41s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6e47fc5dd0b5450c946b0eb3035d0551 : SUCCESS in 21m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4d4a2f140247462c849a3c564465dac4 : FAILURE in 18m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8711544a03e44da191a6f3e583ee345d : FAILURE in 20m 12s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c74fc84c6b6b4452bdeb13ea770569f8 : FAILURE in 14m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3c07fb6e6f704383b20b466b0089c7ca : SUCCESS in 15m 54s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/56582901a21341328de88f882a7f0d46 : SUCCESS in 25m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6e42772c776f4db78cbddf8df7f452f9 : FAILURE in 7m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6bbd2fa39ad5457787164d133b4af5a1 : FAILURE in 58m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e8954c0f319a48cc9c0c608fbf200bf7 : SUCCESS in 39m 50s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e3f36eb2201948d0a1c8efa53dcd430d : SUCCESS in 39m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ca3b2934d7d74f5bae56f537b0b21f36 : SUCCESS in 31m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97dc41447b7e4f49bee376131ef67a31 : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4e8fa00f2ae48149e3e5942c583a5c2 : SUCCESS in 9m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/834337b028b243e8b17ed69a3955b226 : FAILURE in 17m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c271e272e1f4af0b8283135b23c44e1 : SUCCESS in 1h 13m 42s","accounts_in_message":[],"_revision_number":1},{"id":"52584b5034242b348542480b745bdee774a262e5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-14 17:52:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck\n\nrolling upgrade job continues to say\n\n\u003e /usr/bin/python2: cannot import name ensure_text; \u0027virtualenv\u0027 is a package and cannot be directly executed","accounts_in_message":[],"_revision_number":1},{"id":"698d556753a6e01223a334689202031a5db01e23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 19:06:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d7b8c8637c04e5fb5a36da9d71636fc : SUCCESS in 9m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/81633665bcb14f889cfdc65b28153133 : SUCCESS in 1h 07m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/433daede4cbb4e8fa48ea0a06bc6e2a9 : SUCCESS in 56m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8df8c1f4fe53495f80925b9973955cd1 : SUCCESS in 9m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ac524cf1f7cd4d71a6d9eeaab63b27cb : SUCCESS in 10m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0fe3091e286e44e0bc1a981bf7ceb720 : SUCCESS in 12m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2c08f0f0e2754bd9865a7fa0e592ea70 : SUCCESS in 14m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/824d40e155e64b949575c3f0f9147de0 : SUCCESS in 14m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/02bb0b4191114e90bbe8df2952fda40b : SUCCESS in 13m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cb1424dc68ab46c99aedd89ce0df18e1 : SUCCESS in 12m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4ed5d8f70abb456da50b6c5832acbe78 : SUCCESS in 13m 09s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/69ca1c40c53242a5be76ee5ee48af3fc : SUCCESS in 10m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/575321eece274cec87dbe51089b3828b : SUCCESS in 12m 08s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/149f3bfda08e45ab80dd9a0c24b2de93 : SUCCESS in 16m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7ea5a94416eb4a8aa9e52edee897fec5 : SUCCESS in 12m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9ebd6a13492c49b791f4a6cc27ce94f2 : SUCCESS in 13m 31s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/cb375a29ee8b452b9590904dd820e0ec : SUCCESS in 10m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/70ac436aa239405481f4392e8978f095 : SUCCESS in 14m 56s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a21b0f7a29354291bdce332fff2edf0a : SUCCESS in 21m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/52e62ae4098c4f8a9fd060d43b33b9a3 : FAILURE in 6m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cabd738f4dfb4108a20a88518de140e2 : TIMED_OUT in 1h 02m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e26abcc882a43b383cd3486c1c28514 : SUCCESS in 33m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1447d87be7164065a10307acfd5ec265 : SUCCESS in 32m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d3e223bcf4e94da383d84f6f664e5354 : SUCCESS in 37m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da144b48e0c64e5499e0219a7ba36af9 : SUCCESS in 13m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fad718c61754b6798b90d7644f6a6de : SUCCESS in 7m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cfe3bfe3171649f6987b9c4ef8955ef0 : FAILURE in 12m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61173dd287ff426dbbc11e76b715f87d : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a72a4cc0d9eb9b5805813000fce2fc575bbd009e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-14 20:12:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d35b86fa61731cc5d0b90417d4a2238ae47aa3fb","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-02-14 20:23:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"90601e7107db5219126328a5be4cb11a8eb3fcb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 21:30:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf07e18e5c1e418d941fa7ef921a44e3 : SUCCESS in 11m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bfc39f2c161b47419f02019e38619d5d : SUCCESS in 1h 02m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c0e632fcae9549b391e85a8b98f35250 : SUCCESS in 57m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b82f0325b5a94dbfa644070063f6e833 : SUCCESS in 10m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fd6bea79fa684787a5635eccd42a797f : SUCCESS in 11m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ef7448f2351e45c58a47ba5b48c79160 : SUCCESS in 16m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d9b3d8fa23844709860a43592a7c2133 : SUCCESS in 13m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b37656c19c714cd5bcbc429ff03239ab : SUCCESS in 16m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cab6b25ed92c405f9b8070b33ca5253d : SUCCESS in 18m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0b7d22c0a93f4ceaa59232e6a2c86aa0 : SUCCESS in 11m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a9ad5a1290384b578a65b1aec4a6413c : SUCCESS in 21m 19s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/222d7aa6bfb04634b401fd8727e1facd : SUCCESS in 10m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5da182b7874f460494f2bca208a50ed3 : SUCCESS in 11m 53s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/43a44403f2564fc6bac81bbc870cbc48 : SUCCESS in 16m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d1c05ae27c854671a762e81d9d5f4d58 : SUCCESS in 10m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1c0e090b7e034e3d9fe2c312cfadf253 : SUCCESS in 12m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/387a3bc459354fb6bfdc13407f006ecd : SUCCESS in 12m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5b24c808bec9457ba3bfb1e1f5cfc152 : SUCCESS in 11m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3a41749133664164a98199fc5c6b05c0 : SUCCESS in 16m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/317f2e869bd84df380ecbc3a9f57b5b0 : FAILURE in 5m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/af8c27de2ca04a279b50de8611bf210e : SUCCESS in 47m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0da2a5065233495287b140ed50ab6a98 : SUCCESS in 41m 37s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/3d58d66a310b40279f98712f80510f96 : SUCCESS in 31m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/abf90db0a827421f8fa1bd1eb4ee24cd : SUCCESS in 29m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b659ff8a1114558bb47d8334ff44424 : SUCCESS in 16m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfb14056beba4c658674cafc8979068f : SUCCESS in 6m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c6c5a4c266624bdc9f353df30547edf2 : FAILURE in 12m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed54791adc5440679fe3d26eb4769370 : SUCCESS in 1h 09m 05s","accounts_in_message":[],"_revision_number":1},{"id":"55afd41c9ad68d30b6252c9f3a77b9a271fa02d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 21:30:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57f2e055fce154dab15a128c47b17a07925d603f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 22:52:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbc55a22612840f3836e4fd4b9e514ce : SUCCESS in 10m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a2b4c0179db84fc8b5e55884dd8c7a18 : SUCCESS in 1h 00m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/413b8388bb564104b54182a0a07cc4a9 : SUCCESS in 53m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/651a1ba61d914c9082ffa031abe1ac98 : SUCCESS in 11m 27s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1965a6ed72b649deb41751420a53c477 : SUCCESS in 11m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb4f440c96d542c8a2e6baf1a902b1ef : SUCCESS in 16m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e33bdba8f09140b1a2f6bda330cc4a7d : SUCCESS in 16m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/533c2087b4734e40801486cf9af00a78 : SUCCESS in 16m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/53a0e442476c4d0286d2e56f85ab0562 : SUCCESS in 15m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9d2ba8aba52840009a327f70c06673ef : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9e718fa5dc004c7e8594d337eb4a6e83 : SUCCESS in 16m 52s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/220c9ddca0104fc481db7afba1270f6f : SUCCESS in 13m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e7524a22b0614f5aa6ddef902c774c6f : SUCCESS in 15m 11s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/94b8d35589f24ee0bd267f9f623eab83 : SUCCESS in 19m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5e8d5a70d807464f80c01e4bd993f8af : SUCCESS in 14m 41s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a9e8995a7800400d90e380e6f24bb09f : SUCCESS in 10m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0c3d71c19c3148d8ad55e20e38fcaaec : SUCCESS in 12m 05s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/56a26461afde40b0b1a6256ba29e922f : SUCCESS in 20m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ac1c65b9480c44ffa4a2b3619daa84cc : SUCCESS in 41m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d0cb070d8a5479b8e37402247414624 : SUCCESS in 30m 58s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0efda629b71b42f2b8050a6a744677cf : SUCCESS in 30m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cf21d0cbc61f4b3e934d490d85a6e0e7 : SUCCESS in 33m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3b21f0b06254b5f9fd068ff807012da : SUCCESS in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97560e0ab93c49a9be5ce1472d9a9244 : SUCCESS in 7m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b26671d3e234d5e849bd9b5a7b9d3f3 : SUCCESS in 1h 15m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b0d9c7ed51f6940e19ad9a985aa5758821ef686e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 22:52:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c193faaaf477daa24e9cd5d6e8996d0190cae3da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 22:55:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb46f651b79140f9ae07128a0412fa7b : SUCCESS in 1m 18s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/df52bd7cd7d84f2f8fd81c9734ef1dd2 : SUCCESS in 55s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ee1dcfb604144c8c9e35eaf9c6831398 : SUCCESS in 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd4ba90cc1a13c94e2a02f1af015e20d6882e05f","revisions":{"cd4ba90cc1a13c94e2a02f1af015e20d6882e05f":{"kind":"REWORK","_number":1,"created":"2020-02-14 06:02:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/707762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/707762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/707762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/707762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/707762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/707762/1"}}},"commit":{"parents":[{"commit":"baaa5c57e18da5e7d4baa79e1b5420621c7adf0b","subject":"Remove reno from test-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/baaa5c57e18da5e7d4baa79e1b5420621c7adf0b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-02-14 06:02:26.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-02-14 06:02:26.000000000","tz":-480},"subject":"Make rolling-upgrade job non-voting until virtualenv problem is solved","message":"Make rolling-upgrade job non-voting until virtualenv problem is solved\n\nChange-Id: I4c8a3756afccb2b42acbd9aeac5908f3d4febcae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd4ba90cc1a13c94e2a02f1af015e20d6882e05f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd4ba90cc1a13c94e2a02f1af015e20d6882e05f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
