)]}'
{"id":"openstack%2Fswift~708719","triplet_id":"openstack%2Fswift~master~Icae429b4eba796328e767317e204ca30cc105da8","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Icae429b4eba796328e767317e204ca30cc105da8","subject":"Revert \"Make rolling-upgrade job non-voting until virtualenv problem is solved\"","status":"MERGED","created":"2020-02-19 18:10:35.000000000","updated":"2020-02-20 01:04:50.000000000","submitted":"2020-02-20 01:02:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708719-1582160568028-4edda447","meta_rev_id":"ff9f9294db71185db8699320fc552248af92004b","_number":708719,"virtual_id_number":708719,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-20 01:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-19 22:03:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-19 22:03:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 01:02:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57db7a61a8203381599adb32c6b3f54775f90646","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-19 18:10:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d33e85ba824b6d904a0999bb02134ec879ddd040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 21:48:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a74d5ceafa3d481ab0279fec35b174c6 : SUCCESS in 13m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f59ad3158e794ebf93dbed3d82b2c411 : SUCCESS in 1h 04m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e066c230cddb4f0eb88b398df28b4069 : SUCCESS in 55m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5df6d4f9a13a4489a11c5860670f1730 : SUCCESS in 12m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4024e40aad6b43b3901bb24d14226e36 : SUCCESS in 12m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/71388f0d7f2246159c566ea54a89a98c : SUCCESS in 19m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ddbd6f86eaa14ea68bc411bec1704276 : SUCCESS in 14m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d016154eeb894e13b22836e28cc78d10 : SUCCESS in 15m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8ff93a5e3cc14d44bc0012bc52fa8516 : SUCCESS in 14m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/993b0b025f5341ad970ba82008712a35 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a8aa2130c1b94f3aa1b9a7c79def6489 : SUCCESS in 14m 50s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/486956ce821444399a0ce72c7fa55689 : SUCCESS in 14m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee7495b13eb44a7fad2bb8385b123214 : SUCCESS in 13m 12s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ff6fa91c7c9443d98ce5e2f948de6ac9 : SUCCESS in 16m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/438e042e1d9345fa89ba7011e57757bf : SUCCESS in 12m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/92adb86c91264014b0de83be8ca05fa2 : SUCCESS in 15m 32s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/841fafdc0d0a4d7996cd21ef3854c5e1 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4d1d3f8c219e4b1e8fa2b3c22d10fbb9 : SUCCESS in 13m 10s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/72e0d0cd073548489f74615fefad3198 : SUCCESS in 22m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/98541dc6db18498cb471256f3b8f2c9e : SUCCESS in 24m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/392614fb4d384a278e7aa4c9b04e4fe0 : FAILURE in 46m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f205f597b8984ec995b18bc7c0b41cb3 : SUCCESS in 38m 55s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e13ede6fc988404795ec4974f3aee833 : SUCCESS in 35m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d4b5e7ba2d0b4a2ca60977fc12e631f7 : SUCCESS in 37m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc06821a5fc84942b1e6c908dca8d1d9 : SUCCESS in 16m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6275ecf6a26140d39a0edc0fb8514d61 : SUCCESS in 7m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/810376a6a7df4157932581ba9a9be635 : SUCCESS in 33m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c46d27b6b90e4a4baf8e71b16f9ae1fc : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7ee436daaa2d52ef5b5710aa8fc8bfeb254172f6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-19 22:03:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a0cce890ede7e80d57447460326dc4e8a5fdb912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 23:46:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c01767a56959486d96551095fdb75645 : SUCCESS in 10m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f3d35df4322644ecb7323f02710a4cff : SUCCESS in 1h 02m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e2d1d3fc4b41484ca6086d01ab96b4bd : SUCCESS in 1h 00m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/357bc4d8ef9d43c4923c0ed4f638dbf1 : SUCCESS in 12m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ac8fcd6090f84ab5930464ac5735a62e : SUCCESS in 11m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/763d4b20aa4f48b49c3993e4cebea71c : SUCCESS in 16m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9ad7fee5346f4b959bb3a111bf550edd : SUCCESS in 16m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/146888ca07bd40249e9932369ad9494b : SUCCESS in 13m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9f25c6e48c804e1d83886eef8fce858a : SUCCESS in 15m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/44a25e75479b4faabadaad00a735e67d : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/051beef88b7b4213bcc7aebe231831b5 : SUCCESS in 19m 43s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/0ec3982e9f2c488c97c4071bd7a43df7 : SUCCESS in 15m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/12e1275552b548b185701086ffbc83aa : SUCCESS in 12m 35s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/b1a9ab73d1f24d9abab0b3f623c5e003 : SUCCESS in 19m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/414cc78ffc2d4bd099d34ba801d82ca5 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d9792ae02c9e4ca69f94b2fb9318988d : SUCCESS in 13m 59s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/08a3fa6fb49640bd8676f73358080ac2 : SUCCESS in 11m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/970fbaa403c64a808a9f4e815f68877d : SUCCESS in 14m 47s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6380806159bd4005b7cb1060f41cb62b : SUCCESS in 20m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cefbf14c4d1741198bbb2943464394be : SUCCESS in 25m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ebe87076e3934b249c6f06a920288462 : SUCCESS in 40m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/111d2497bac64439b1c5f5dffa17c786 : SUCCESS in 32m 17s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c89f503b69c144b787db6bdcc56a1261 : SUCCESS in 30m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c236a0f1c088433b8d86f89ddfb1cc45 : SUCCESS in 33m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87ef0f9e635148a89f32a7fa3af32797 : SUCCESS in 13m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/467fdfdfdfdc4c05b96622b72571403c : SUCCESS in 8m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24cf289714634f17a7c305ae34d6d16e : SUCCESS in 30m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c17b2476d13a437397e947c854bd4e96 : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c655b4f17511534973c682fcfd63de0d1b536acb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 23:46:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e539a195a45e8302a308f5fa2de68596637dd179","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 01:02:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36e3b82eff0f4cabb4c97d1b9b2579f5 : SUCCESS in 11m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/60a5205c4f65420eb25f961ebc89f7b8 : SUCCESS in 1h 02m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e47dcaa841924ce9a7ab46f6cb5208f3 : SUCCESS in 53m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a03ef19a1ff24d8b86110f0c9af672bb : SUCCESS in 10m 34s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/da1f9d65388a4479bf4f28be8317aff4 : SUCCESS in 11m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2190fd35c45f4bda97314e7f32e9ade7 : SUCCESS in 16m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a3f878b2345471bbde3bb78c7948192 : SUCCESS in 15m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/414b0ec0d46648d3a26306ac7279a365 : SUCCESS in 16m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/731d3fdf9a164466bb444d4151772c53 : SUCCESS in 16m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a487f5c92b9a4ad09a74477f3024eb26 : SUCCESS in 13m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bedef33f8d03474eafb48069555e4332 : SUCCESS in 16m 15s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/60c87f25d7e44fc98629f3038e2ab313 : SUCCESS in 13m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a4c8852610414f7bbeff20163d00e406 : SUCCESS in 14m 20s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/fa5a7dec9a1f47b0a6fc3b19d63e6229 : SUCCESS in 19m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ebbed1f626ba4001a26dc13e6bba4468 : SUCCESS in 15m 18s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/50d16955146a4f55a038d97a9c89d1cf : SUCCESS in 14m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0f1d87e1e5ac40b1ac6d9d5d27bb4e23 : SUCCESS in 15m 35s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5ebaca3470a64ce68ab13d719854390e : SUCCESS in 16m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/834ed7d3c9174a21ba221380a05e1e48 : SUCCESS in 42m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b9948eda440445f39505f242e6ee4850 : SUCCESS in 30m 03s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5a8d12e682eb44d29c6ad0a607b104d1 : SUCCESS in 29m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d8da14746d0e4db5acb5023882b264a9 : SUCCESS in 30m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7410247605c4e508a3f7b9e6fafbb15 : SUCCESS in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a32461a8a771424eadf1f97b3f133ac4 : SUCCESS in 8m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff1e19660b4f44cdaa2b57ed0ce9c052 : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f9970f088e79e67a7823f86c493ea320be18beaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 01:02:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ff9f9294db71185db8699320fc552248af92004b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 01:04:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3aa845c061248d09dc4e00277669b17 : SUCCESS in 1m 08s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/dd03c8fbe81b42ea8db936713a38be57 : SUCCESS in 1m 02s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f3a8984d71694746a7780231e9ce8339 : SUCCESS in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5afe767581d2cb97cf3690067e6d626c7682c2c","revisions":{"a5afe767581d2cb97cf3690067e6d626c7682c2c":{"kind":"REWORK","_number":1,"created":"2020-02-19 18:10:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/708719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/708719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/708719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/708719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/708719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/708719/1"}}},"commit":{"parents":[{"commit":"6a47d9e4f95fb8554228b890346058d93ed1d6e9","subject":"Merge \"Ensure domain stored in memcached gets utf8 decoded on py2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a47d9e4f95fb8554228b890346058d93ed1d6e9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-02-19 18:09:49.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-02-19 18:10:25.000000000","tz":-480},"subject":"Revert \"Make rolling-upgrade job non-voting until virtualenv problem is solved\"","message":"Revert \"Make rolling-upgrade job non-voting until virtualenv problem is solved\"\n\nThis reverts commit cd4ba90cc1a13c94e2a02f1af015e20d6882e05f.\n\nThe virtualenv problem seems to have been resolved.\n\nChange-Id: Icae429b4eba796328e767317e204ca30cc105da8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5afe767581d2cb97cf3690067e6d626c7682c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5afe767581d2cb97cf3690067e6d626c7682c2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
