)]}'
{"id":"openstack%2Fswift~709323","triplet_id":"openstack%2Fswift~master~Ibd9fe5c9a1e75b86eb7d540594d5cf516758e17e","project":"openstack/swift","branch":"master","topic":"waterfall-ec","hashtags":[],"change_id":"Ibd9fe5c9a1e75b86eb7d540594d5cf516758e17e","subject":"Extend eventlet_debug logging to GreenAsyncPile","status":"MERGED","created":"2020-02-22 20:48:47.000000000","updated":"2020-02-28 19:10:22.000000000","submitted":"2020-02-28 19:07:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709323-1582916869036-b16e9fe7","meta_rev_id":"edf03ed1f5e6fad3557e199c1f4a2feaddf790e9","_number":709323,"virtual_id_number":709323,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 19:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-02-22 20:51:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-28 16:31:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-28 16:31:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-28 16:31:13.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-02-28 19:07:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12adfec5ea81455be45051ef0f9dcbf5af7345b9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-02-22 20:48:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4150ba32b926ddda6b2e4eb5479d764b5c4cd5d6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-02-22 20:51:07.000000000","message":"Patch Set 1: Code-Review+1\n\nthis is a huge win for debugging tests - i\u0027d be happy to make any changes needed to get this merged","accounts_in_message":[],"_revision_number":1},{"id":"dd403a85543ab000f92aa5286945410d8ec5bf0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 22:19:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4c9daee2e44ebdbef4235da15228d6 : SUCCESS in 9m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bebdc2dacc154036963604e64dd1dfa3 : SUCCESS in 1h 02m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/06ee0119069b447ebefad0659acce0fd : SUCCESS in 52m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/75030c7ad82a445aaa4ad3d8a9a020de : SUCCESS in 10m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/246f6ffd375e4fb3ae6df8dd48cd0fe1 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d642d48ddaec4bbdb5fa30556e81ad50 : SUCCESS in 14m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/77a1b748eddb41d3b2d441d22f57e4ca : SUCCESS in 13m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8209beb57b174f8cb12265245c8d4ab6 : SUCCESS in 13m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0883351654314ce5b7686811bd8544f9 : SUCCESS in 12m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ab4844dac93f4733978c2537b73840ea : SUCCESS in 10m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4180a4f6c09b401cb4d603f62b5f614e : SUCCESS in 16m 45s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a1d636b4ee4f456bb6872e5efa6a24a5 : SUCCESS in 13m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/94982b3c3db042f68f41a2bfc51f4efd : SUCCESS in 14m 20s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/4365fb433d9a4ede979445ed025b7be4 : SUCCESS in 16m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/03284e210b0d448d9841063404d0b146 : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/81ff8cdcd342479e97886c631b327d34 : SUCCESS in 13m 06s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ce67738d42604391bc20d4da98ebafce : SUCCESS in 14m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cfe55be47f41430faaffac9a3bb3927d : SUCCESS in 15m 50s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d3e7731b32c243a789aab9055d1af91f : SUCCESS in 16m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/caf950502e6e4b7a880df7c16f0ef678 : SUCCESS in 24m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/857c846352a0418ea3ce3baf248c1a29 : SUCCESS in 52m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c5c7a79d58744e38037825f46222788 : SUCCESS in 34m 04s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7b8cd562724c46e09f5363bc5d94a6b1 : SUCCESS in 30m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c65f1ef561c4c1992c9a90b70a564b6 : SUCCESS in 30m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44999b99f4f4492d800c9ad084ba9a20 : SUCCESS in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1e96b7e9674465e98d801722bbbece7 : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/909fa85de6e247068c080944ef6935ae : SUCCESS in 31m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abf9052bdbdd497abbad184476c9c69f : SUCCESS in 1h 13m 42s","accounts_in_message":[],"_revision_number":1},{"id":"51cd5c6e3867980ab5ce5e0245812a7df13ee8fc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-28 16:31:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95ee63cd649c8cc1ff86a3bb201ba153757a650f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 16:31:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"afbf9ff8821374039ac268b551307e7d03042742","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:07:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/511f0d86979648f9b9e86933e0a61b9c : SUCCESS in 10m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5f6cac954fe34e7b904215583651380f : SUCCESS in 1h 13m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e5b889f2ee92404badf6bcb038d989da : SUCCESS in 53m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/013653b78e23439c83a9ee6981fcc2cc : SUCCESS in 12m 59s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f4759164380b4ec686b8cc494369440c : SUCCESS in 16m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5c427efcfd7d40519a35d0221e3a5697 : SUCCESS in 20m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2d6a86882ecc4f8ea7be31f178cd1b43 : SUCCESS in 21m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8ed3489c1f344b6b8c590b3009d1cb59 : SUCCESS in 15m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a25645c9cdae45c886431a051353b316 : SUCCESS in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/38f9464fb35a4256abed8d7adb1576ae : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/95208c3c59f94afe8239d5d2fc78dce2 : SUCCESS in 22m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/09d669e6cd1e4e3b88f873e87a24aeea : SUCCESS in 12m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/15b338761e2c4931814c1e33835bfa8b : SUCCESS in 14m 17s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7afec7a53c5f4a8a993be7166624835a : SUCCESS in 24m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/77776e08d14e4d3597b711e0dddf9b28 : SUCCESS in 14m 14s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fcfb5f7d6a6049e188667808b7264377 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a591072b71fb4d41804614453ee374b3 : SUCCESS in 14m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/612ff2449ce249be9664a49baee29f06 : SUCCESS in 23m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1d7487059324406787f69a622f564ab1 : SUCCESS in 46m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/324306461ace4fa3af6243ce830bf388 : SUCCESS in 40m 05s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a8727695107e434daa7eac886500a005 : SUCCESS in 35m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/71d136765b4e43ef8296d6e419e9d698 : SUCCESS in 35m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a810c47da09e4d05b4f55ea8d89c092e : SUCCESS in 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/824927add4b041fbb6d96689b5a952bf : SUCCESS in 8m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1c04865d41f403eb764df914c6cc19d : SUCCESS in 1h 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"62fc38304ae2b346aec51dd7ea2da6ca30f8a342","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:07:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"16a44666296bb70a96448bd5bc88c8f257ff5c96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:10:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f87f149dbab64cae8f8b0ee6eccd5592 : SUCCESS in 1m 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4c63bf709cd74886b4d3d8cf0b75e801 : SUCCESS in 1m 01s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1032af7c02e841b7ac4457602580c42a : SUCCESS in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ecca23eb806e11cf6517f0456483da7a065350a8","revisions":{"ecca23eb806e11cf6517f0456483da7a065350a8":{"kind":"REWORK","_number":1,"created":"2020-02-22 20:48:47.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/23/709323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/709323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/709323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/709323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/709323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/709323/1"}}},"commit":{"parents":[{"commit":"a5afe767581d2cb97cf3690067e6d626c7682c2c","subject":"Revert \"Make rolling-upgrade job non-voting until virtualenv problem is solved\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5afe767581d2cb97cf3690067e6d626c7682c2c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-02-21 21:33:21.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-02-22 20:48:18.000000000","tz":-360},"subject":"Extend eventlet_debug logging to GreenAsyncPile","message":"Extend eventlet_debug logging to GreenAsyncPile\n\nChange-Id: Ibd9fe5c9a1e75b86eb7d540594d5cf516758e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecca23eb806e11cf6517f0456483da7a065350a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecca23eb806e11cf6517f0456483da7a065350a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
