)]}'
{"id":"openstack%2Fswift~709324","triplet_id":"openstack%2Fswift~master~Ib45b69cdbf29dac701388f1df18e0d5ff665a973","project":"openstack/swift","branch":"master","topic":"waterfall-ec","hashtags":[],"change_id":"Ib45b69cdbf29dac701388f1df18e0d5ff665a973","subject":"Make NodeIter greenthread safe","status":"ABANDONED","created":"2020-02-22 20:48:47.000000000","updated":"2020-02-26 16:26:14.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fc1ea4b6e166881ce1f23567114d8249a6157b53","_number":709324,"virtual_id_number":709324,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-02-22 20:58:05.000000000","_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-22 22:13:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-22 22:13:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65f558596abf4873eab0bcd7823d6637a843837b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-02-22 20:48:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c438125ae94b06abe3742502989a42005e32699","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-02-22 20:58:05.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nIt might make more sense at some point to squash this in with the next patch \"Lift concurrent_gets concurrency up one layer\"","accounts_in_message":[],"_revision_number":1},{"id":"1a653818ab0e65133e369c159a323a3abc6bb969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 22:13:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85fcab6952ec4c779c186a6afebc886e : SUCCESS in 10m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6d3d69eff7af44a7b697cc9432967dc0 : SUCCESS in 59m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/604c1e6d82804fe59204557652fbb3ba : SUCCESS in 53m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0520134e0664b5f887c0cb72374ee81 : SUCCESS in 9m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/482e751d24e34b6c8017ddcb371fa26a : SUCCESS in 9m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/af993702ec2b453abbdf3521ff0b1d93 : SUCCESS in 12m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0346028e7a2d4554a69e1913dad1c95a : SUCCESS in 16m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6c4f304e403e45bb819a89e1613e30ee : SUCCESS in 16m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/efc658adb06a4e30a08b3365ac78217c : SUCCESS in 12m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0ff261376fbf4f55a307a6d39f7442d9 : SUCCESS in 13m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3ebe064caa5e4d288eb647f6ebbeccc7 : SUCCESS in 14m 12s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5f291bb512b74723a502ff393f1867f1 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0bab31cfd00f4bc6b8008bf1bcdd785f : SUCCESS in 15m 14s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/82e41b0ba56e495ebf0531704d69e6db : SUCCESS in 16m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/657898f99e3542518ef05b3d18ca1fa6 : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/af22daa2d022434aaf30daca5ae5d880 : SUCCESS in 14m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a2b0c6ab1bd44e6ea3e8ee80c678a4bf : SUCCESS in 14m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/377a75c5d1fa4afe802a73e773b902b5 : SUCCESS in 15m 31s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2f939e75a11e46ac88d447440788812a : SUCCESS in 16m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/db4bb9ef17ad4e828a4e3feb030518f3 : SUCCESS in 23m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/caa9b59d291b40159b1c482d5824040b : SUCCESS in 45m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0c5f7300e5942408a6aeb633b4e1596 : SUCCESS in 29m 52s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/08f7e56aa67240559291acef3ec1396c : SUCCESS in 32m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c378e874c80f47e5a95c7e9f6fc7a937 : SUCCESS in 29m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cdd7973b9b64854b20946d34b6d162b : SUCCESS in 14m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63b65d3d160442d8abb357e8937be93d : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/465048044092423badc61b5c05bea172 : SUCCESS in 30m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e44b50199d5c4c65b8b86db33e75b7bd : SUCCESS in 1h 02m 25s","accounts_in_message":[],"_revision_number":1},{"id":"97cac9368def97cfabc28bb9163114a6a8ad5f76","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-02-26 16:26:14.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/#/c/709326/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f75f5341dd436d71448d7b70a3deeed30eac475a","revisions":{"f75f5341dd436d71448d7b70a3deeed30eac475a":{"kind":"REWORK","_number":1,"created":"2020-02-22 20:48:47.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/24/709324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/709324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/709324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/709324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/709324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/709324/1"}}},"commit":{"parents":[{"commit":"ecca23eb806e11cf6517f0456483da7a065350a8","subject":"Extend eventlet_debug logging to GreenAsyncPile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecca23eb806e11cf6517f0456483da7a065350a8"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-02-21 21:36:19.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-02-22 20:48:18.000000000","tz":-360},"subject":"Make NodeIter greenthread safe","message":"Make NodeIter greenthread safe\n\nChange-Id: Ib45b69cdbf29dac701388f1df18e0d5ff665a973\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f75f5341dd436d71448d7b70a3deeed30eac475a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f75f5341dd436d71448d7b70a3deeed30eac475a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
