)]}'
{"id":"openstack%2Fswift~710475","triplet_id":"openstack%2Fswift~master~I433c97df99193ec31c863038b9b6fd20bb3705b8","project":"openstack/swift","branch":"master","topic":"float-client-timeout","hashtags":[],"change_id":"I433c97df99193ec31c863038b9b6fd20bb3705b8","subject":"Use float consistently for proxy timeout settings","status":"MERGED","created":"2020-02-28 16:00:43.000000000","updated":"2020-03-04 07:31:46.000000000","submitted":"2020-03-04 07:29:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710475-1583306982431-b70934f3","meta_rev_id":"b36df0002d338bfc88c5861e3ca020a990e7accb","_number":710475,"virtual_id_number":710475,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 07:29:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-02 17:57:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-02 17:57:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-02 17:57:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-03-04 07:29:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d620255678bfe1583a736a87f4771785a9d50684","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-02-28 16:00:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01f0b5afae05012d5a8dcba666ea170c06631d81","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-28 16:29:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nYes, *obviously* that\u0027s what we should have been doing.","accounts_in_message":[],"_revision_number":1},{"id":"ea2d2d98f3a3a7df35c87b532a9bee160dec0376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 21:25:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a970a8ff87414d69b0c5f8d6e3960014 : SUCCESS in 10m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/95424853efa84d648e2b683b168b7eae : SUCCESS in 1h 08m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/576a23d5d24f466c9d4105c43a54715c : SUCCESS in 57m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/72a0fc9ebd9a4192b763d62d6b9c071f : SUCCESS in 12m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4649b82c0911440f99c8162eb6dbecaf : SUCCESS in 12m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/05140ceac347427086517097cfc685ef : FAILURE in 13m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e6401ebbb4e547c9b6147ee9f49fb5c9 : FAILURE in 15m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/32faf90478cb47d3afcd39d96b140a8b : FAILURE in 15m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/28b0381050324670b33a9dd6c2b73d55 : FAILURE in 14m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4c4c8560dcad43a39e67f3e385f11ab9 : SUCCESS in 10m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/86c737907f5548bc888d1b1f8019d676 : SUCCESS in 17m 29s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/33bfbbabe3f24f2b9e7e55834231042c : SUCCESS in 14m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/54e545eada3644dab0390af93bc82cdb : SUCCESS in 14m 28s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/338d2f8fd4ed451dbb1fa655fa1ccc40 : SUCCESS in 18m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/77447d7d863b406f9cbe518db6dbdfb1 : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/537c40e91e6343968e63579c6cb2a476 : SUCCESS in 18m 11s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/40d67802cd73470b8729c1b9687a25ba : SUCCESS in 15m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/234adc8face94937af0a0f199e9706c1 : SUCCESS in 19m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/75642a12d64b4722a1f352f401e63ed5 : SUCCESS in 21m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8f31e0c0153045a3a3d4cb1127810666 : SUCCESS in 33m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/076ebbc78efe4e1f96cba4daa60b4ec4 : FAILURE in 53m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0da8ef3086b845af9c04954f631e9f6f : SUCCESS in 39m 20s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f9a5c20b4ba64bd2928ad4f6dea65e96 : SUCCESS in 38m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/037eb62d23714642b9fbe3a40602a96e : SUCCESS in 33m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9193b7722cfc428fb3e4e97f972932a3 : FAILURE in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eba712e302d4ba4849f6d342f2e1e9a : SUCCESS in 8m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/df51861f469c4ba2a1ee510e706b91d9 : SUCCESS in 41m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8044964b46a4606a3d9915e363e90ac : SUCCESS in 1h 03m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2dc0e45f9c3e758d2e7e7383bb61ced9737de437","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-29 02:30:14.000000000","message":"Patch Set 1: Code-Review-1 -Workflow\n\n(1 comment)\n\nBah","accounts_in_message":[],"_revision_number":1},{"id":"1e0cc26a1783669a68134c4a1bccd74bceb375bd","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-03-02 16:44:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d85475ce3a994794f4e8f1d615b45a5c0e340c0e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-02 17:57:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"97039d101eea4b388446b3181ef0132cdf308cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 21:17:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c653f4ddc5c44269383eb8a9bd481e9 : SUCCESS in 12m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cf5d41fc2df947f59980873c28068480 : SUCCESS in 1h 20m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db616ee170464fa89a53a757d9ab2c16 : SUCCESS in 58m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cb7f254ff08f4412a951adfe64acbee0 : SUCCESS in 11m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2f9083a483784a1a8c008124bd2cee46 : SUCCESS in 11m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e3b1f190f50d4159aa7337aaef607845 : SUCCESS in 14m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d5aa2661ca2f45c984a84400c31e023d : SUCCESS in 15m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8960e966f24d4e2db0f4519d7e0269ab : SUCCESS in 14m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1861cb9254964fb397622442b963e515 : SUCCESS in 13m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b0a73df32fa24258bfb77081dd860d7f : SUCCESS in 13m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54a363dafbdc49ce9de869f9429d00c1 : SUCCESS in 20m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/e5634ceb02994133a3551093ce776dd2 : SUCCESS in 15m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a34c209c18b648ffb6b24af0fa5a1f9b : SUCCESS in 13m 33s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e4545c88ee4547d5a7ef7302cdf735bc : SUCCESS in 17m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/916a409fe2904446bdb2f3d9afe2832e : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f8c67232a076424386ec043a2ec09d56 : SUCCESS in 17m 14s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0607c8675961464c8c7adaffb02b4c32 : SUCCESS in 11m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/72a0cd37b55f4c22a94f25c87337eb51 : SUCCESS in 13m 57s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1faeeddf8d6d4ee49ef4aace6bdd1469 : SUCCESS in 24m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc650129446f40698ead2ec99a2c12a3 : SUCCESS in 32m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/78931803851f47149bdda85c8eb5027e : SUCCESS in 59m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/50b606d2f89544ecb92c8f0b1ccf4d36 : SUCCESS in 43m 00s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/96ba223d973f4b4bbd51fd7621457004 : SUCCESS in 44m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8759308233a94bdf96dd0893068a4bad : SUCCESS in 34m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d9b7206f4ab4d3fb96ef623d62cadc7 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e2b41450cd54f7f860d5b7a7d636de3 : SUCCESS in 9m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee01cc4986654882b82bb35b0ad72de1 : SUCCESS in 36m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5eb68d74c27446a09ad0ef779f23c268 : SUCCESS in 1h 02m 40s","accounts_in_message":[],"_revision_number":2},{"id":"9d4ef0c347f8ace453545ef5718d091a78493284","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 21:39:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9989079fc5c7500aaaa560b9d4f89fbfd0203b22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 04:55:53.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9623cde986964f29b340eb6d88c3dfd2 : SUCCESS in 11m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/62d0ecb471954922b4be6666b9d784f4 : SUCCESS in 53m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ab0fadee8b85443f81290d3081113837 : SUCCESS in 48m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/70fc7ea6d1a44d809dbf7e2a56e05d89 : SUCCESS in 10m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/cc6315702f864f73b35220e502937a3a : SUCCESS in 11m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/312d2ad2f2704261be6c6e203bb10c0e : SUCCESS in 16m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0fba9375f11b44fe9ef5505911bceb22 : SUCCESS in 21m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9d22fad22c594650aa2cb4df672b2c07 : SUCCESS in 14m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/aeb862f754a048a1ae603121edf65600 : SUCCESS in 14m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a124caf254274b0cb915b9b84772632c : SUCCESS in 11m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d61ed9a42e6e4159b83cfe2177e5634c : SUCCESS in 13m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/dd06ffbad084447b85a26a0de2fc0dc6 : SUCCESS in 13m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f1010efd457447d491501523e7655ad6 : SUCCESS in 12m 16s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3fc40c6ef26f4252af051cb3cea7a336 : SUCCESS in 16m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a5f7ebed581549e3ad430d16700c0fa7 : SUCCESS in 10m 48s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6aa64f5910de46e1ab7537a705b80481 : SUCCESS in 12m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/aa30a01fa43f448abf7766483a2eac30 : SUCCESS in 14m 14s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/7138bb4415204d31a6c88e7a2c6db800 : SUCCESS in 17m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f07a0ab1263f401d8aad9c79c6707752 : FAILURE in 47m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/202d650c36f945ca9f2e9fc4fd3abfab : SUCCESS in 31m 35s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/40c72dd5c04e48abb4cabf5431b0634e : SUCCESS in 32m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b2fcd808f07d4d98b568e1bf2c3c3002 : SUCCESS in 33m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e7a170b832f434582fb6d386c6e0562 : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe13cf452a0a42c88168ea34e3f604ff : SUCCESS in 8m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f75f7b50356b4e15bf75f06aa1e5ed45 : SUCCESS in 1h 00m 16s","accounts_in_message":[],"_revision_number":2},{"id":"498626a20f08bec2d0240b2159be277e673ea346","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-03 05:01:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"eb61e86940a821d0f8744ecff1dc1e631e9518dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 06:08:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc30867b05324c8e92394dfe54723b9f : SUCCESS in 12m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/078b5781b2094c4e8c293be3bdfbcd6d : SUCCESS in 1h 01m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b90dcd3c4b1432f996f3b0e2a163ad9 : SUCCESS in 55m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a90f6f72ac374e81ac7c9e7df820929f : SUCCESS in 15m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3a5cc527d8c54280bc4e7c7caf2b2532 : SUCCESS in 14m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/11389b8801974e2ea927a3dd74fa9c8a : SUCCESS in 15m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bbaaf37de0114f038b0095c92f4aea17 : SUCCESS in 26m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fe15e1fc543845e0a3e29716aa343c9f : SUCCESS in 26m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/411e384060394aa59db1ff1c28f15755 : SUCCESS in 27m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/66d2e30a95dc466fa6508a6d6ee0dc56 : SUCCESS in 13m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b029b0a344624ffbadb0a3640387fc8b : SUCCESS in 17m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5365d2d84226426abf87c0937b3fc48a : SUCCESS in 15m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/35ff65cf7f7042e38c8cc0e81d017053 : SUCCESS in 14m 51s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0e9f1c350a3d4f0fa30eef3881a68885 : SUCCESS in 23m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/eb0a682c7846466ea049f3da67d605d7 : SUCCESS in 25m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a7aa813c8012466d80868999975d68dd : SUCCESS in 14m 15s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/33a1605ed99146de8b5aa32aade5a12c : SUCCESS in 15m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/282cd1e3cc95455a8780231e513bc2d5 : SUCCESS in 12m 27s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/dbdd16bc9a1f4db1b9e59f5bf5c0601c : FAILURE in 20m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5681898814be4b699a05cb3fcdf61b29 : SUCCESS in 26m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/26451c90466f4400994605f487f7b9af : FAILURE in 1h 01m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a58d1b70524f48b9983fd4fada560fd6 : SUCCESS in 40m 36s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1800d28ccd704764997c14e300f611ba : SUCCESS in 33m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/af7482842b924f6d9439216a917ea664 : SUCCESS in 39m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/241156fcd78649e9848b13913136735d : SUCCESS in 26m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d016931e6444441d85ff2ca26bad7d22 : SUCCESS in 7m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6df3ac4fe7eb4ec3ac3ea70c8cefa0bd : SUCCESS in 38m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8dfbae5c3eca450a88bf2895dd48decb : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":2},{"id":"35771ab8867125e42651410527f15472b9ed394b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-03 19:23:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d92d0aba4f12366a49acec3006b5033ce5822bc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 00:47:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c77a1a39648447178a492e29f51826be : SUCCESS in 10m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/da84fdf64fbb499fac0ffd4631c492ff : SUCCESS in 1h 00m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/764131ab7c604b2f9b3c393988a1f2ff : SUCCESS in 56m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2c9510a0a24429daddd3407d8b716f6 : SUCCESS in 12m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fbf5152d4d02441f8cc9bfeb58344733 : SUCCESS in 11m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/72f32e982e84424caf0335824dfdcb68 : SUCCESS in 12m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/965d6e066cd249c3916038c6ff794368 : SUCCESS in 14m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f212d9a444bd4ad49220153da736921c : SUCCESS in 14m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/86b6416fffc84364864dfdc952c76417 : SUCCESS in 14m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9d7da0bc7dd845309c632b62440c4809 : SUCCESS in 10m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7d0aaef808404151ab537592fe821e08 : SUCCESS in 12m 57s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1803acc0e5ec4292a5bc55ca0cb9365c : SUCCESS in 10m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/21cd52ef6e4941f18c2ae30ced96e600 : SUCCESS in 12m 14s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5804ae8fc3b547f2841fa92ee275eb4a : SUCCESS in 16m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9ad46f302e63410197e9ce17d06eef18 : SUCCESS in 11m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2fb3de3b019e4544a84be5e84c13c95f : SUCCESS in 13m 08s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/463135b776284917a614e7ebdac36777 : SUCCESS in 11m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b3bc10fc5739431390be275f5433f96b : SUCCESS in 11m 41s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/cdd42f4a54484c8d9aa84cf7d7dad5e5 : SUCCESS in 16m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/884dbd3cdbc846f0be6d1cbd88a600eb : SUCCESS in 21m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/99c0b4f2cfdb42d1b0a7d7bb1928bdf1 : FAILURE in 50m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9cd102dd029645a1bb2974ad42b60eee : SUCCESS in 33m 27s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0619293314be46bc8e9c3878ff0af2fe : SUCCESS in 32m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/163a75c96b904a4db84efdf10b228491 : SUCCESS in 33m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a17a4aa4c5f44c8d81661ce3b7a4d1de : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d04d3a3294483c98a1b0bacb0ec9ee : SUCCESS in 8m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fe8ce02491eb42d2a4a916a66c655d94 : SUCCESS in 42m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72b20f7e32f949839247263744a7de9a : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":2},{"id":"eb2e527f2179ecfa1d1b02ded9b90495945ab14e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-04 04:39:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1bad510ae574c6429ce4dd4349a5017bef0138b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 05:42:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ca7f5367f043148f31350148fd4b1a : SUCCESS in 8m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/89f91636b8844c1092bc35342f1573fa : SUCCESS in 1h 01m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/07efd4a45a5347b3b74a900b0e961367 : SUCCESS in 55m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/becbda7b79cc4edaaf8b8a47c2499e8c : SUCCESS in 10m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/20dfb1dc0bb0402da11d410ea7b65cf4 : SUCCESS in 10m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a2ff5450e3ad4834b3c924c93c455221 : SUCCESS in 12m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d475ac5fcdfb44d68baec5f4b4495079 : SUCCESS in 12m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1aac0fc66ea74277b0295785d7e6b14e : SUCCESS in 15m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4b59504bd4b14db39c09cc391d8c9725 : SUCCESS in 12m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9abe52d804b840b6ae8dc48b2f2f4a90 : SUCCESS in 10m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9530194620d64cac80b2a84b12be677b : SUCCESS in 12m 01s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/9f00e02cff004180997d4d7e5fdbd5ff : SUCCESS in 10m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/95dcdfb9abc7449d8ba6d2fce3821bd4 : SUCCESS in 11m 06s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c29b160325094fee8b665ef26852892f : SUCCESS in 15m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f108dcab5902444284196837ae9ebb9c : SUCCESS in 11m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0991e1c7eca4404d9de9e2c07d74223d : SUCCESS in 17m 30s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/200cf8674e734370837bcc87bd8f60e3 : SUCCESS in 10m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5dda081302df4fa69967bffb25c72bc4 : SUCCESS in 12m 14s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5337e061c2c043ce94da802e20804a24 : SUCCESS in 15m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2293366518a64846bcddacec6797155a : SUCCESS in 18m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/719cab2f5d704052aba318c064832e10 : SUCCESS in 50m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f43664a7a1c34e63bbf34bf448ce87d3 : SUCCESS in 32m 17s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9e005169a6bc462ea932fc58e4c5e3b8 : SUCCESS in 31m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff0e138320594dab8a58de51354fae12 : SUCCESS in 28m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/814a9c85a0fe4fe0a80c51bfb9ce0ebf : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d55f281ff5d429ca45b6c6d66ee2390 : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/184db9ac89cd423a8550a876f9017d35 : SUCCESS in 21m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8604e4595c234b38acf50652b6303280 : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7b1d927b657070e0e9453730512380559e73fa60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 05:43:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f3d348550284532c384f88bacac25a9e20a12b8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 07:29:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/912e47073840459790edf100d248650b : SUCCESS in 10m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b4fbe6ab06274026b86cbc089381d666 : SUCCESS in 1h 07m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b0409067d59490a82e09d996cb45d2e : SUCCESS in 55m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/dca38884603f46c19d2d59bf9c147378 : SUCCESS in 12m 46s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/820629e0caa1419ea1231b47be9763a8 : SUCCESS in 13m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8bc2b9e5a13149e39c59f3c13fad3ce9 : SUCCESS in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/699344386b0646c7afda62a2d50ebecb : SUCCESS in 15m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f1fcef5f48314f19b19bf2fd884f04e3 : SUCCESS in 15m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e6104522e3f74dd096dfbd5d5f81f11b : SUCCESS in 15m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0aa97faba58346a5ba2d19204caad9fb : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/48ab2c1933be4299a61fbb7384c111e6 : SUCCESS in 16m 34s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/2d3e3ef6138f4b318821be04fba2b680 : SUCCESS in 13m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/311627a1af86408ba76b92d2f1f4d180 : SUCCESS in 15m 50s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e49da575d1c14b2dad9bdf740e4deec0 : SUCCESS in 19m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fab5998bf1b84263b794e6a9607822bb : SUCCESS in 21m 04s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/76f9048b782d4b0ba1c40fab59bf79bb : SUCCESS in 21m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0695b6e60bf940c69dc4b630be689104 : SUCCESS in 20m 56s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b67cd59bd4664166b2a71708e9d456ad : SUCCESS in 18m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ede2ababc6de40f0913b83ab488dd0c7 : SUCCESS in 50m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bcb290e7c0a74f9aabdece5200231f66 : SUCCESS in 40m 45s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ec9548b93b1947c88a8317ccce5b27f8 : SUCCESS in 31m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d54a8a01450545728465ac04d9a5ffc7 : SUCCESS in 33m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53a6c32b01c7443b8106e6cba3ceb5ee : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9aa83922dfb43ebb39d02d8570501c9 : SUCCESS in 8m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8862fff700a64b32a4167c2315bce00b : SUCCESS in 1h 06m 28s","accounts_in_message":[],"_revision_number":2},{"id":"8367537454c8e7bdb2124a94ebf58ab6f080b4e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 07:29:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d13b1016f75302aa636e76ff22c5aa640d94bdae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 07:31:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35f0a726c9f4bc2bc044aacdc1b1fd5 : SUCCESS in 1m 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3942ee71e4c8412dafe8b075c1d92837 : SUCCESS in 59s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0d86620a490b4362b112d2c11401b8c8 : SUCCESS in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dc40779307095b8d0b2761b77b9cb2904ec721ae","revisions":{"58409630b9d9f69c62130cf080214d5c1b5e754d":{"kind":"REWORK","_number":1,"created":"2020-02-28 16:00:43.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/75/710475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/710475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/710475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/710475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/710475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/710475/1"}}},"commit":{"parents":[{"commit":"a5afe767581d2cb97cf3690067e6d626c7682c2c","subject":"Revert \"Make rolling-upgrade job non-voting until virtualenv problem is solved\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5afe767581d2cb97cf3690067e6d626c7682c2c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-02-28 16:00:25.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-02-28 16:00:25.000000000","tz":-360},"subject":"Use float consistently for proxy timeout settings","message":"Use float consistently for proxy timeout settings\n\nChange-Id: I433c97df99193ec31c863038b9b6fd20bb3705b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58409630b9d9f69c62130cf080214d5c1b5e754d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58409630b9d9f69c62130cf080214d5c1b5e754d"}]},"branch":"refs/heads/master"},"dc40779307095b8d0b2761b77b9cb2904ec721ae":{"kind":"REWORK","_number":2,"created":"2020-03-02 16:44:53.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/75/710475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/710475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/710475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/710475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/710475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/710475/2"}}},"commit":{"parents":[{"commit":"a5afe767581d2cb97cf3690067e6d626c7682c2c","subject":"Revert \"Make rolling-upgrade job non-voting until virtualenv problem is solved\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5afe767581d2cb97cf3690067e6d626c7682c2c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-02-28 16:00:25.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-03-02 16:44:48.000000000","tz":-360},"subject":"Use float consistently for proxy timeout settings","message":"Use float consistently for proxy timeout settings\n\nChange-Id: I433c97df99193ec31c863038b9b6fd20bb3705b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc40779307095b8d0b2761b77b9cb2904ec721ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc40779307095b8d0b2761b77b9cb2904ec721ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
