)]}'
{"id":"openstack%2Fswift~710543","triplet_id":"openstack%2Fswift~master~I8f65ccd7f2a79d5b877bfbef0274fb7857e21391","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I8f65ccd7f2a79d5b877bfbef0274fb7857e21391","subject":"tests: Use timedelta to adjust dates, not string manipulations","status":"MERGED","created":"2020-02-29 02:59:58.000000000","updated":"2020-03-03 05:03:54.000000000","submitted":"2020-02-29 05:34:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710543-1582954446613-3bf77f9f","meta_rev_id":"10d492bccb81e926a8c2eb3d053137029b6a03ee","_number":710543,"virtual_id_number":710543,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-29 05:34:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-29 04:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-29 04:54:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-29 05:34:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74969ce460898e80471e8a91d2c5e9f2cb04ff19","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-29 02:59:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"249c976344471550856bb9a9c6d9141616e5f930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 03:40:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a478788cbca24f70a242900c5a53bf9a : SUCCESS in 8m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c34b67c8f8a441e4b3f81253be416395 : SUCCESS in 11m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/28f406e784d14720abda527c9dd6831d : SUCCESS in 11m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/081b7d39de604fe1b2f67d62224e1aee : SUCCESS in 16m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/32b4e223fa13428b8c1ada1d9174b8ba : SUCCESS in 15m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/37241957505648538d734bca650e7069 : SUCCESS in 16m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9a4e3f6d84d249918cee5ac73e323f6b : SUCCESS in 22m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/98e7f76d9a5b4648a921412f41f1874b : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fac7710014974bd6a08b9c70576264e8 : SUCCESS in 16m 39s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ff2ac5c9a96a418784b80623449819b4 : SUCCESS in 20m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/379d3b5869614596937c0d8cb6574e43 : SUCCESS in 20m 39s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7996653eff154e2d9be42fa4eacf3ed2 : SUCCESS in 26m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d51583d3ce014dc0afb279174883eb03 : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/14a966b81bf74edbaba8182efb57f40a : SUCCESS in 21m 24s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8fe30c5b822845d2995a576329552974 : SUCCESS in 18m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5c96bf1cdb8540f4b75e6155e20e1548 : SUCCESS in 21m 20s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a4935a1f15514cde9d29a09979682126 : SUCCESS in 20m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ac339d365e44641bd4121b6af58a2c9 : SUCCESS in 36m 00s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/affdc12715214cf396773e161f3bd133 : SUCCESS in 35m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d50efff6bad24ea195d1d6b98211bb85 : SUCCESS in 30m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8785d0683d3446309ec66b2031e13c97 : SUCCESS in 13m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ac6c7a2e064672a2f506c6a5e2eeba : SUCCESS in 8m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5afaae833faa479eb54a12f9fc185d22 : SUCCESS in 32m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8d97f225f52885a70cad6912ee4db6593d0fa92d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-02-29 04:54:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a429a59946062db15f647b7d4c79fdf86c7b7553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 04:54:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95dcc801e07a1aae388207a450c05662b88e421e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 05:34:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98cdff4ab64f45c7bdae6efa0a98fc07 : SUCCESS in 9m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a185aa8305464015a84d05cd1e1ba165 : SUCCESS in 10m 47s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7ced082933b4450ca1afa4ffa50cc017 : SUCCESS in 11m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3c717ac242f74b99bec84684717971ec : SUCCESS in 13m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cac55f4eb2114328a9e98702204930fe : SUCCESS in 13m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0c6ea3af86ff4bcf8b93cd2c02b16fda : SUCCESS in 16m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d1d5a2914e044f4199e0428734da445c : SUCCESS in 13m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9cbe70a318a54cbf8e4fd2c6b404e4e0 : SUCCESS in 11m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/546477e9a463459a9b8f91ef45e72c86 : SUCCESS in 18m 07s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/178be0a4282749efa7b00c27ce412d78 : SUCCESS in 10m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6bde5b232ae54ebaa5b3d41f912e97a7 : SUCCESS in 12m 04s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a12d0879511b48539fc94017372909bc : SUCCESS in 16m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/266784bd6da140c9bb71dc2f070e43ee : SUCCESS in 11m 18s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/346b1614a8dc41b9bd9fc8b27722f202 : SUCCESS in 11m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/35b240ce515a41f286a5a4ff22f6b2cc : SUCCESS in 12m 33s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/01e22125c8944826b0e19caf050212e2 : SUCCESS in 17m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d58a750372bf447998c68955125ea17b : SUCCESS in 33m 02s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/82c9af0f1ff5432a931da2b3fdd6e064 : SUCCESS in 33m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5efc8dfff76f4a26a6863214f1bf9e42 : SUCCESS in 34m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa840f5fe5764aa8a7038fa9f2977081 : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77db87ead82a4c958663c158aa67079c : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"bbaf67e481de962a113d352a5a6aae415c67297d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 05:34:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a6bba336465474b18836b15b4e61a00ece7356cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 05:35:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/911c78e5ec6543d98d3cb314ff49db4f : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b59886108c2b4ffaa6b55d1f10d61694 : SUCCESS in 55s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/06c7123fc5ec472b921108fe72c68b7a : SUCCESS in 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10d492bccb81e926a8c2eb3d053137029b6a03ee","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-03 05:03:54.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit b76ea70b0883b2ba60761ab690eff05979a57279","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55049beda5b9d7038a3604a87f28312d7702ccb2","revisions":{"55049beda5b9d7038a3604a87f28312d7702ccb2":{"kind":"REWORK","_number":1,"created":"2020-02-29 02:59:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/710543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/710543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/710543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/710543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/710543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/710543/1"}}},"commit":{"parents":[{"commit":"a5afe767581d2cb97cf3690067e6d626c7682c2c","subject":"Revert \"Make rolling-upgrade job non-voting until virtualenv problem is solved\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5afe767581d2cb97cf3690067e6d626c7682c2c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-02-29 02:59:32.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-02-29 02:59:36.000000000","tz":-480},"subject":"tests: Use timedelta to adjust dates, not string manipulations","message":"tests: Use timedelta to adjust dates, not string manipulations\n\nChange-Id: I8f65ccd7f2a79d5b877bfbef0274fb7857e21391\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55049beda5b9d7038a3604a87f28312d7702ccb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55049beda5b9d7038a3604a87f28312d7702ccb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
