)]}'
{"id":"openstack%2Fswift~710905","triplet_id":"openstack%2Fswift~stable%2Fstein~I73b5af9645f3f7349144384609bf18a79620e92f","project":"openstack/swift","branch":"stable/stein","topic":"bug/1862115","hashtags":[],"change_id":"I73b5af9645f3f7349144384609bf18a79620e92f","subject":"Ensure domain stored in memcached gets utf8 decoded on py2","status":"MERGED","created":"2020-03-03 05:20:19.000000000","updated":"2020-03-16 23:38:45.000000000","submitted":"2020-03-16 23:37:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710905-1584401830125-c6fdaf72","meta_rev_id":"767d2de3f0c79d9e30c1812dbb820218af1ecf64","_number":710905,"virtual_id_number":710905,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12050,"name":"Charles Hsu","email":"charles0126@gmail.com","username":"charz"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-16 23:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":12050,"name":"Charles Hsu","email":"charles0126@gmail.com","username":"charz"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-12 04:13:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"date":"2020-03-03 05:20:19.000000000","_account_id":12050,"name":"Charles Hsu","email":"charles0126@gmail.com","username":"charz"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-12 04:13:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12050,"name":"Charles Hsu","email":"charles0126@gmail.com","username":"charz"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 05:20:19.000000000","updated_by":{"_account_id":12050,"name":"Charles Hsu","email":"charles0126@gmail.com","username":"charz"},"reviewer":{"_account_id":12050,"name":"Charles Hsu","email":"charles0126@gmail.com","username":"charz"},"state":"REVIEWER"},{"updated":"2020-03-16 23:37:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"976e3d57fe2d970d8e1247152930d776c73eb43f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-03 05:20:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4795adbe33e4f3b2b07c97ef79761aa57fd032d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 05:59:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3615808972144d1c8216700ef247163b : SUCCESS in 7m 44s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6dddbc48201c431bac60ebd8cdff71ed : SUCCESS in 12m 27s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/d4e126622d9746429dddac598d124536 : SUCCESS in 10m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e31eef16ef0448ccbc212a6a71f40541 : SUCCESS in 14m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6643090339ab4080b7cde5c135584a4c : SUCCESS in 13m 20s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dc9cc51bdeda4e5d8d99d80c078133d3 : SUCCESS in 10m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94ee526123714c56b75fa1f95d8e8fde : SUCCESS in 14m 40s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f3ef9ef2d4ff40ccbdba3dffeec45037 : SUCCESS in 10m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b1a8a278a12f4e9e93a98e8f7b4b1f36 : SUCCESS in 10m 16s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/83b5ecefe6f14954b1d338f882c84a8d : SUCCESS in 13m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/47a5dc7db30746aeb01b979f2dfb9a20 : SUCCESS in 19m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2f817895b0b742e6a97f5d13437c83da : SUCCESS in 34m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/30c3b3d4982141709faa5190999e3e4e : SUCCESS in 30m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cbdc34ac7dca452eb64353c62fbe87e9 : SUCCESS in 27m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf122033f6f148a68e8e204bd83723a7 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cea137f551e543aa89af0c024e4a3e69 : SUCCESS in 6m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1bf6c62f32604601a5d15e7f386c4b8c : SUCCESS in 27m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d7bdf2c679bf7f981b464e98003132510fe2d45","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-12 04:13:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dafe7f1a231b026f189a22ceb8aaa76a0e4e1ad4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 04:13:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af0d01d091923609e3c619b64c116fdfca172ee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 05:10:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d83c26caa11f44faabc5afa62a85340a : SUCCESS in 4m 31s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ce4c437d6474859aa0847daa30dcc39 : TIMED_OUT in 41m 13s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/630f157db4164f19b349459fbd80f3a6 : SUCCESS in 21m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/90ff85f42cd64caca21646e74460dccd : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/37ee4d6a98c940cab5719f20e31f8ccb : TIMED_OUT in 41m 07s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ebeff570b70848b2a80c7ebfb4e4d91b : SUCCESS in 8m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5a2642e73bc148d395a98485b27fc0d3 : SUCCESS in 9m 51s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/afdc9e0503ae440db759c3b43625da95 : SUCCESS in 19m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1254b5725de64d759e838dfd39aa13b7 : SUCCESS in 35m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9994c2721b14c42a4c9b760e251b92c : SUCCESS in 28m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5b9e6ca88d8d4b16880a852af5723221 : SUCCESS in 28m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78b5e9b3a30d40f79351600ba172cb16 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef2d32be96064b7aa9b296c5527a58b0 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0f77b2969f82e97dd8cec3b4bf5a9b77f0d49f4f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-16 22:09:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"10d366bca31f21b46d60c8702d4f7c609787639a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 22:48:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad97797f12b5442eb87d11e4c24c7ce4 : SUCCESS in 5m 41s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68adec70de3f4b819c6c299331b0a8d6 : SUCCESS in 17m 27s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/a83b60bf134d4e9a97bbad401de7b757 : SUCCESS in 20m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/26114f02d44048eaab703164f98325d3 : FAILURE in 12m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9157333ddd1a499eadc7025a51a639b4 : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/10c394d8bcbf48d2b9f79983441e52d4 : SUCCESS in 22m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bda3092afb9c436581ab23b165c647db : SUCCESS in 22m 15s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/79d2643e44ee42058f3f00c0801e969e : SUCCESS in 21m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fdfa4644a6a64c57a5bd368ccad4dbf2 : SUCCESS in 22m 50s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f0053bdc1f1c424584c2564b988a7682 : SUCCESS in 25m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/855addabcc2a4925a1d4204f79c8771c : SUCCESS in 25m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7f76181d0f034be5ae6b0ee2347f4624 : SUCCESS in 30m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/47ab9c8591b64ac7bcb4abfd3f16299b : SUCCESS in 26m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d505f051d1f4a71a8cf5f1b1e91273c : SUCCESS in 24m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af4a8a99f6254240959bb2a37acbb951 : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/394799f63a574492b3a69e26719915f1 : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d042b50372d2477d89c01d43aef2a591 : SUCCESS in 20m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aadfbfc42eed070e96001e20da47e1e0850bf856","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 22:49:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af4e4cead18758e260ae8e9b8bebf41c7df790d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 23:37:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ad5a1d45b94027b16edff003685570 : SUCCESS in 6m 23s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/020301546af4458e8b79631d38d5c6be : SUCCESS in 16m 35s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/368baf48a0934a2fb18d0102d474e7ca : SUCCESS in 13m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/103ca54c601e46d8bf905e31b196417c : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20dd355562f8449dabeb9f5f71e970ea : SUCCESS in 16m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/dac0c24a5cd9467eb004536568ad9efa : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c749936343d24d9185d703b8812fed89 : SUCCESS in 16m 46s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/812c269968c6493483a907d18a29b835 : SUCCESS in 18m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a436b8f6c19542a7b8736c4b6180259a : SUCCESS in 35m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b376fd3e336940b2a68070db0ed34ac0 : SUCCESS in 32m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c957a5be2c5544bdb316cfb15f5ffc45 : SUCCESS in 32m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8da42bd2dd334109a55b7a26528e8e98 : SUCCESS in 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a5168ac63b54066abe92a9a9332f569 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ddf3360035414d372fb55affe27bc03c7e83cdfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 23:37:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a385c18be0116ae00a5f12ec1a89880af4d5c9a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 23:38:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a55908633d9b40f69a86ebdee6ddf21d : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24bb15c2914c23b02c801eb29e3da452b81ebf3d","revisions":{"24bb15c2914c23b02c801eb29e3da452b81ebf3d":{"kind":"REWORK","_number":1,"created":"2020-03-03 05:20:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/710905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/710905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/710905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/710905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/710905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/710905/1"}}},"commit":{"parents":[{"commit":"778767f324e82dd7a075306cf8468aabbbc363b6","subject":"s3api: Better handle 498/429 responses","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/778767f324e82dd7a075306cf8468aabbbc363b6"}]}],"author":{"name":"Charles Hsu","email":"charles0126@gmail.com","date":"2020-02-06 07:35:11.000000000","tz":480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-03-03 05:19:43.000000000","tz":-480},"subject":"Ensure domain stored in memcached gets utf8 decoded on py2","message":"Ensure domain stored in memcached gets utf8 decoded on py2\n\nChange-Id: I73b5af9645f3f7349144384609bf18a79620e92f\nCloses-Bug: #1862115\n(cherry picked from commit 61bf5ee1c428410b70bfab0a6eae361df948d03e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24bb15c2914c23b02c801eb29e3da452b81ebf3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24bb15c2914c23b02c801eb29e3da452b81ebf3d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
