)]}'
{"id":"openstack%2Fswift~712984","triplet_id":"openstack%2Fswift~master~I5068ab4ca8e7d4237fe66c5eed7ae3123485b7fe","project":"openstack/swift","branch":"master","topic":"py2-uc","hashtags":[],"change_id":"I5068ab4ca8e7d4237fe66c5eed7ae3123485b7fe","subject":"Use local py2 upper constraints","status":"ABANDONED","created":"2020-03-13 15:44:42.000000000","updated":"2020-05-13 10:17:13.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cb6c3b657938c045e294704cdc4ca3bc20a13c03","_number":712984,"virtual_id_number":712984,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2020-03-13 16:17:31.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-13 17:20:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2020-04-15 19:56:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 17:20:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-15 19:56:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7dde6d6b83fa48a109c4f8921fc7b544c0d09a3d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-13 15:44:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"913f38714191d8299770ecc2d877f948ddfb5053","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-13 15:45:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3de477df33c57e10dfe3bdf9256993441520f840","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-13 16:15:07.000000000","message":"Patch Set 1: Code-Review-1\n\nhow will this be kept up to date, or is swift fully managing their own constraints?","accounts_in_message":[],"_revision_number":1},{"id":"5ca13e74be487c093ba96dd940ed0cff12289c1c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-13 16:17:31.000000000","message":"Patch Set 1:\n\n\u003e how will this be kept up to date, or is swift fully managing their\n \u003e own constraints?\n\nThat\u0027s a good question that we should probably discuss with the team. I think like lower-constraints.txt, this would need to be something managed by the Swift team. So it is probably fine staying where it is, and packages can be updated if and when there is a need to pick up a newer release.\n\nWe could maybe look at running our requirements repo update scripts against this too, but I think that may require some tweaking to get it to pay attention to py2 only things once we drop that in the global requirements.","accounts_in_message":[],"_revision_number":1},{"id":"c5efefac1f96363baefe66a9315acf9e59c492ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 17:20:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/41267dcb96b847438d3c081826bbcbf6 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9452a50e58547c992eb038fdc371265 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7563bd084cc4be988fbec6173c93c96 : SUCCESS in 8m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7cffa94394154c9389e96b5ba407dc33 : SUCCESS in 1h 24m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a140bccde08243e68d5dcffd3ebc29b9 : SUCCESS in 1h 11m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b4579300400b424f9ebeef090272e27a : SUCCESS in 10m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4b9858b5c03e4cab825f87ef08310497 : SUCCESS in 12m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b51fb8cb66ef4d6dadfe7deef2f07d67 : SUCCESS in 12m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a69a355f9a948918b0ab5296f6f7ecb : SUCCESS in 14m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ee7c6d7ca7be4d3fbc6059ca9a606097 : SUCCESS in 15m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9bf9e2db35514c3a96d240f6e93cbd34 : SUCCESS in 16m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/46c7402221ce458c8f43f4d44d34a976 : SUCCESS in 13m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b75b58cb9e284237aeee830140be708e : SUCCESS in 15m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/63d3f06932bc4a97a2fa0754f84236d5 : SUCCESS in 13m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7718097667c148d493fbbb9a84f3e673 : SUCCESS in 15m 10s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e1f08c642d4741bebf1ffc6880791a46 : SUCCESS in 20m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/38dea39e451b4ef9b44df4d1f98a9f7f : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b14c2ec65c0e40eabb22e4b45f78fc5b : SUCCESS in 17m 39s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/937b0ab8c6c7436597803c676c20ba6d : SUCCESS in 13m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/88bed8e53bb74887a57031c1b4b0aa54 : SUCCESS in 15m 24s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5a4e37f794a143c8a04fe62f3b0b2b33 : SUCCESS in 23m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d02fa385abc74306ac9253a3a66c3885 : SUCCESS in 26m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/68bedeacce7a402981a4e8bbd0c2606b : SUCCESS in 50m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e24219552e654ed7b11574129c90db17 : SUCCESS in 28m 46s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/eeffea4e89bf4c2b8c8e8533488174a4 : SUCCESS in 29m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b82e1e75812f41769c79d2d288ec675a : SUCCESS in 32m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93a15e3326b44e4f83a2698d55c693c2 : SUCCESS in 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1fb095634d647e8bce7cd8ba06f31c9 : SUCCESS in 7m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c5942691c4df4a08b150788fd30076c3 : SUCCESS in 25m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efdaacda68ea4136a086f4b26083ac8c : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":1},{"id":"533ad4c5ea46d55faa29c3b711d8bf730f8d5827","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-15 19:56:46.000000000","message":"Patch Set 1: -Code-Review\n\nok, just wanted to be clear.  I also expect the py2 constraints to be on the swift team.","accounts_in_message":[],"_revision_number":1},{"id":"b6fde932f50304c277067d8a0b310a0faac0d7ed","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-13 10:17:13.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/724422/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"440e63d96bff965e55938c801ef186a0e2448e12","revisions":{"440e63d96bff965e55938c801ef186a0e2448e12":{"kind":"REWORK","_number":1,"created":"2020-03-13 15:44:42.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/84/712984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/712984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/712984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/712984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/712984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/712984/1"}}},"commit":{"parents":[{"commit":"9d4dc29fb37db1ad767eae3e39468d76e996af63","subject":"Merge \"Apply limit to list versioned containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d4dc29fb37db1ad767eae3e39468d76e996af63"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-13 15:43:07.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-13 15:43:07.000000000","tz":-300},"subject":"Use local py2 upper constraints","message":"Use local py2 upper constraints\n\nSwift is the only OpenStack project as of Ussuri that still supports\npy27. In order to be able to clean up the global upper-constraints file,\nand to help avoid accidentally breaking Swift by changes to that global\nfile, this attempts to use a local copy for py27-only upper constraints\nenforcement.\n\nChange-Id: I5068ab4ca8e7d4237fe66c5eed7ae3123485b7fe\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/440e63d96bff965e55938c801ef186a0e2448e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/440e63d96bff965e55938c801ef186a0e2448e12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
