)]}'
{"id":"openstack%2Fswift~713059","triplet_id":"openstack%2Fswift~master~I45298d690fa0ac6dad5dd792bffd73c09f86ad2f","project":"openstack/swift","branch":"master","topic":"performance/md5_alternative","hashtags":[],"change_id":"I45298d690fa0ac6dad5dd792bffd73c09f86ad2f","subject":"WIP: Make the hashing algorithm configurable","status":"NEW","created":"2020-03-13 20:06:34.000000000","updated":"2020-03-18 22:03:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dbf66f207a108b6188af5d473c08ef84cc66856d","_number":713059,"virtual_id_number":713059,"owner":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-13 21:10:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-03-13 20:08:29.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 21:10:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e38c288e9e48134988197d3781f059576b82412a","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-13 20:06:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c97602a71648ab6f8cd04808e141af6a091e716b","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-13 20:08:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5d58cde260b2606f9912dcf647480d267b86ea17","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-13 20:08:29.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"62c20082d5e158d15039c5b43b250f37a9803515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 21:10:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d861cac3bb98481fbd1877f07315c4d8 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/201b34abf6004c748211e113f29051ce : SUCCESS in 9m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/05015a696bfc4b939c5dfb6f4e76501d : FAILURE in 3m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6e4c13d450e547e09b27096634fce18c : FAILURE in 50m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ef97da412c5649e28d86bbc902125fa9 : FAILURE in 26m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e386a1d0d5af443e8b904bf7feb0f515 : SUCCESS in 9m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7d7a73f735454a56838d280891fd6e54 : SUCCESS in 12m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c79da13b09e247c2a0321c5399fbe338 : FAILURE in 14m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1232e6becc51468d91b73bf4420712e1 : FAILURE in 15m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/da66f45071284f99ba84c9983e4e3b1c : FAILURE in 14m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f358ec8d6f0e4cecb54c5eb1a0827e95 : FAILURE in 19m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/afc9210a511a4d63832e1a8fa5ae4677 : FAILURE in 4m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a0f77b028fff4f869ab397ffa7de2e87 : FAILURE in 5m 05s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/201491441ef64df8af40b61221ad3fa6 : FAILURE in 4m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf0255f675824f70b8864566de5bafbd : FAILURE in 4m 53s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5407d2419ff8407295b0db7490ec8bd9 : FAILURE in 4m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d043647a81e244bda66bcc16c77ec71f : FAILURE in 4m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4663c3df619c49f7addea44c6e8b82f8 : FAILURE in 4m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0de2e9f4a84f4241acc33ccad2926404 : FAILURE in 4m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/75ccedf5838d4b0f98b17acd090af4b5 : FAILURE in 7m 13s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/01c508ea5aa54c8eab3984286deda173 : FAILURE in 4m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/de913916adaa46529c6a9375a0fb43e3 : POST_FAILURE in 7m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/de865d069080425a84be30db8322c4f2 : TIMED_OUT in 1h 01m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4df69492ed4440a9ad02db95aa7f5759 : RETRY_LIMIT in 17m 54s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/36808302b57b43b181b7068c41e05f13 : RETRY_LIMIT in 16m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d8ee10991999431594678247b8c7fc21 : RETRY_LIMIT in 18m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c31b949affb84036ba0f6bc9f9de6a37 : FAILURE in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/693bbc8d16834adfaa3f4b1fd8430a5a : FAILURE in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ab4686b219fb461db48429158c06f0ce : SUCCESS in 27m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c766023e6124b7e9d438f423c1353ea : FAILURE in 26m 29s","accounts_in_message":[],"_revision_number":2},{"id":"c2a44f1c2770f079b8b09e087fde5738c3c27812","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-18 22:03:16.000000000","message":"Patch Set 2:\n\nOoops, I forgot one file. i\u0027ll upload it asap","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0bd43b64e06a90c5ffd1baaa239056963d9cd2ea","revisions":{"6459750ef0f249aa74e4e1759637888736ae503e":{"kind":"REWORK","_number":1,"created":"2020-03-13 20:06:34.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/59/713059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/713059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/713059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/713059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/713059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/713059/1"}}},"commit":{"parents":[{"commit":"9d4dc29fb37db1ad767eae3e39468d76e996af63","subject":"Merge \"Apply limit to list versioned containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d4dc29fb37db1ad767eae3e39468d76e996af63"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-02-18 19:40:00.000000000","tz":-300},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-03-13 20:06:07.000000000","tz":-240},"subject":"md5 first version","message":"md5 first version\n\nChange-Id: I45298d690fa0ac6dad5dd792bffd73c09f86ad2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6459750ef0f249aa74e4e1759637888736ae503e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6459750ef0f249aa74e4e1759637888736ae503e"}]},"branch":"refs/heads/master"},"0bd43b64e06a90c5ffd1baaa239056963d9cd2ea":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-13 20:08:12.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/59/713059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/713059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/713059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/713059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/713059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/713059/2"}}},"commit":{"parents":[{"commit":"9d4dc29fb37db1ad767eae3e39468d76e996af63","subject":"Merge \"Apply limit to list versioned containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d4dc29fb37db1ad767eae3e39468d76e996af63"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-02-18 19:40:00.000000000","tz":-300},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-03-13 20:07:01.000000000","tz":-240},"subject":"WIP: Make the hashing algorithm configurable","message":"WIP: Make the hashing algorithm configurable\n\nAllow to select something else than MD5 to ensure data integrity on disk\nand on wire\n\nTODO:\n- fixe tests\n- write tests for new middleware\n- update crypto middleware\n\nChange-Id: I45298d690fa0ac6dad5dd792bffd73c09f86ad2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bd43b64e06a90c5ffd1baaa239056963d9cd2ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bd43b64e06a90c5ffd1baaa239056963d9cd2ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
