)]}'
{"id":"openstack%2Fswift~713910","triplet_id":"openstack%2Fswift~feature%2Flosf~I0e4ab89bbe3d0aee305272fdd920122e80530b8b","project":"openstack/swift","branch":"feature/losf","topic":"dev/alecuyer/feature-branch-catchup","hashtags":[],"change_id":"I0e4ab89bbe3d0aee305272fdd920122e80530b8b","subject":"Fix get_stats","status":"MERGED","created":"2020-03-19 15:47:50.000000000","updated":"2020-03-20 00:10:13.000000000","submitted":"2020-03-20 00:08:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713910-1584662904428-603078c3","meta_rev_id":"29a7a94d1a89881ff01660dd9a0e239f76f58a7d","_number":713910,"virtual_id_number":713910,"owner":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 00:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"all":[{"value":2,"date":"2020-03-19 21:41:49.000000000","_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"all":[{"value":1,"date":"2020-03-19 21:41:49.000000000","_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 21:41:49.000000000","updated_by":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"reviewer":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"state":"REVIEWER"},{"updated":"2020-03-20 00:08:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b83caacf3fdce632f5bf686c404d15f6a41afbe1","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2020-03-19 15:47:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe500b7b700d89cfaa23d715e575fba3f56de2ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:10:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6a198c7f6e047cea65e795fdf300a33 : SUCCESS in 12m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/783d01f3bfde4837a407609e267a468d : SUCCESS in 14m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f508a4f21b9342c69c0333564728ef74 : SUCCESS in 11m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/05ca58528a884f15be42927a481e4747 : SUCCESS in 15m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d9cd9508d41943f7822a2702196414e4 : SUCCESS in 14m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a7be07ecd98c43e2bf6febf0558037f7 : SUCCESS in 18m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f4c4a5d097574c40920a3d953b53143a : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9c30e51759dd475b825f594b225eae47 : SUCCESS in 16m 31s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/34fcfb777a6343ebafdbe5465aa4dfd1 : SUCCESS in 14m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/37eb4d902fcb4773859d09ae5644b417 : SUCCESS in 13m 20s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a74f3dc342854edf9da2cefae29b2394 : SUCCESS in 18m 05s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/92ba171e73bc44c490c40722d246c6d1 : RETRY_LIMIT in 8m 09s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/77c4f4a29ca742d4a23a4fec404d1bf1 : SUCCESS in 13m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e90f495182204cf4a4a0ca12c34b1a2a : SUCCESS in 16m 13s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2f9794d10fe748729247d287e4b46ee7 : SUCCESS in 13m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a9c74e1665ce44eb9be6f1c83ba4858e : SUCCESS in 16m 58s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f7fb0ebf62dc42e9aa0fb2f7d95569b4 : SUCCESS in 23m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e6d3572127e3476dad7493b82342c1c4 : SUCCESS in 21m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0a300d8a7a88436592095a51df6e7ce2 : SUCCESS in 50m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c539975ce58c4aad9b9cae32b31c19a4 : SUCCESS in 31m 07s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c783ee4444da481fbc7db8a2157d78b6 : SUCCESS in 31m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6427957ce0a64b70be8d7c2b06ff6730 : SUCCESS in 36m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3e0184f11c7494393e316345c8cd38a : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10bfb001b17c4627b1f835903f8a82f0 : SUCCESS in 6m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/61e494a9279b48b49c3c5ac309e67bbf : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d30f795d621b32701188a236f2af5ace51896409","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-19 21:41:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b417b6afb20a53b279d75a28d9df3da0138fddc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 21:42:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"170972497cab19d78048d8b5e3c80361a7678c28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 00:08:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1c7d6ca942c41f9b306f5c10cd49fae : SUCCESS in 10m 45s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e2fb65d8b8784c70aa4badb9ad0ddb7a : SUCCESS in 11m 22s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/44ed6d4e11864a82aac7c97f82487784 : SUCCESS in 13m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/37c4374bf9d54a86a5006d2ec8615f06 : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c870c3294ed04036b12877dbc8a5de77 : SUCCESS in 12m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/81a62b9d213e4fd89c69bddc2323198f : SUCCESS in 12m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7f519a9af7ff4cf59f45ae0783521622 : SUCCESS in 13m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2858d6d10f15496796e76dc99fe9d1b4 : SUCCESS in 20m 21s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5add213dd2ae456b8fc031f89d789131 : SUCCESS in 14m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/06f11aa8da754217a32cad9261da1259 : SUCCESS in 18m 11s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0c44b4fc6606449bbc22ac669d16b1ec : SUCCESS in 23m 57s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/8d0001d5a514413eb1d3b75b7837893b : RETRY_LIMIT in 7m 18s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/85ed54dc74f1442ca0d5bc52a863ef42 : SUCCESS in 16m 14s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8a55e65845d54b0ca3fb3a8a19b8f7f9 : SUCCESS in 14m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4794f3cdc5cb44bbabd2c7cf55505206 : SUCCESS in 18m 45s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a6f2285ab54048afb4ddff698e7acdb3 : SUCCESS in 21m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6a9b18b5cfd049f7acbaa56481b6136c : SUCCESS in 47m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a0b86fd9d514e0fac5ab1f826b23a27 : SUCCESS in 29m 10s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8f59ef7dba954fb4adc0aa2dab22ee98 : SUCCESS in 32m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e7f428fcb4b9470c9c64def1b807beb2 : SUCCESS in 38m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cf972139c07437490df5b68fd565f5c : SUCCESS in 15m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a923937a142c4d37a8adf4687c715ca1 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8f47fbf036fd8ab58bc8f0b3e403f45915407482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 00:08:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6d4e9cbdcb107796bfbf46c9994bbb4636ed304a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 00:10:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b9379249f64d53a4e8832dd3497ae3 : SUCCESS in 1m 29s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6c37ffa69e4348a09ae1c28abf9de141 : SUCCESS in 51s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/747c46e8915e40be981205ea6bb78022 : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6e5f407b46f6cde08c9945fcef96535c5d42483","revisions":{"b6e5f407b46f6cde08c9945fcef96535c5d42483":{"kind":"REWORK","_number":1,"created":"2020-03-19 15:47:50.000000000","uploader":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"ref":"refs/changes/10/713910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/713910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/713910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/713910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/713910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/713910/1"}}},"commit":{"parents":[{"commit":"08e7521b9304caa0f1c3b853562d1ea76251aaa3","subject":"In _get_vfile(), handle the case when a referenced object has been punched","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08e7521b9304caa0f1c3b853562d1ea76251aaa3"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2020-01-06 21:03:49.000000000","tz":60},"committer":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2020-03-19 14:25:48.000000000","tz":60},"subject":"Fix get_stats","message":"Fix get_stats\n\nThe response is now the stats themselves (remove \".stats\")\n\nChange-Id: I0e4ab89bbe3d0aee305272fdd920122e80530b8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6e5f407b46f6cde08c9945fcef96535c5d42483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6e5f407b46f6cde08c9945fcef96535c5d42483"}]},"branch":"refs/heads/feature/losf"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
