)]}'
{"id":"openstack%2Fswift~713916","triplet_id":"openstack%2Fswift~feature%2Flosf~Ib6cbcd3f84e800a843519c01ba4821d575166a2d","project":"openstack/swift","branch":"feature/losf","topic":"dev/alecuyer/feature-branch-catchup","hashtags":[],"change_id":"Ib6cbcd3f84e800a843519c01ba4821d575166a2d","subject":"Remove migrator.py","status":"MERGED","created":"2020-03-19 15:47:50.000000000","updated":"2020-03-31 23:46:49.000000000","submitted":"2020-03-31 23:43:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713916-1585698200232-afcf8495","meta_rev_id":"645b32f0ad36a2b6b0c5e9b35d2313d7fcdc684b","_number":713916,"virtual_id_number":713916,"owner":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 23:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"all":[{"value":2,"date":"2020-03-31 18:24:33.000000000","_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"all":[{"value":1,"date":"2020-03-31 18:25:40.000000000","_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-20 05:45:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-03-31 18:25:40.000000000","updated_by":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"reviewer":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"state":"REVIEWER"},{"updated":"2020-03-31 23:43:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9f6fe9185bd3f9c22f86f5ddee3765d8131ff40","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2020-03-19 15:47:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a71fe6a73c25e932cb0141b5336c918d57fd4734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:51:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6b68570f34460cbee6608e9a2551a6 : SUCCESS in 14m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/89d1b6194d5e455aa47f2d87758fe43f : SUCCESS in 13m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/19032b8cbd8c4695a8bec6ac08fa9cca : SUCCESS in 14m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83c0f73f85a345be995e0205f20f0572 : SUCCESS in 14m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d7d0000798a426399663d79ef758f72 : SUCCESS in 14m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8960f42775ba4b0fb15b705f818c0541 : SUCCESS in 14m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f64b29864c9a4052ad136bd257ce13dc : SUCCESS in 12m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/537bd6ca33064323b083e90a97754999 : SUCCESS in 14m 13s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ba9937caee4348329daf52169bfca594 : SUCCESS in 13m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/118b84471bc543dc90a69c98180a616b : SUCCESS in 12m 30s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/8b3f8e87678b403c8092d06659d0c014 : SUCCESS in 15m 48s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/884ae72ec21f4041bac9a3ad39d15666 : RETRY_LIMIT in 11m 27s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9ff2f24bed4a43bd89f86f33e2f3d998 : SUCCESS in 12m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1b56eebb6b2b4f94841a4bec13fbe7ef : SUCCESS in 16m 48s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/82dfa549d6cf4f3e8bbb23d84a42f25f : SUCCESS in 14m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dfe8e5630c6e4441b5ef10480cc27f2f : SUCCESS in 12m 44s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4287c7a3e11c439b959b0706f451579d : SUCCESS in 23m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ac5fd750f6f43b7a8210e45c59dda19 : SUCCESS in 23m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cd53ed4bf34945cd9943571c291e320c : SUCCESS in 55m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4cc31ab119a4410a80f334117423506 : SUCCESS in 33m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/71132c888a9d459fb472ab88689306c2 : SUCCESS in 31m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8b12c2af36b24ed78b7e1bdb6f98adb3 : SUCCESS in 37m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a738d0fb597b4d16be582081e91b57c3 : SUCCESS in 15m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51151c8274da49a382d6ed021c760071 : SUCCESS in 11m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f9731e7b06949309afbdcf295a4af04 : SUCCESS in 40m 04s","accounts_in_message":[],"_revision_number":1},{"id":"55183dfea07643dce3b40e56d604f92658c15d9c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-20 05:45:09.000000000","message":"Patch Set 1:\n\nShouldn\u0027t we also remove object-migrator from swift/common/manager.py?","accounts_in_message":[],"_revision_number":1},{"id":"b30fb8c6e7e67551b1e76c7183142d9679e3af21","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-30 14:08:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8169e8750b63b54be5284084e16d7e15a897369","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-31 18:24:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8e0565703502217f05737c66b1773b843b9eb8e1","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-31 18:25:40.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6e287fe1fce73506f87231b37eceb21df23592f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 19:46:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f2f02ed6728461182c62536409c623b : SUCCESS in 9m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6f1973968884792857366b2cc5b165b : SUCCESS in 12m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5e5dff20b8c2420d898ed9fd2c0aee2d : SUCCESS in 12m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/805293ea17ca4dbd98c033cef3aa2ba1 : SUCCESS in 16m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5962ca1e8c5f4d5eb48f299dada9399a : SUCCESS in 14m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4fe83d64f68b4330ab32b4e3a379d8d1 : SUCCESS in 15m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d12ce9a3868c46939baaa7c55b12c833 : SUCCESS in 15m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a76fda7fcdc74d09b2a5ac130029d09b : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/280b3e08cc214de2a18277fe3ab23825 : SUCCESS in 14m 31s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/781044b70c2e491980251372b0b08ad3 : SUCCESS in 12m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5e396b6618f643d4a3af955838dc1593 : SUCCESS in 12m 41s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/736282ec62964884b9fed1eac60c6e03 : SUCCESS in 19m 41s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/2d86250b43744064ba35de802046627b : RETRY_LIMIT in 6m 10s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6af834bd53974cf6823915b42408e489 : SUCCESS in 12m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c646ec0ddf6a4b43986164553a9cc6aa : SUCCESS in 13m 27s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e594db824cc6433f89dacbf9030a9753 : SUCCESS in 13m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9c3621ed04b44508a14931ff7aa507a6 : SUCCESS in 12m 22s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/28bae6c6b0144916956017084640b788 : SUCCESS in 16m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8df8781de65e472582e858b93edd7943 : SUCCESS in 23m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ce048923baec4703be1ab0db7bf8e807 : SUCCESS in 52m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a23c4bce6e064ad7b926a91e4adcbe0c : SUCCESS in 30m 59s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/882fa55059bf4ab3930878f3d8a65ce5 : SUCCESS in 27m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c28abbb43d254f5dbab55e557c739ad6 : SUCCESS in 31m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53ff2a232bae43cbba99261d7cf82ffe : SUCCESS in 15m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed440263eda342918e2a9e27734264c6 : SUCCESS in 7m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a1b0aadae35e45bb9118764ed70c11fd : SUCCESS in 25m 52s","accounts_in_message":[],"_revision_number":2},{"id":"92920b4fb77b2076503f4a0c3e4d0dccbdf08054","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 19:47:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"610be248be43c4fb059e67ab17cd084f30cf9826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 23:43:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92dd096e787f4441aa7d4a269e5aba84 : SUCCESS in 9m 18s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9dc4439c53ad402fb57f4bb7c72ca3a3 : SUCCESS in 10m 39s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8bcd33470182478fa3be93e0b72c1459 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/adde4df0bf32403fade43d16258d7411 : SUCCESS in 15m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f2105321726c4d80bf2596b55e2a4b94 : SUCCESS in 13m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cedb0004d67843d18d18ae9373680053 : SUCCESS in 15m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/94863b20033d46fb9bfeb6e113f173d2 : SUCCESS in 14m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e3b879d1cdf74e38ac3d9b6479790ff0 : SUCCESS in 10m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/70d256bfd3624bfab63e416679b72f5d : SUCCESS in 12m 26s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8a627f1df0974a82bc6b0de019dec75b : SUCCESS in 11m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cc60bdd3e586450d92c0d5ce37fc4b4a : SUCCESS in 15m 20s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/82f17d6059034be2bcc6fb5a7700aee1 : SUCCESS in 16m 43s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/438a1096523f4f64ad74cda6d4d24202 : RETRY_LIMIT in 6m 15s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3b7e789afb0141e59ef719790ec4e1d3 : SUCCESS in 10m 25s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d625171b03c6407fb24ac8bc6044c84e : SUCCESS in 11m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3b14502222b046aabf4379ba9944510e : SUCCESS in 13m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/19adc3f9e5fb4549acb7c3f580a77a0b : SUCCESS in 16m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e3ed4a6c2096485a93aeffcfd971de6d : SUCCESS in 51m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fffb678438754af4bcf0b62fd0e1f108 : SUCCESS in 32m 28s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f7fe493e503a4a82a676e878071fbf2f : SUCCESS in 30m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0cabb628ebd14e0e8cddf52fec2dc1e9 : SUCCESS in 29m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56f9b8ad91ad4847ae8615a54609427b : SUCCESS in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec7faa9141b14fe0a47236803f608563 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":2},{"id":"95c7081ce454f4ed4445b70281de2806ded8c517","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 23:43:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8b3ca8d089c5975c4db709f3b544f2e30e9d2f80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 23:46:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7dade5dbc0449388b2403b10693b99f : SUCCESS in 1m 23s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/22cac7a8bef1401bb818049819df5795 : SUCCESS in 1m 10s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/763f5eaaa22b48a78fc92717ad22d52a : SUCCESS in 1m 18s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"55da9a615b80e29a52ab7c7546d93a15c05e1061","revisions":{"dfe0cda2bdb1d91049dab9e2635ac97e46f7fa23":{"kind":"REWORK","_number":1,"created":"2020-03-19 15:47:50.000000000","uploader":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"ref":"refs/changes/16/713916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/713916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/713916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/713916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/713916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/713916/1"}}},"commit":{"parents":[{"commit":"c1a78082a999d4a0564de866f911f7ccf4a1aed8","subject":"Add missing option in for the volume check script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1a78082a999d4a0564de866f911f7ccf4a1aed8"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2020-03-19 15:41:19.000000000","tz":60},"committer":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2020-03-19 15:41:19.000000000","tz":60},"subject":"Remove migrator.py","message":"Remove migrator.py\n\nIt was a preliminary version, cannot be used as is to migrate from a\nregular diskfile to LOSF\n\nChange-Id: Ib6cbcd3f84e800a843519c01ba4821d575166a2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfe0cda2bdb1d91049dab9e2635ac97e46f7fa23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfe0cda2bdb1d91049dab9e2635ac97e46f7fa23"}]},"branch":"refs/heads/feature/losf"},"55da9a615b80e29a52ab7c7546d93a15c05e1061":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-31 18:24:33.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/16/713916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/713916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/713916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/713916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/713916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/713916/2"}}},"commit":{"parents":[{"commit":"a1c9b50802581050b86e1c512cf13414155a0855","subject":"Add missing option in for the volume check script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1c9b50802581050b86e1c512cf13414155a0855"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2020-03-19 15:41:19.000000000","tz":60},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-03-31 18:24:15.000000000","tz":-240},"subject":"Remove migrator.py","message":"Remove migrator.py\n\nIt was a preliminary version, cannot be used as is to migrate from a\nregular diskfile to LOSF\n\nChange-Id: Ib6cbcd3f84e800a843519c01ba4821d575166a2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55da9a615b80e29a52ab7c7546d93a15c05e1061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55da9a615b80e29a52ab7c7546d93a15c05e1061"}]},"branch":"refs/heads/feature/losf"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
