)]}'
{"id":"openstack%2Fswift~715492","triplet_id":"openstack%2Fswift~master~Ib79bfd23b7b6f40673396aafa47a0f8c4f533a97","project":"openstack/swift","branch":"master","topic":"715492","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-05-19 11:06:56.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-05-19 13:26:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib79bfd23b7b6f40673396aafa47a0f8c4f533a97","subject":"s3api: Better logging for non-JSON when trying to do bucket listings","status":"MERGED","created":"2020-03-27 17:31:35.000000000","updated":"2023-05-19 13:27:27.000000000","submitted":"2023-05-19 13:26:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"715492-715492","meta_rev_id":"d9e4f35e262a5ef88a3c85bf10574826b4304693","_number":715492,"virtual_id_number":715492,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-19 13:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-05-19 11:06:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-05-19 11:06:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-28 13:26:16.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-04-03 21:48:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-02 08:27:53.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"9536f88537e81bf09d20f09e2c23c6895529ee47","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-27 17:31:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bbd6ffac5810804a70a2be4fe240a83c2110330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 19:35:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d07c68049ea4e099ddcfff1f6d0b0c5 : SUCCESS in 12m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e458630c053e444ab7d8ebc06142a2a3 : FAILURE in 41m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d1b43427533546c7a600a4b9bb29c1e1 : SUCCESS in 55m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3bb691c8e0764d988e3266c5dcd04c4c : SUCCESS in 13m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9d589441ff204ce3ad88528e0b8d1324 : SUCCESS in 14m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d0a04f619137418da9c8206aaf877411 : SUCCESS in 14m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e1e749bd648d4e25bae0e7736c6a4f18 : SUCCESS in 16m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ebacc41a8ccf4bf28bcd44acacb535e5 : SUCCESS in 16m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a64d81c5bcf84b96a63ae78be3021140 : SUCCESS in 20m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/12c60f733c6f436bba15b5e62ba797a0 : SUCCESS in 12m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/57920ba922d8479699be3ce84232d338 : SUCCESS in 24m 23s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/879ffd5ba18d4509843eeace70addb63 : SUCCESS in 20m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/282125266f624f17ba39c470e9b9b2c1 : SUCCESS in 19m 19s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c5a1462b3587449f979f667a12c73fe2 : SUCCESS in 26m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3893572e435a46eb8d28f41f927b9237 : SUCCESS in 22m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4d94989cc9594f1ba470477fffe7eb52 : SUCCESS in 20m 54s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7a209aa009b64995bec9d06737415dae : SUCCESS in 14m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/99daf74e2a324f2781bf03b31d4ad27a : SUCCESS in 15m 01s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f1e77e73f51e4ce584b199b4d1ce88fe : SUCCESS in 23m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0836b0d115d54d0f8084352afd7eb9d1 : SUCCESS in 23m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/98bddb29ee3d48e38161aad1b9a13757 : SUCCESS in 50m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/070b77f9c24749e8b08f28fe07f7f30b : SUCCESS in 32m 13s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5cef757acde74cecb37cee5a9d304226 : SUCCESS in 26m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f76835c65d6b4f2ebc9935ab5ca97483 : SUCCESS in 37m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7e5f294b4364c6fb8fc508f9b739cf8 : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e328b758ee0944908fcac472237cf0d0 : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/78e54f76e57c4f639296f85d7c219830 : SUCCESS in 25m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3d16dde6d4e4285844ec7ecf6e42e89 : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2bc0765d117e7a104cb3a8ebc12d4d6282d53492","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-03-28 13:26:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nthis seems better but it needs a test","accounts_in_message":[],"_revision_number":1},{"id":"c9221fddb9e3077dbfd65b81a5fe560cca88db39","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-03 19:27:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"83b70c0826da151da95b2b9c8426d9d7e4a4b0fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 21:48:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6ae9cc428cf4f14ad2a98fcad5474e0 : SUCCESS in 10m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e25ee56595134fe99879f0e427af25cb : SUCCESS in 1h 06m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/271db40cb21440dba7f1fd0e9f6127e8 : SUCCESS in 1h 00m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0867c52acdf14a98b576eaf2f0436289 : SUCCESS in 11m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/87c88d5d57ca487d8273061f034be0c2 : SUCCESS in 10m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1565513cc4df483ebe532202b7f53ebc : SUCCESS in 20m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35d6feb2cfba421c8eb6a292492d2820 : SUCCESS in 12m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d2dd8e65b5784a60b359e996b0321b40 : SUCCESS in 13m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/48f6b1456b874941b5729b8834bb0981 : SUCCESS in 18m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bf371e10953e42d797a998ebced17507 : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e1bbcec840d34660bef580537134ea52 : SUCCESS in 16m 07s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c8161b8b329d4cddaab3a18b3c846d90 : SUCCESS in 34m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d4035f8ef3d6478d896582871023a9ad : SUCCESS in 13m 20s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/4d1350f620f345f08aae6f04a033bb84 : SUCCESS in 16m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5ebf32e83e064649bb977702ede9e1b8 : SUCCESS in 10m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2bd7ca06660e4165a1da7930c583dbec : SUCCESS in 13m 01s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5cdcf3404b3b46baac6ae3d2afa16753 : SUCCESS in 11m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e53aadea0a784c66b9cebf44471ef9b6 : SUCCESS in 12m 29s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4ec04295662742dc96278db93c4d4dab : SUCCESS in 17m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/25ca27b7306d4ceba73fb58d5ce3172b : TIMED_OUT in 42m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/87534d183cc840bc8b9019e38b54ef57 : SUCCESS in 45m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b16fb7383bc14d6ca02765e46f134e83 : SUCCESS in 36m 00s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4f078fcf63e64a5ab8951774da598223 : SUCCESS in 45m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4557a01e09b6498f8f9b7e930f3c63c7 : SUCCESS in 31m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82db343ddbbb49f189879d2217906cb2 : SUCCESS in 15m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da71ab0aa37644a99fae21838b938ab1 : SUCCESS in 8m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c22723fd71f944d58fbe2f6b01e254b5 : SUCCESS in 26m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22662b99928d4cdda8110bb24f114f2b : SUCCESS in 56m 51s\n\nWarning:\n  Job swift-tox-func-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-py3/lib/python3.6/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-py3/lib/python3.6/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-domain-remap-staticweb-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-domain-remap-staticweb-py3/lib/python3.6/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-domain-remap-staticweb-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-domain-remap-staticweb-py3/lib/python3.6/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-encryption-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-encryption/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-encryption-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-encryption/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-ec-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-ec/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-ec-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-ec/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-encryption-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-encryption-py3/lib/python3.6/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-encryption-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-encryption-py3/lib/python3.6/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-ec-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-ec-py3/lib/python3.6/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-ec-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-ec-py3/lib/python3.6/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/netaddr/strategy/__init__.py in the commit\u0027\n  Job swift-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python2.7/site-packages/setuptools/dist.py in the commit\u0027\n  Job swift-tox-func-s3api-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-s3api/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-s3api-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-s3api/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-s3api-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-s3api-py3/lib/python3.6/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-s3api-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-s3api-py3/lib/python3.6/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-multinode-rolling-upgrade: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-multinode-rolling-upgrade: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-dsvm-functional-ipv6: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-dsvm-functional-ipv6: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-domain-remap-staticweb-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-domain-remap-staticweb/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-tox-func-domain-remap-staticweb-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-domain-remap-staticweb/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-dsvm-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-dsvm-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func/lib/python2.7/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-dsvm-functional-py3: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-py3/lib/python3.6/site-packages/os_testr/ostestr.py in the commit\u0027\n  Job swift-dsvm-functional-py3: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/func-py3/lib/python3.6/site-packages/os_testr/ostestr.py in the commit\u0027\n  Comments left for invalid file .tox/func/lib/python2.7/site-packages/os_testr/ostestr.py\n  Comments left for invalid file .tox/func-py3/lib/python3.6/site-packages/os_testr/ostestr.py\n  Comments left for invalid file .tox/func-s3api/lib/python2.7/site-packages/os_testr/ostestr.py\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/netaddr/strategy/__init__.py\n  Comments left for invalid file .tox/func-s3api-py3/lib/python3.6/site-packages/os_testr/ostestr.py\n  Comments left for invalid file .tox/func-ec/lib/python2.7/site-packages/os_testr/ostestr.py\n  Comments left for invalid file .tox/func-encryption/lib/python2.7/site-packages/os_testr/ostestr.py\n  Comments left for invalid file .tox/func-domain-remap-staticweb/lib/python2.7/site-packages/os_testr/ostestr.py\n  Comments left for invalid file .tox/func-domain-remap-staticweb-py3/lib/python3.6/site-packages/os_testr/ostestr.py\n  Comments left for invalid file .tox/func-ec-py3/lib/python3.6/site-packages/os_testr/ostestr.py\n  Comments left for invalid file .tox/lower-constraints/lib/python2.7/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/func-encryption-py3/lib/python3.6/site-packages/os_testr/ostestr.py","accounts_in_message":[],"_revision_number":2},{"id":"e18c6b2626c74f9ed39c69d1cb88bd92e64cbf3e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-06-02 08:27:53.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)\n\nlooks fine, I\u0027m just a little concerned about the potential length of the log line.","accounts_in_message":[],"_revision_number":2},{"id":"e7789217546653d0ab22c0a63121f32af565c526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-08 18:28:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4fe3c52726a28e395c8a69749200db358c09a037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 19:37:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7310c87de4bc483b9c0e41a9e3098670 : SUCCESS in 14m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/6f44366e90af4039973bb96f654ea8d3 : SUCCESS in 1h 05m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4f59ae7cc6c9451dac6d4f57ae662e40 : SUCCESS in 45m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d9610ddd1828487bbdf1729c9f1259c3 : SUCCESS in 11m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c5cbf59d6ac74f089b1794f70dbf97df : SUCCESS in 12m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ce47e9ed49f04d9b936f108ef103d5b9 : SUCCESS in 18m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b240518d3b604d1e97bacd1aab254b62 : SUCCESS in 18m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/799e0410d9eb4ec1bbd896db3951e93e : SUCCESS in 19m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2434f71ea3f24e34ae7fd20c0b0eee2e : SUCCESS in 18m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6941c04f751442c2a1332be8471e1bb6 : SUCCESS in 18m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b45fd8efce0441acb90732361c3fdc25 : SUCCESS in 19m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a65c46a749be41e992dd42c2c2953358 : SUCCESS in 52m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2e458d8f4313446f88ffcc9db2d2fbb4 : SUCCESS in 20m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2cb414c1d4b54a9fa21de4013746c179 : SUCCESS in 21m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/29e40403819343289948207d42c42c1a : SUCCESS in 52m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/91234ea4baa3415e9f125ea0cabaa903 : SUCCESS in 19m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d3eb8b46a9ec4d8d8cd8e040ef4e0179 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/01b957121d994539bfc1707e1e17d244 : SUCCESS in 22m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/61bf1c3067c7421280359b722af9ad65 : SUCCESS in 49m 30s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/847788d9797846ac84e22d22ce4fbd14 : SUCCESS in 50m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1180cf46a1274f51824e27dd0b0f321b : SUCCESS in 40m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c489cb90c7a418ab89dcb4aeecdbcd7 : SUCCESS in 39m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/232478ceb80a479487d6db8ac9e9dd80 : SUCCESS in 19m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d88f5092c60e4f448678af2569ca760e : SUCCESS in 8m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68b6fa2d683146e4a2325a1dc40566d3 : SUCCESS in 25m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8272a5cd7a424d818499cd68546a89ed : SUCCESS in 59m 49s","accounts_in_message":[],"_revision_number":3},{"id":"41827083ab3927f715ea97b244c00dcf993af460","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-06-14 18:54:04.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)\n\nthis change is pretty good - but it\u0027d be better if the cap_length refactor was pulled out and the json error handling was more targeted","accounts_in_message":[],"_revision_number":3},{"id":"ccba634627fbe77651c3284e01f4dfdf5ac5b45d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-25 18:58:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3bf2e0fb8db65b2491725616f88173a38ea2efb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 20:26:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e579a30b83f422d8be873639f14e274 : SUCCESS in 11m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/5c40fd71fe4c48fe9e61db7da62b15f0 : SUCCESS in 57m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d669c8152864c99be52c73e9c00f9f5 : SUCCESS in 53m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/860346b038fa48978d9ffb5c1011a261 : SUCCESS in 54m 21s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8061b1d680094ea6916b98f3a7e8bf44 : SUCCESS in 23m 21s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e9864e7a52fd47868783c01ae5be5556 : SUCCESS in 24m 35s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/706a49e412c14b5b91ae0e233a6660e9 : SUCCESS in 24m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3b230807e344e6db50b299eb6fff5a9 : SUCCESS in 11m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1be719610bc54e699b31b8f8d7fce3ac : SUCCESS in 7m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b17fc665e3af485c935244426894f8d2 : SUCCESS in 18m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a22120f2f41422e82e1caef508e0598 : SUCCESS in 18m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e958edc65c014a4e967cc5d1103fea23 : SUCCESS in 23m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/43a9f0de570f451bb45fe4ddde0e08f7 : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e6fd56aaad6e4ed4b7d143374d2c7f9d : SUCCESS in 20m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd72506f90e44b358f05b267c11fe9b9 : SUCCESS in 18m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/336703c0320f4d28a623c7ac571a3b56 : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5b847ca7b8304f6397e8f314a3117474 : SUCCESS in 19m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3c18ccb2e92b42178b66c18a05fd8c19 : SUCCESS in 19m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/23dc42e573bf486f965e6daffce5981a : SUCCESS in 8m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bf23b8b24c524279ab4abee329219c02 : SUCCESS in 22m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/976613d8f6164877ae2634e49ee5297c : SUCCESS in 1h 00m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9db02d50feee4396aec3b2bdeb8cb765 : SUCCESS in 55m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5682db80633a4e559e88c675c42d0d38 : SUCCESS in 46m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a87950ade33b4e9aa215db14d88e9ead : SUCCESS in 45m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d75effdc53b04d5e8e266b703a59e9bd : SUCCESS in 19m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b6b1ab9a3be400db55006bc7c50c18c : SUCCESS in 8m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ff9362112c4488db40ea7227674537e : SUCCESS in 26m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1a31b2dd4e344e9bca0cdd9cf6901fa : SUCCESS in 47m 40s","accounts_in_message":[],"_revision_number":4},{"id":"7b4caa388df360d6e41c9cd397a52d0794f04c8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 22:42:18.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f394f11f7bfd4fb9a560db714c71148e : SUCCESS in 55m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9521c423adb14fd6a43a96d5391bdb01 : SUCCESS in 1h 12m 26s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/573fefebf823499b91ed0f1c6da13a09 : TIMED_OUT in 3h 02m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8b9cefa53854229a1f3a884561e4255 : SUCCESS in 1h 28m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/162d059a0edf4ce7af7649ffb98a3810 : FAILURE in 1h 13m 56s","accounts_in_message":[],"_revision_number":4},{"id":"1c5151ff7701ceef8dec4bfea33f9dedff5e4163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-24 17:32:16.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"62fa28b39a4f406ec0b30ad2d4ac11d24007dce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 20:05:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81fba7ada661483ebfef227b56b63dfa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3ac14eef264c52ad73bcbc2aabae73 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/e1135f41d49c4e7dbfd84fdfeed1f5a2 : SUCCESS in 42m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e30f7a013b364ff3919e25284a44a82c : SUCCESS in 58m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83f8105f8d26491c95e10e7633e43576 : SUCCESS in 59m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54c61b6b36ec4b09a25107f87906fc57 : SUCCESS in 44m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67172fab47e6497e9ade5542205f2496 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cafce4d465aa46fab25085d82114b2f0 : SUCCESS in 18m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf19ae427ebf4f1792a9ec4acec22367 : SUCCESS in 8m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6e4c7f64dde14486bdfe813e03f34f06 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/34c89c4450f043648ffb2259555b62ba : SUCCESS in 16m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d3a4ee9ec0c949779679a12d771a7465 : SUCCESS in 17m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ce30e478973a4301af6a4e5a37701411 : SUCCESS in 13m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/83b11143ec094449952e1c3bdff40a80 : SUCCESS in 13m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ffd70160c9064aecb2bc0be2dbb2c863 : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1afe93ce427640a698a3a0756801d5ad : SUCCESS in 21m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da8967f2d2c8459eb7a14e5e30834d72 : SUCCESS in 18m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc7d961fc8bf4692ac2496d8b78fe954 : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/331089a1ba3340688e2fdff399f62ec0 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6928f4e2c8c24d7299e547b3ab107ff5 : SUCCESS in 14m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0800417f3de04af5be59681b210da325 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0b75ed0611bf49ca9a537503a811d99b : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/18671dadec2c4336a842087aaf3ea60e : SUCCESS in 8m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dc39fe7f1cf544b3931694800dafaeee : SUCCESS in 50m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a848a7f8006c419599db315dc3a2719a : SUCCESS in 57m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8a6ef54d57842458221d4fe39a07735 : SUCCESS in 46m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c73fa4a11ea49e0a22867d13efaaa47 : SUCCESS in 57m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40d9a7276c7645d8ac68cae70c02b467 : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/723954938e0e42069a827aff5c5a58a7 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7821cdfa3bc94d1c88e4a1d925e18813 : FAILURE in 24m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abe2ec471b8840bd99043deba9e03477 : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":5},{"id":"ff155d0a8d54bb95f7ec043e49bcb5fae8e81133","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 22:14:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e013a1248545b7b2fa286778437ae1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/347352a913804067907017005c88d971 : SUCCESS in 1h 01m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a6c64eae218644d2b81c938740f36307 : SUCCESS in 54m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/89329d5017d24e5fa135090ab5de0ceb : SUCCESS in 58m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9df7cf59b0d488b91443602f533e0c6 : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":5},{"id":"b8d61daccbb3cf4c601ad057b44fa6380869283e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-06 03:54:59.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"6b673bc31c0fc01332ca6652b319027a43b52fb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 04:44:05.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f24f652d7a1649a69ee1cbeb21612a58\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/181757f61ef446998d27586119c9917b : SUCCESS in 17m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdee950248684064abcb933ea4f3ec43 : NODE_FAILURE Node request 300-0020917176 failed in 0s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b42ad8d3ae374f2e901386edbd2cd471 : SUCCESS in 25m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fac45ff12c494043a96e77e12f90fc12 : SUCCESS in 46m 19s","accounts_in_message":[],"_revision_number":6},{"id":"3061351faa2be638d70ead168d9a7dec8f86a89d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 05:07:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/548c5de185d04950bd549df555761e16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b744f67ca3452b81ba236aa72ba44d : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/ecc3a91cfab04bd38142ea295ef69079 : SUCCESS in 1h 02m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d90f8a2ebe24344bef2cc53b94bd268 : SUCCESS in 45m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e13d641409a4db6af389d841acf70d4 : SUCCESS in 1h 04m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37b3e081c4504ba384e537e93a9845f0 : RETRY_LIMIT in 3m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8ef179a9b924c1aa6ca4c0ebfc277a5 : RETRY_LIMIT in 4m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26b8fa225b5e47cd9e5adb7566197c59 : RETRY_LIMIT in 4m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f08a6a6a0d7f4e99955efbba4288fa63 : SUCCESS in 8m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bfc96a5235e14cd6a9bff404ada0b941 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f425f6e8c57a487c911b5fd3195f8a38 : SUCCESS in 25m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/93180292f6574e2db6688e3bb2516e59 : SUCCESS in 16m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1404580831534b3f9ae37485950f3343 : SUCCESS in 14m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8eaadfb37fa44882be2282f5ebadd92d : SUCCESS in 17m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9b374bc322d74a1ab7403f0b60328ff0 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/97650d1e83e74176bf747bb2f2dd617b : SUCCESS in 17m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da0dafb45f4d4fe28fd93478ae45d5c0 : SUCCESS in 17m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0e0c33dda1bd4a59b6a8237341e205e6 : SUCCESS in 13m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2ff6ef98471b46f7bfaa0a4ff1e5b7d3 : SUCCESS in 13m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e8aed12d88da4c7795a777750916e06e : SUCCESS in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4154c642bb9e4461ad946a51d6f720c5 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c6503bdd77c4aa8a00819c81d0e1f09 : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a2cbacc15f284ae393f0328018f95e8c : SUCCESS in 6m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/96bc102da2514e55afe272e4f28e0db6 : SUCCESS in 44m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8c1c816d86a741cfbcaab29e84d59269 : SUCCESS in 1h 02m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/099daa79797b4601bc79b17c1febcf21 : SUCCESS in 52m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c393cb82f40b42a49db3242493c7fb01 : SUCCESS in 45m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1913d79e910541629c2803ce0c517b8a : SUCCESS in 16m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/383206ee6e6a4e8aa220a6033f7f51c0 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2807aa5d25404f2eb05fc2f734425410 : SUCCESS in 18m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bc5c1a396fd4f3fbc59ff2c8fdad4d5 : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":6},{"id":"2db9d71e80969dfa7292054c3b2449de54245360","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-19 11:06:01.000000000","message":"Patch Set 6: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"46c97da6ec7100aa6efe6195f6c3a08aab7a46b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-19 11:06:17.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"6ba9b315458ae16d3012e00e44ea3b80facf38c4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-19 11:06:56.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5c0eb2b56ad0a9e028fecd4db9b4939e8f2778a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 12:11:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d9c187859eb4175a7745d629fd0a096\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c70d2d07070459ea9f17aa5418c2a76 : SUCCESS in 7m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/1ea67a9d5edb4781a32744f84207e12d : SUCCESS in 43m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/280e1ec0652b43dfb84b90d7eddaa614 : SUCCESS in 50m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c230134d9dc498e85fb13300b010a8a : SUCCESS in 56m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2907a0c2bf784eda901760bcd75e3172 : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d694f3901f3948299b73df9dfa9ac2df : SUCCESS in 16m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9cd9447579a440781e66c43d7f449ac : SUCCESS in 17m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f9a84ebb4b2e4d7ca53c5f0f1ac13f5d : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f58851544b2a42edb90f7778bff46abb : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/191cb87f62ab423998d7d4e8cd67f826 : SUCCESS in 17m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e9d63213946488d95a5a029c5e48f94 : SUCCESS in 16m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/724b1263e7944bb8aedbfec7ff4dd3e0 : SUCCESS in 14m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f07cc33dc86e4932a56489c41e23d4db : SUCCESS in 16m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a04f673c03304995975d5467bd95c09e : SUCCESS in 21m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fdee107ac6134d508452c6a243de8dfd : SUCCESS in 21m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/10507a48f4a44df3ad703a7db2230251 : SUCCESS in 17m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7aef34f8ff92412eb494671321494b37 : SUCCESS in 20m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/95669360b8b449b18ae45a4dc203a507 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/448e3eb8fa934c89b2a0b82aa4a70137 : SUCCESS in 14m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0e0b3e26fb4a43f9b443fbeae379a55d : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3125e2a9b96544bbb9af248022404dad : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/65e0f0b6d93c4c878c1771d8fd72a0d4 : SUCCESS in 6m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fbd4b30c35444a70a3f5dc97c24dda0b : SUCCESS in 52m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1a86b9d6e914d14a328770e71d166d1 : SUCCESS in 1h 04m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8beab0aaa88f4a60b915296deef9a8d2 : SUCCESS in 56m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/653e3a8e46fb4e7f95c34b583e092a14 : SUCCESS in 48m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/182d2bb0e125437cb5ade89a3e83a80f : SUCCESS in 21m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3b95c619c743a883962d64c8bf02b9 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8a33bf7f8e4941a987af79037db90c8a : SUCCESS in 23m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/675eda70c42b473c8a65d932045517d0 : SUCCESS in 52m 57s","accounts_in_message":[],"_revision_number":7},{"id":"2710dab60267cfbce69d07fd3b178923cf210873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 12:12:05.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"4f1471f24fe4ae2112e04a55a8ce94d371926d5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 12:13:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/859722f6edd542f5b03c5655bf3cc91e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5479efc085742dbb5d1dd2efb0b2949 : SUCCESS in 17m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae905b53a0394193995af6e268700762 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6b812f0c2c04d4c8571f3581d46d0ff : SUCCESS in 19m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ddf0d5017a8743eaa5be393d9f9fb543 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":7},{"id":"7dfc2d77aba83aa4ec9c8fc9aa09d633c9535540","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 13:26:26.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31ff398825f24d64ac540fded9d5495f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52ae3e9713574e97a52b5081e3a091e5 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/46815d661aa74d3990d3200b5521df90 : SUCCESS in 57m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/67c59fca8ec945448b3157081d2c156e : SUCCESS in 35m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f8df0fd556b44b698afbaa292380e5a : SUCCESS in 51m 06s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bb0daa171a57461c8653b0ae0ab63b02 : SUCCESS in 9m 05s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/39619aaa9b0c48f98d74dea968087cbf : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6a63c8f0d3944bcbb7b6c34fb61be3ee : SUCCESS in 17m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2fd84bb5a4a4ad696dd89732672ba0b : SUCCESS in 15m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e381dc0ca3b64c41b3c27f96df37107d : SUCCESS in 12m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f8ef52fd1adf42f4873130f43faaf3f6 : SUCCESS in 13m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7729390958b646978ed80be6c8993062 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7f87a4d2167c460ebcdace93b7a59f63 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/56ad6033528643babe2d45e8fdfdd467 : SUCCESS in 17m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aabc871d61ac462cb6c0ed0c83cc9405 : SUCCESS in 12m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c24cfa6a2f4c4c6c9606de2754cf7a2f : SUCCESS in 17m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/30ca4c3f85e34bf0a96308c80f8c7d04 : SUCCESS in 13m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5fa9af1ef69e4c929fa28535e1406b5e : SUCCESS in 6m 12s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e260a7e7fcd54de6beeacd1aa9569023 : SUCCESS in 6m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/023a2f76a4ab40419066810322c9a891 : SUCCESS in 1h 00m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ff9095b7ccd4eb28ac24b379a586db1 : SUCCESS in 1h 03m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/34be32c62a9647be8ff2cd490e850de4 : SUCCESS in 46m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f7e29cfcb2944bf289cdb30e42c30add : SUCCESS in 42m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/410a06dd6c484e2b9582c1c0e1b5b786 : SUCCESS in 21m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98208b34e80d4184a9e2c120188f70d2 : SUCCESS in 6m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73b7783c1fdc4d67b0ab9bfc9188e3db : SUCCESS in 41m 40s","accounts_in_message":[],"_revision_number":7},{"id":"7f219f705a7307f849b5fd72d11ab5b59e370000","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 13:26:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"d9e4f35e262a5ef88a3c85bf10574826b4304693","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 13:27:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2e04d26fc0c468da6c9aac977a5014e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2ec5555fb64778b03cf8b443e921d8 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2b58315f93e24cac8eb4e863957a6af2 : SUCCESS in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/a5838543fc07492980145f7b9a3b8710 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2deab27dadb2f765a9abac43d5ecc6ad1851c6de","revisions":{"cbd39b377cf3ae0da194b95418132aff4ee589ca":{"kind":"REWORK","_number":1,"created":"2020-03-27 17:31:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/715492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/715492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/715492/1"}}},"commit":{"parents":[{"commit":"39c81597021c5e21ae45e78512f63483ea914846","subject":"Merge \"Recommend better rsync.conf settings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39c81597021c5e21ae45e78512f63483ea914846"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-03-27 17:30:22.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-03-27 17:31:04.000000000","tz":-420},"subject":"s3api: Better logging for non-JSON when trying to do bucket listings","message":"s3api: Better logging for non-JSON when trying to do bucket listings\n\nChange-Id: Ib79bfd23b7b6f40673396aafa47a0f8c4f533a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbd39b377cf3ae0da194b95418132aff4ee589ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbd39b377cf3ae0da194b95418132aff4ee589ca"}]},"branch":"refs/heads/master"},"5b57dc3f6015f5393b248a1c34d560c554d90544":{"kind":"REWORK","_number":2,"created":"2020-04-03 19:27:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/715492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/715492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/715492/2"}}},"commit":{"parents":[{"commit":"39c81597021c5e21ae45e78512f63483ea914846","subject":"Merge \"Recommend better rsync.conf settings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39c81597021c5e21ae45e78512f63483ea914846"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-03-27 17:30:22.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-03 19:27:32.000000000","tz":-420},"subject":"s3api: Better logging for non-JSON when trying to do bucket listings","message":"s3api: Better logging for non-JSON when trying to do bucket listings\n\nChange-Id: Ib79bfd23b7b6f40673396aafa47a0f8c4f533a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b57dc3f6015f5393b248a1c34d560c554d90544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b57dc3f6015f5393b248a1c34d560c554d90544"}]},"branch":"refs/heads/master"},"42c6eeec8cb3b7b8922b2d7797c50c0447e86320":{"kind":"REWORK","_number":3,"created":"2021-06-08 18:28:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/715492/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/715492/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/715492/3"}}},"commit":{"parents":[{"commit":"5bc28a7febe8c45df6f489d3de12870fb1b81f58","subject":"Merge \"sharder: Send broker path as referer when updating root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bc28a7febe8c45df6f489d3de12870fb1b81f58"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-03-27 17:30:22.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-08 18:26:59.000000000","tz":-420},"subject":"s3api: Better logging for non-JSON when trying to do bucket listings","message":"s3api: Better logging for non-JSON when trying to do bucket listings\n\nDrive-by: Introduce a string-length truncator.\n\nChange-Id: Ib79bfd23b7b6f40673396aafa47a0f8c4f533a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42c6eeec8cb3b7b8922b2d7797c50c0447e86320"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42c6eeec8cb3b7b8922b2d7797c50c0447e86320"}]},"branch":"refs/heads/master"},"c42f3d9b4ca8c7350626b347fa28e345ff7314f1":{"kind":"REWORK","_number":4,"created":"2022-02-25 18:58:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/715492/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/715492/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/715492/4"}}},"commit":{"parents":[{"commit":"689dab9a7f82640a8cd151f1b740c6e0cb45a2f4","subject":"Add cap_length helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/689dab9a7f82640a8cd151f1b740c6e0cb45a2f4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-03-27 17:30:22.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-25 18:56:10.000000000","tz":-480},"subject":"s3api: Better logging for non-JSON when trying to do bucket listings","message":"s3api: Better logging for non-JSON when trying to do bucket listings\n\nChange-Id: Ib79bfd23b7b6f40673396aafa47a0f8c4f533a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c42f3d9b4ca8c7350626b347fa28e345ff7314f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c42f3d9b4ca8c7350626b347fa28e345ff7314f1"}]},"branch":"refs/heads/master"},"d47018da76778eb2f6c7790352b9106b61712b1a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-01-24 17:32:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/715492/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/715492/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/715492/5"}}},"commit":{"parents":[{"commit":"32081ad100de2610cd69fb5238108fa87b056f7d","subject":"Add cap_length helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32081ad100de2610cd69fb5238108fa87b056f7d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-03-27 17:30:22.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-24 17:31:50.000000000","tz":-480},"subject":"s3api: Better logging for non-JSON when trying to do bucket listings","message":"s3api: Better logging for non-JSON when trying to do bucket listings\n\nChange-Id: Ib79bfd23b7b6f40673396aafa47a0f8c4f533a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d47018da76778eb2f6c7790352b9106b61712b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d47018da76778eb2f6c7790352b9106b61712b1a"}]},"branch":"refs/heads/master"},"6bedde1cc07cccab1a1e9c0018aa8b20731a00c6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-04-06 03:54:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/715492/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/715492/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/715492/6"}}},"commit":{"parents":[{"commit":"0a4e41701dbb5795ff4cab7a2c68a41c90bd51e7","subject":"Add cap_length helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a4e41701dbb5795ff4cab7a2c68a41c90bd51e7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-03-27 17:30:22.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-06 03:54:39.000000000","tz":-420},"subject":"s3api: Better logging for non-JSON when trying to do bucket listings","message":"s3api: Better logging for non-JSON when trying to do bucket listings\n\nChange-Id: Ib79bfd23b7b6f40673396aafa47a0f8c4f533a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bedde1cc07cccab1a1e9c0018aa8b20731a00c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bedde1cc07cccab1a1e9c0018aa8b20731a00c6"}]},"branch":"refs/heads/master"},"2deab27dadb2f765a9abac43d5ecc6ad1851c6de":{"kind":"REWORK","_number":7,"created":"2023-05-19 11:06:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/92/715492/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/715492/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/715492/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/715492/7"}}},"commit":{"parents":[{"commit":"0a4e41701dbb5795ff4cab7a2c68a41c90bd51e7","subject":"Add cap_length helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a4e41701dbb5795ff4cab7a2c68a41c90bd51e7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-03-27 17:30:22.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-19 11:06:08.000000000","tz":60},"subject":"s3api: Better logging for non-JSON when trying to do bucket listings","message":"s3api: Better logging for non-JSON when trying to do bucket listings\n\nChange-Id: Ib79bfd23b7b6f40673396aafa47a0f8c4f533a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2deab27dadb2f765a9abac43d5ecc6ad1851c6de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2deab27dadb2f765a9abac43d5ecc6ad1851c6de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
