)]}'
{"id":"openstack%2Fswift~715576","triplet_id":"openstack%2Fswift~master~Ice224fc2a6ba0150be180955037c13fc90365479","project":"openstack/swift","branch":"master","topic":"systemd_support","hashtags":[],"change_id":"Ice224fc2a6ba0150be180955037c13fc90365479","subject":"Make wsgi server uses systemd\u0027s NOTIFY_SOCKET","status":"MERGED","created":"2020-03-27 23:58:00.000000000","updated":"2020-04-02 01:42:29.000000000","submitted":"2020-04-02 01:40:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715576-1585791643202-6894be15","meta_rev_id":"c94d11c1ccbdbeeddbdf6bb6d3f6fcad9a9a0f2b","_number":715576,"virtual_id_number":715576,"owner":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"date":"2020-04-01 22:55:45.000000000","_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 01:40:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2020-03-31 19:46:04.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-01 15:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-01 15:50:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-28 13:22:14.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-03-31 19:46:04.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-04-01 15:50:37.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-04-02 01:40:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d02f96c3898a4674f5e43967f8ee1e2d5c96f4ef","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-27 23:58:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6e3f2f63e5a7beeb4a6fc8e873ccea078c52205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 01:06:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e28a755d5d746cd83fd1b1ee3e7b743 : SUCCESS in 8m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/dd88abe22cd746b98e661888d5e643d7 : SUCCESS in 1h 01m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0902d6fedf9c42c0bde078cb448b2f25 : SUCCESS in 48m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de8c14bb0f604883bc57043e81e8f4be : SUCCESS in 12m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dbfe1b17a0e14eee8f33208e4b41b323 : SUCCESS in 11m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/758c92a4c37241188ff876961d031702 : SUCCESS in 21m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7bf33fbe5804446397abe6b7ccb648ae : SUCCESS in 23m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5e96e60c9d554e1db70efa60d445eaba : SUCCESS in 12m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7eedca37257046d3a4c36d5f1db59d5f : SUCCESS in 12m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d47b5e55a07c4cec862888ff7d0d12e4 : SUCCESS in 9m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2c2f3737687e4ac8bc96ce58711f29c6 : SUCCESS in 16m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/bd8ce2cc458b4e4fa7bdd221fe8e7382 : SUCCESS in 13m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e6ad95bbe48042bb9d8ce55fd0a4e158 : SUCCESS in 12m 01s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3f5ad0a749304e60b33d5287de2e7f2d : SUCCESS in 15m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3204d49edf7c48d48cccbf3d34d3ef37 : SUCCESS in 11m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/894c427214e34960a8791d56624281ae : SUCCESS in 21m 18s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e6e68e52d669430a8dd5eda74dcf9b9f : SUCCESS in 12m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/871c2a6cce5a46a9990f0234b738a6c7 : SUCCESS in 10m 36s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/584cbf59af2740c9a15056ff25b9cf01 : SUCCESS in 26m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7bf383c87c7244639ba39424a0aee622 : SUCCESS in 29m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8eb4c09d18b94bbaa7ba53258ce47586 : SUCCESS in 53m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8e1a8ddf94574e2a9889b560b98f1a3c : SUCCESS in 35m 52s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/bddc4fd6cfb34ddcb61e5889f148d75c : SUCCESS in 35m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1afd94248dcf47cda2c4319b65a45a69 : SUCCESS in 28m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd8e5a006190408a8e0949b08efb714e : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddaa3e89ae5e4cbb8cff253e9dbda042 : SUCCESS in 7m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87812ae3e0c048729e4eb6952c892ddf : SUCCESS in 33m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0bdee1ddd5446f5852f2f377792b576 : SUCCESS in 1h 03m 17s","accounts_in_message":[],"_revision_number":1},{"id":"16b60ec6d1ebdf8b0318e0cb83a85ef814546b22","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-03-28 05:25:14.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f8c024603c9f4c3d458a89bba1f7720fba98b3db","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-03-28 13:22:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nPlease provide more detail (links?) on what this is supposed to do and how to functionally test this.  May benefit new docs as well.","accounts_in_message":[],"_revision_number":1},{"id":"c63c583259ff04951725eef7e2d6ba524861a592","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-28 22:22:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6c4e957a827e2de7cca0ff9e675f205ad1433d58","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-28 22:25:41.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nPlease note that I mostly copied code from oslo.service:\nhttps://github.com/openstack/oslo.service/blob/master/oslo_service/systemd.py#L36-L47","accounts_in_message":[],"_revision_number":1},{"id":"298b783a46993d0cfa26e7d889bc44e1a90f7253","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-28 22:32:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"de45096df0c9d2c6997687b9db7bd1d010d562ec","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-28 22:32:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e70715926085a28d15bde669ef3d620e68858c17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 23:33:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6acb035f80124a78a284d1496fb5791f : SUCCESS in 7m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f3f5b6e9d2a14951827964bca3d0a1ee : SUCCESS in 59m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5eae5405bd0c49bdb0518f22066f854e : SUCCESS in 51m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de69e390901c419aa5860ddee1c16220 : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1ee100cc9ee74f7198d94db3358adf54 : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f57e9bdc7d2b43818d49d7850befbfac : SUCCESS in 14m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d9db40069fdd44a9b60828ae2dc07e0b : SUCCESS in 11m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3b31fb4cdb884965b03e317b0b6b60a2 : SUCCESS in 11m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c37526c1c83f4a3daa40888a231cbdb4 : SUCCESS in 14m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3189028ecb6d4734aaf6b7d45f971bc5 : FAILURE in 4m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f03d3bb0fb184b80b1020772680b3f5e : FAILURE in 3m 22s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b949212deea843018c7b210b2b1cfc26 : FAILURE in 3m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f1a254f152104a19bce1e5840e70f9c2 : FAILURE in 3m 37s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/2373d2255884419abcb471d18229bef6 : FAILURE in 3m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dbab10a3236543f8887e3e2fae1f666d : SUCCESS in 9m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f6bd98345cbd47c38fd0c278ef8afad1 : SUCCESS in 12m 12s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/44ddbcf7c511447780d9f11bb8514f42 : SUCCESS in 9m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/44fddbf85e9f4e4a809d0f0f9161af54 : SUCCESS in 10m 17s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f06eaf4e65e44665b841fc59526df035 : SUCCESS in 14m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df21a587b3a0459e8ca392b843e9e82a : SUCCESS in 28m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c9806f02b4134a14aab3df76bbb391fa : SUCCESS in 42m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39a6ba2812724b3e8e14b8befe6bc857 : FAILURE in 17m 53s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5df02869f9e34a27b73ece599e7bf0ef : SUCCESS in 26m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34347585f3a54edd8c5d7d9e18dbe2b2 : FAILURE in 15m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b8342e2799d4337b68fcfbc72728e80 : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ae15884b80400e8e69e6de4b4d2582 : SUCCESS in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3264ba0e34143f8b535ffd2da79c9e1 : FAILURE in 8m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba890524fd064368bbe6fa9da0719c90 : SUCCESS in 52m 51s","accounts_in_message":[],"_revision_number":3},{"id":"ce9da12575fcf663acac6d4bb3b2308840fdd87a","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-03-29 05:49:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bc62385081929d10b09de47f09c4e17255574e9d","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-29 14:38:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c1340bed9e29845fb2e7f49d91be5528a1a4a425","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 15:40:36.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c40f60c2614ff0a5cc8dc96992d6b4 : SUCCESS in 10m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/907897735b5f4b5ab568036768f3d675 : FAILURE in 1h 00m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ec56aa26bae74c71ac0065e141bb2d1c : SUCCESS in 49m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e35504601c22455a813415838b34075c : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b152edaeb42e48fd91e6fa38fb216c08 : SUCCESS in 11m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/738364b469ac480fae07a54a49d3c031 : SUCCESS in 12m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/756b52a6b9c3446b94a62c2d72381223 : SUCCESS in 15m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/25e951c6ff874f31865f30fc011fd5a0 : SUCCESS in 14m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/354460f452a5472baba712709c12da69 : SUCCESS in 13m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/52933c071f484242bc106e198920ad38 : FAILURE in 4m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cf760684781a437083287f60e88c6f01 : FAILURE in 4m 00s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b226a7b4fd0747399f7b15c6a7eacd36 : FAILURE in 4m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a823adac1bd4683ab037067fc97fdde : FAILURE in 4m 46s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/465588df507c4394982a69c87b35a282 : FAILURE in 3m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/192719ecaef04f77a4417c7bf5572172 : SUCCESS in 11m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c577d103b25e420f854f772bc619efcd : SUCCESS in 17m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a735d692a0d24094b1cb4380099488f9 : SUCCESS in 11m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/354972b33b4a408fa8957868a89b9fa7 : SUCCESS in 11m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b576281307dc4c73af1a914955fd750b : SUCCESS in 23m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0a64492771e04da28c208aca5e1e0745 : SUCCESS in 25m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/97e0d028432a4c22afc487daeedd05c2 : SUCCESS in 48m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f3958377d4e4b178ad69555e867c808 : FAILURE in 17m 47s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8b5f115fa5cb4c5c9ed6fb60bfe49ccd : SUCCESS in 26m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/509a322a717b4373986f9b314f6804e8 : FAILURE in 18m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d6506ed5c224ba781b79b1c4f6dadbe : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbf18600caad400497599f615fbfa16f : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa3af48d6ff54426b741a735316ed07b : FAILURE in 8m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd2d0571603e4087af109da8c6e22861 : SUCCESS in 1h 01m 26s","accounts_in_message":[],"_revision_number":3},{"id":"125a8f3d5efbe3aa9315e3e9a8528873985527f7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-30 20:46:57.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nFancy! One quick question on ordering, but I genuinely don\u0027t know when is best. I\u0027m mainly asking because I remembered the comment.","accounts_in_message":[],"_revision_number":3},{"id":"62cc10e037894ffd194ef47f53a9a887504868b5","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-31 19:22:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5147e232defdb2879c190318d9899c0f2816845f","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-03-31 19:23:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca2b6842b5df62d8ef71b1272203e2dd873be2ea","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-03-31 19:46:04.000000000","message":"Patch Set 4: Code-Review+2\n\ntbh I don\u0027t see a difference, but sure, this is fine.","accounts_in_message":[],"_revision_number":4},{"id":"6d4d5f95d6c9e1ad992cfd4764aa8ecb28ac6959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 20:34:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a71d13c29d84c85875fa202cf3fc145 : SUCCESS in 10m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2ed0b53b611e4f519a65d4b027141039 : SUCCESS in 1h 08m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6fb75a24a8654bfdb3ec834a5abfe51d : SUCCESS in 53m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b945fac2638418190096a0ee686d22e : SUCCESS in 11m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2f83c2a9a3d74f23a55646068dba35a6 : SUCCESS in 10m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3cff4c2fa2724ab5865b4f8edab98b97 : SUCCESS in 16m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8bfc191f10141668993139651769262 : SUCCESS in 17m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1ad29e22f0864738a9071ec91d4c43cb : SUCCESS in 16m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e4428e5f26174921996a98dc6eeebbe1 : SUCCESS in 16m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8a767ca52ac442ed85135db4ef0ee427 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a2f18fca299e42ea8ab2c3e9e230e7f1 : SUCCESS in 15m 07s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c4e42c9e4ba9473d8550e5c84b67c857 : SUCCESS in 10m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c30be8e6df7a48f6b50b4a19573c3e41 : SUCCESS in 12m 12s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e4c917b5a72848f8a2348d66fc16c34d : SUCCESS in 21m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b2ac3044445945ba8099dac6c25296d0 : SUCCESS in 11m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/edb117a252de4ab3bcb5a1dbbce819c2 : SUCCESS in 18m 27s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b64aa08d28944ae4ac8097bd0bbf0b95 : SUCCESS in 13m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a619468b39924a34a78a3fd6b6c69f86 : SUCCESS in 11m 30s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/41f2625549e24793bafaed3ae3574ded : SUCCESS in 20m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/91a21ea4b69d4a238e528795e0f59a96 : SUCCESS in 25m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c6576e96bbbf4f8fbd0b5ebf46b097fe : SUCCESS in 54m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eb3955c6f043446f8e58b7804b990f8d : SUCCESS in 31m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/66080e12cb5944a384749fe3eadd6f63 : SUCCESS in 31m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/00fd283fef584eaab9e28bd1d6e8b90d : SUCCESS in 33m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43f4ab790bfa41a0946de1249f076641 : SUCCESS in 15m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af67fade8f7487aa8aced1cd06322a6 : SUCCESS in 6m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b6861b6a2129466a8664afdb5a9b1db5 : SUCCESS in 21m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/073800ebb4c8441a91ea14c1e617e3ae : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":4},{"id":"4f7a17c5d70a4d1b5991afe181354cafbc698576","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-01 15:50:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)\n\nI\u0027m starting to think that we really ought to maintain in-tree systemd service files. We should compare notes on our service files at the virtual PTG, if not sooner ;-)","accounts_in_message":[],"_revision_number":4},{"id":"44ba94dc1c2a4feb64f34b5c4e400d78e207aaa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 15:53:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a9a1f638ee2ec6e6dcc3068befefd334c05060cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 18:06:59.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04ea8f8ea8814e0db82f2459637537d2 : SUCCESS in 14m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8a7207ea64ee4198bddefee3d8714cd4 : SUCCESS in 1h 08m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba46ee50ddb04dbb8215036c5491de41 : SUCCESS in 47m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/deda9edb083d45fba380c1f69a07d088 : SUCCESS in 15m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3bdbcaab1026403285300eb7b7d7b137 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f19ccc3668f443ef892fe4863769a624 : SUCCESS in 22m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/16191195bdef45dab936270fbdb21fc7 : SUCCESS in 26m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bbfca820807e4da9b10db263bffd88c8 : SUCCESS in 26m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/acf21a03e6c14009b4456fa6cb360549 : SUCCESS in 25m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a3746c1b484f4dca8ddf1a9c584e5564 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/893e516f50ad4ecaa9eec102afa8bed9 : SUCCESS in 16m 01s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/214da6092d4e4cbeb5108ad38dce2f94 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9a2d4e0033cc435894be3a6eeb4bfb41 : SUCCESS in 23m 04s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c1105f13c05f4c708230e2e36502f0c0 : SUCCESS in 17m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/480bde25d8074fffa19119490e4762e4 : SUCCESS in 11m 00s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a88c4947e4b14d5fbb57b0744053f786 : SUCCESS in 12m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e0a935b390a04194a3d39b1a62d31c7d : SUCCESS in 15m 05s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8cae14beebca444b9a656dd8dd83f6c2 : SUCCESS in 17m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/277106f763de48f5a7387a0e7bb86e3f : TIMED_OUT in 1h 01m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c566a14cc54c4fa4b64c3349e4ffef88 : SUCCESS in 34m 34s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/3a43ef807b0b458b836c91f57f93b57c : SUCCESS in 32m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8bb94a9811c34f6db0a9c0498178362b : SUCCESS in 40m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a2be50a068f434595704e7dbca59356 : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02551e8adcf04fa3b739bd8c048eeb3d : SUCCESS in 11m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdedbed30aff401995dc40e2dc80f78f : SUCCESS in 1h 01m 33s","accounts_in_message":[],"_revision_number":4},{"id":"71db73b96a4843219109d86058fecc192c28b542","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-04-01 18:33:23.000000000","message":"Patch Set 4:\n\n\u003e I\u0027m starting to think that we really ought to maintain in-tree\n \u003e systemd service files.\n\nIt was one of explicit goals of Systemd: to let project upstream maintain their unit files, instead of having distribution maintain distribution-specific rc or init files for every daemon or service.","accounts_in_message":[],"_revision_number":4},{"id":"e6e64c945a31228a97ed28bc35522e26a4a6d375","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-04-01 18:43:01.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6cb433083c5b81729dac21d859fe85def49389dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 19:48:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/875c957596a24e89899169ae18ba18f6 : SUCCESS in 10m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/65e8baea682448148835dd083cfcaba0 : SUCCESS in 59m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/48fbcbfc7c414364a3755837cd5fdc92 : SUCCESS in 56m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8357173e189b41e4ac281b47db8a2613 : SUCCESS in 10m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a5b0fcaba8df41599f240c72693d6ab3 : SUCCESS in 10m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d075e2f4c2f04f23aca57a9ea81845cb : SUCCESS in 13m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1f070e7840d49d99417a3f818b64899 : SUCCESS in 13m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/921236c3cc5c4964aa460919096f9979 : SUCCESS in 16m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e9d0227ec3864abcbf5ddff43bd6682a : SUCCESS in 21m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2cb6fca8aa95422abe363d781462260e : FAILURE in 22m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7a9ffff8d91c453a88643663311d7dea : SUCCESS in 14m 18s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/46e92b17ab9041aba392a5da1c1d389e : SUCCESS in 11m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/97d4a9674d5a4785a7b840901953ef4b : SUCCESS in 10m 47s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/8a3053f054b04d2ea01664f43be1c68c : SUCCESS in 31m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8440bae9adab401da885a38aa08f6cf0 : SUCCESS in 10m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3404e3d2c87741faac4b77a26bf8a7ef : SUCCESS in 26m 46s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2695650082a84bdeb5d8560e386f9c3f : SUCCESS in 13m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e0b46420da1c439aabc612a249235895 : SUCCESS in 22m 14s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b05bb872a2ad4b2fb6c026bfa3f2a1f5 : SUCCESS in 30m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dd8cd96e748546699f0f9fa777354c1a : SUCCESS in 28m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a105dd6020b54680b26203117967d6a3 : TIMED_OUT in 1h 01m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fbd48401c4854945b3db8fb12aff019a : SUCCESS in 39m 45s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8ec7c274a81f4f2ea208abcd7ffa8919 : SUCCESS in 31m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f239ceeb98034b729219174d7a93df70 : SUCCESS in 39m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46e8674a770e4b4281b992f6c4b633e9 : SUCCESS in 21m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27bd4fa651c847a785b5de0e5ca5629d : SUCCESS in 6m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec4c79dc5f85474fb74463b05186602c : SUCCESS in 38m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b62e8a3c28c437ba82d71a1807da591 : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":4},{"id":"5db98dcef629afd95f61386c24597a11422db453","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-01 20:37:31.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"07d7467a1dab7ab0c10f31bed020a7f74194c4a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 22:00:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/968132dbce254d4cbfdcb17a2e8fe9c7 : SUCCESS in 9m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/68c70dd130d94303b8b7fd50e86ca86b : SUCCESS in 1h 00m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f60dd65fa064886b9e9fb04fed28ceb : SUCCESS in 54m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/76f7cb6e72fb474c9ae0b7479eef06bb : SUCCESS in 10m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6e1f8246c0f24971990a0e2a5868b40e : SUCCESS in 11m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2fce10e329c8410a9d9ad9223620e971 : SUCCESS in 13m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c7f60905f4d543e6bee65418e6be6f47 : SUCCESS in 13m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b958a1b2d19e46bd8c579901dcf0c188 : TIMED_OUT in 41m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b4e8008b0f0946d4b6df53d08efeba04 : TIMED_OUT in 41m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8248c293816f447b828c220ee8e991af : TIMED_OUT in 41m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f55b326444134e9f8ca6bbc750ba8d01 : TIMED_OUT in 41m 27s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a1e49beef7ae4a6080eeaa59ea3749c7 : TIMED_OUT in 41m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/27b855d8165540189fc2ac82b7c66d4e : TIMED_OUT in 41m 19s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ec545e219961474186809dd9e64f783d : TIMED_OUT in 41m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0952efa5812149e2bfc31918de52086f : TIMED_OUT in 41m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dbc7bb0ce74f48f183101ad309e5498b : SUCCESS in 15m 07s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9cfe056a7aed4fceb7d3f258376bb526 : TIMED_OUT in 41m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a16c071d37d546eab0d3563f71dc14f9 : SUCCESS in 10m 53s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/86644b077250489a9d225f0181edc65a : TIMED_OUT in 41m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4e851863eace4480a51ade108f8c9a7a : TIMED_OUT in 41m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e989e51dbf604992ac8367d440545902 : FAILURE in 46m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91f23c95d500490e926893e304dbb89b : TIMED_OUT in 47m 33s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/01d8fd49f8a9473fa2f91dfba10d5ee7 : SUCCESS in 41m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5d86d86a6e84e2faadc5e74f01eb35b : SUCCESS in 31m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36662299cff9435387c55340255886db : TIMED_OUT in 40m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ecfaa3da787445dac712ad160113d53 : SUCCESS in 7m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/df8395185d904460a4b6fe3cfcbbfd57 : FAILURE in 48m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9953bf0f4c0445fba87196d777ad1839 : SUCCESS in 1h 14m 06s","accounts_in_message":[],"_revision_number":4},{"id":"22fc83da802b43a8bd11cfa91a9e327ea7779b4c","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-04-01 22:55:45.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4eae859c0f7e75ff1092295208953ee6028870a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 00:04:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c02dba25c2489283413db30b4db0dd : SUCCESS in 7m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5a293e1ce35549bb9d3ff93f4ec11d8f : SUCCESS in 56m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53e3f56ec0bd4eb0806c6786ac40c607 : SUCCESS in 51m 36s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/df2cd79b6eb04378bbf73ac4fdffcf03 : SUCCESS in 9m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/91ebf374780a41d5bee7ad5e76cdd24c : SUCCESS in 9m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/849477c3c4494346ad32a434e1f4a3e6 : SUCCESS in 11m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dc39b9d7d21940c99a0b402bd3a61a72 : SUCCESS in 12m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/76b1e05adb53403a80b4a2a292fdb370 : SUCCESS in 11m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ba80d43d6b0b4c65af52d9d23a086dbc : SUCCESS in 12m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5703d42575ab47cc80b85400c7355bfd : SUCCESS in 10m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2caa0199872648b682949b8215eb4f2a : SUCCESS in 14m 45s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/afb58e03a57b4d2bb1d18b279b194173 : SUCCESS in 10m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c741641eb5864a0b8c0212475027ab27 : SUCCESS in 10m 32s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/48bd16bff2384691ba3fc423c136737c : SUCCESS in 17m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dd303e0a5a5f43fa9f69c62bb18b6bf6 : SUCCESS in 11m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5f334cdd7b8f4a77ab07ece6d8761ed4 : SUCCESS in 12m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fa7eea0c667c4bb0b9eccb2709645f46 : SUCCESS in 10m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/61b37d10e9754decb13425b68bd085ba : SUCCESS in 10m 48s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/86ba60772cbf4b69944eac904b97593f : SUCCESS in 16m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9af781dc6db44379a94b711370671922 : SUCCESS in 17m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/541810e3f5874f42af1f88664d69dac5 : SUCCESS in 41m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cdb51c55e4a14826a52e53f3d3d5ee9d : SUCCESS in 30m 18s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/399064037b8141c3a47c27d064cddc38 : SUCCESS in 27m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1e2b2f672fa74977b63edf41eac13ac4 : SUCCESS in 28m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed10734fcda84d6f898bd40fb38fb349 : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ef4ea45893e4803bcb9661ef8ef3ce6 : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/720c54ad77ba4c61b4740d800532162e : SUCCESS in 21m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/574162dbb4f8462598f37d9f6b7decb6 : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":4},{"id":"2238c6198e367a1f84b5bcfcbbdcf2337717cac5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 00:05:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"011113d6f28da23e786f8836e797829374eadc67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 01:40:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ece4b6a4d904ca6b1c693ed38844bb0 : SUCCESS in 8m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c971281c995b479aa2345cd899956ba5 : SUCCESS in 59m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cc7cfcea230c41b38fac7a7ac194c7c3 : SUCCESS in 49m 35s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2a5fd991de4046e6a69a5e127a237b25 : SUCCESS in 9m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/af21361748af4a83ad774f9a239d1a5c : SUCCESS in 10m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a43f39c29454b06a59c453329b92d03 : SUCCESS in 24m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/655e9de80a6d4f909bdce5d2db54be73 : SUCCESS in 11m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d141b04cd87641b0bdb7df742e95d9ab : SUCCESS in 12m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/302d6e6a478d46a98fec81874c1f97f3 : SUCCESS in 12m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4ac846b65bcf48ffb908461e4bfbc392 : SUCCESS in 9m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9376f7d45fe340b7b06bc53de3f0cb1c : SUCCESS in 11m 43s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c6e3837e553c475a9cc4f421f90fded2 : SUCCESS in 19m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5176c00f7de14962a7d0cccaedc57995 : SUCCESS in 10m 55s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5a63700bafdd4dea94d8714e32a852b2 : SUCCESS in 14m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8fd5fcd2a6994608a5226c9f9ecb5d60 : SUCCESS in 11m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f46d8f6b3d484d89939030d4e690ff1e : SUCCESS in 9m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bbd9737d0c8a46a99f7f9a2a68511042 : SUCCESS in 10m 27s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/67566b18dcf8437fa9aa1fa776bafbe3 : SUCCESS in 15m 07s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ec8b1cbf38f442d5821022f60343a884 : SUCCESS in 46m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a8cc071dbe0418599472208b66e2298 : SUCCESS in 29m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e4df73e99f074e20b398f77a75ff0ce7 : SUCCESS in 26m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5fef31357ecd4dbeb5284671e73018bb : SUCCESS in 27m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/063a19e0636c44a1b93ee992dd4f4b66 : SUCCESS in 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eb7803172bc4be98bc1676543bc1ec1 : SUCCESS in 7m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69ea80161e804bb99e34e61806945bb1 : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":4},{"id":"767b458362babf15a432ebae3f6797de14a8a427","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 01:40:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c8747063b0088e3e015b323bf253a9c635922ea1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 01:42:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c1955e43354ca3ae20013e10a51c1c : SUCCESS in 1m 23s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/20c26d0921e94102902447706e465a46 : SUCCESS in 49s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/cb2539f5b3d44d7a80ce068224ed73ca : SUCCESS in 58s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d361e5febfbf516e8eb38871ea32dca36d686dcd","revisions":{"202161edecba4767168f0feb22f25e40a0e034f6":{"kind":"REWORK","_number":1,"created":"2020-03-27 23:58:00.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/76/715576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/715576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/715576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/715576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/715576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/715576/1"}}},"commit":{"parents":[{"commit":"712bf3c9fb3720d37e8c4f0c5b20db9bc1446747","subject":"Merge \"ring: Flag region, zone, and device as required in add_dev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/712bf3c9fb3720d37e8c4f0c5b20db9bc1446747"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.ledisez@ovh.net","date":"2015-12-17 14:40:16.000000000","tz":60},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-03-27 23:57:38.000000000","tz":-240},"subject":"Make wsgi server uses systemd\u0027s NOTIFY_SOCKET","message":"Make wsgi server uses systemd\u0027s NOTIFY_SOCKET\n\nChange-Id: Ice224fc2a6ba0150be180955037c13fc90365479\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/202161edecba4767168f0feb22f25e40a0e034f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/202161edecba4767168f0feb22f25e40a0e034f6"}]},"branch":"refs/heads/master"},"6facda025c4c259f3ac53d953bf17ff20dbe7cd1":{"kind":"REWORK","_number":2,"created":"2020-03-28 22:22:38.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/76/715576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/715576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/715576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/715576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/715576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/715576/2"}}},"commit":{"parents":[{"commit":"712bf3c9fb3720d37e8c4f0c5b20db9bc1446747","subject":"Merge \"ring: Flag region, zone, and device as required in add_dev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/712bf3c9fb3720d37e8c4f0c5b20db9bc1446747"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.ledisez@ovh.net","date":"2015-12-17 14:40:16.000000000","tz":60},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-03-28 22:21:58.000000000","tz":-240},"subject":"Make wsgi server uses systemd\u0027s NOTIFY_SOCKET","message":"Make wsgi server uses systemd\u0027s NOTIFY_SOCKET\n\nChange-Id: Ice224fc2a6ba0150be180955037c13fc90365479\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6facda025c4c259f3ac53d953bf17ff20dbe7cd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6facda025c4c259f3ac53d953bf17ff20dbe7cd1"}]},"branch":"refs/heads/master"},"061f89e4c81941df409d0cc02cff7d91db72e4df":{"kind":"REWORK","_number":3,"created":"2020-03-28 22:32:17.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/76/715576/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/715576/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/715576/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/715576/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/715576/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/715576/3"}}},"commit":{"parents":[{"commit":"712bf3c9fb3720d37e8c4f0c5b20db9bc1446747","subject":"Merge \"ring: Flag region, zone, and device as required in add_dev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/712bf3c9fb3720d37e8c4f0c5b20db9bc1446747"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.ledisez@ovh.net","date":"2015-12-17 14:40:16.000000000","tz":60},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-03-28 22:32:10.000000000","tz":-240},"subject":"Make wsgi server uses systemd\u0027s NOTIFY_SOCKET","message":"Make wsgi server uses systemd\u0027s NOTIFY_SOCKET\n\nChange-Id: Ice224fc2a6ba0150be180955037c13fc90365479\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/061f89e4c81941df409d0cc02cff7d91db72e4df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/061f89e4c81941df409d0cc02cff7d91db72e4df"}]},"branch":"refs/heads/master"},"d361e5febfbf516e8eb38871ea32dca36d686dcd":{"kind":"REWORK","_number":4,"created":"2020-03-31 19:22:59.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/76/715576/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/715576/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/715576/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/715576/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/715576/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/715576/4"}}},"commit":{"parents":[{"commit":"712bf3c9fb3720d37e8c4f0c5b20db9bc1446747","subject":"Merge \"ring: Flag region, zone, and device as required in add_dev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/712bf3c9fb3720d37e8c4f0c5b20db9bc1446747"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.ledisez@ovh.net","date":"2015-12-17 14:40:16.000000000","tz":60},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-03-31 19:22:48.000000000","tz":-240},"subject":"Make wsgi server uses systemd\u0027s NOTIFY_SOCKET","message":"Make wsgi server uses systemd\u0027s NOTIFY_SOCKET\n\nChange-Id: Ice224fc2a6ba0150be180955037c13fc90365479\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d361e5febfbf516e8eb38871ea32dca36d686dcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d361e5febfbf516e8eb38871ea32dca36d686dcd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
