)]}'
{"id":"openstack%2Fswift~716926","triplet_id":"openstack%2Fswift~master~I3a2eb7bbe4b4091d2567897904d939df1df6b251","project":"openstack/swift","branch":"master","topic":"bug/1817966","hashtags":[],"change_id":"I3a2eb7bbe4b4091d2567897904d939df1df6b251","subject":"docs: Encourage usage of UUID inside /etc/fstab in examples","status":"MERGED","created":"2020-04-02 10:44:51.000000000","updated":"2020-04-04 22:50:09.000000000","submitted":"2020-04-04 22:48:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716926-1586040510752-e8a3316c","meta_rev_id":"9d4ffbcf15d4e3afbfed08be936f541006bac2ca","_number":716926,"virtual_id_number":716926,"owner":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-04 22:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"date":"2020-04-04 10:13:57.000000000","_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-04 22:37:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-04 22:37:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 12:33:49.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-04-04 22:37:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-04-04 22:48:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb71c59fbb8d860c92442056b0c86b6be8e7e594","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2020-04-02 10:44:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f6bc6d8f90cc48a8ad6a37f3d0e8eba69a79bb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 11:03:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bae4ac51a50e4682a9bc0291c2675eba : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b874ae989ed0f11672a9463f66d5f6c2f5f2de50","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-03 00:06:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nThanks for this! One thought inline; it applies to the other files under doc/source/install/ as well. Worst case, though, we can fix it up later; it\u0027s certainly not the worst of our doc-formatting issues ;-)","accounts_in_message":[],"_revision_number":1},{"id":"5fe4d1fd6bae692b1552ffbdee46bc5cb3b4a8d1","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2020-04-03 10:04:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"abefa10deff1ff731c40cc4e6ddb1ad892ced036","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2020-04-03 10:15:58.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThank you very much for pointing out the correct styling practices. I\u0027ve fixed it and I believe it should be OK now.","accounts_in_message":[],"_revision_number":2},{"id":"4e5501656c296cc1ee9fc1c02cf55ea973e943c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 11:04:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adde7fc3ae554589bf1b2b0c3d9b0c2e : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":2},{"id":"b55ca406d29122407ce7189edd22b10f15e43710","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-04-03 12:33:49.000000000","message":"Patch Set 2: Code-Review-1\n\nRendered output that CI build doesn\u0027t look quite right to me:\n\nhttps://070dd20660dfb1eb329f-995faa2a016e4f4e5cfe3a2e70ac9f8f.ssl.cf2.rackcdn.com/716926/2/check/openstack-tox-docs/adde7fc/docs/development_saio.html#using-a-partition-for-storage\n\nWe\u0027re all gunna be RST experts by the time this is merged, lol\n\nKeep going!@!!","accounts_in_message":[],"_revision_number":2},{"id":"007639eeda000b444826da7940b9ead2e4412d80","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2020-04-03 13:44:40.000000000","message":"Patch Set 2:\n\nThis definitely looks very wrong :D Let me find how to render it locally, and then amend it. That should avoid some unnecessary back and forth!","accounts_in_message":[],"_revision_number":2},{"id":"cdc0197cb4ad7fcf10c4bc40734e341c016a6394","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-03 15:11:21.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n`tox -e docs` should build to doc/build/html/ -- then I often run something like\n\n python3 -m http.server -d doc/build/html/\n\nto serve them locally (since there are places where browsers treat file:// URLS differently from http:// ones). That might be overkill for our docs; they should be pretty stock HTML/CSS.","accounts_in_message":[],"_revision_number":2},{"id":"9740998a1cf7b9183158e437eeac5ee9dbb49539","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2020-04-04 10:09:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"97afdcb00e21f655e2b77e4af83db8c4629bbbd5","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2020-04-04 10:13:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"077b285a2096730ee6a3c70d9215cb28a5d3b50f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 10:18:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd4040e1399041c98ac0bc0c79c3993c : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":3},{"id":"11fb312159708b986a44efd7efdb8ff12afd9d3b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-04 22:37:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLooks beautiful! Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"8ec67c831c2485aa5f16976a1bfab92182512005","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 22:37:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b216d379db67b9fbce6cb71d11bef560141c64b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 22:48:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea1dd63b7d3d4971afca017da9146db8 : SUCCESS in 9m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/14c618edfabd4cbdbaa604f02a2bc2e8 : SUCCESS in 9m 17s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d1da469e46004ef6bbac78153d2963c7 : SUCCESS in 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8c6b82f3a5706659391f4990731403b3e455bdc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 22:48:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"13a69e010262980d2adab627030c301b1e5ae3d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 22:50:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6fb2cf8e1874840a118053c45dfc4c4 : SUCCESS in 1m 20s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0b199efed144450eb457229be1fb4d59 : FAILURE in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6aae6c8bb6a0455c835337c054258031 : FAILURE in 50s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"875a142980b77f2b6eacb3a4d7b8bcb8bb29eacd","revisions":{"66939c7dba6d6f980383377cb8ad51ef4fd35635":{"kind":"REWORK","_number":1,"created":"2020-04-02 10:44:51.000000000","uploader":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"ref":"refs/changes/26/716926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/716926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/716926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/716926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/716926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/716926/1"}}},"commit":{"parents":[{"commit":"9d4dc29fb37db1ad767eae3e39468d76e996af63","subject":"Merge \"Apply limit to list versioned containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d4dc29fb37db1ad767eae3e39468d76e996af63"}]}],"author":{"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","date":"2020-04-02 10:44:14.000000000","tz":540},"committer":{"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","date":"2020-04-02 10:44:14.000000000","tz":540},"subject":"docs: Encourage usage of UUID inside /etc/fstab in examples","message":"docs: Encourage usage of UUID inside /etc/fstab in examples\n\nUpdate doc examples to use explicit UUID of devices, instead of using\nunstable device names.\n\nChange-Id: I3a2eb7bbe4b4091d2567897904d939df1df6b251\nCloses-Bug: #1817966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66939c7dba6d6f980383377cb8ad51ef4fd35635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66939c7dba6d6f980383377cb8ad51ef4fd35635"}]},"branch":"refs/heads/master"},"f751dfddc73cb1f24404961fd665113528e7798f":{"kind":"REWORK","_number":2,"created":"2020-04-03 10:04:41.000000000","uploader":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"ref":"refs/changes/26/716926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/716926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/716926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/716926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/716926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/716926/2"}}},"commit":{"parents":[{"commit":"9d4dc29fb37db1ad767eae3e39468d76e996af63","subject":"Merge \"Apply limit to list versioned containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d4dc29fb37db1ad767eae3e39468d76e996af63"}]}],"author":{"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","date":"2020-04-02 10:44:14.000000000","tz":540},"committer":{"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","date":"2020-04-03 10:03:39.000000000","tz":540},"subject":"docs: Encourage usage of UUID inside /etc/fstab in examples","message":"docs: Encourage usage of UUID inside /etc/fstab in examples\n\nUpdate doc examples to use explicit UUID of devices, instead of using\nunstable device names.\n\nChange-Id: I3a2eb7bbe4b4091d2567897904d939df1df6b251\nCloses-Bug: #1817966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f751dfddc73cb1f24404961fd665113528e7798f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f751dfddc73cb1f24404961fd665113528e7798f"}]},"branch":"refs/heads/master"},"875a142980b77f2b6eacb3a4d7b8bcb8bb29eacd":{"kind":"REWORK","_number":3,"created":"2020-04-04 10:09:14.000000000","uploader":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"ref":"refs/changes/26/716926/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/716926/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/716926/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/716926/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/716926/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/716926/3"}}},"commit":{"parents":[{"commit":"9d4dc29fb37db1ad767eae3e39468d76e996af63","subject":"Merge \"Apply limit to list versioned containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d4dc29fb37db1ad767eae3e39468d76e996af63"}]}],"author":{"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","date":"2020-04-02 10:44:14.000000000","tz":540},"committer":{"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","date":"2020-04-04 10:03:12.000000000","tz":540},"subject":"docs: Encourage usage of UUID inside /etc/fstab in examples","message":"docs: Encourage usage of UUID inside /etc/fstab in examples\n\nUpdate doc examples to use explicit UUID of devices, instead of using\nunstable device names.\n\nChange-Id: I3a2eb7bbe4b4091d2567897904d939df1df6b251\nCloses-Bug: #1817966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/875a142980b77f2b6eacb3a4d7b8bcb8bb29eacd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/875a142980b77f2b6eacb3a4d7b8bcb8bb29eacd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
