)]}'
{"id":"openstack%2Fswift~720098","triplet_id":"openstack%2Fswift~master~Ic765305da394a45e73ec17dcecb58da0c0b54a40","project":"openstack/swift","branch":"master","topic":"cors","hashtags":[],"change_id":"Ic765305da394a45e73ec17dcecb58da0c0b54a40","subject":"WIP: s3api: Allow MPUs via CORS requests","status":"NEW","created":"2020-04-14 23:01:06.000000000","updated":"2021-03-01 21:11:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"21b3b1e8d08c747f4cfa901a4a2174ddadc3ee4a","_number":720098,"virtual_id_number":720098,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-01 21:11:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 01:27:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52fc39aae95854deed0ff45efec76e5771ecb095","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-14 23:01:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef7aa2bc60559f5ea9fc669b48b7d4a346053fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 01:53:12.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00b32ef630674a4c81e7dd8358427fd0 : SUCCESS in 8m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3db4390516f04788a0bd84c8800d1d29 : SUCCESS in 1h 10m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6003275b0745402b9a4fd473aca3c97c : SUCCESS in 48m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/97a0e579190f49d4b3889a63002b17c5 : SUCCESS in 14m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bff4ee5a243746c5ade63d92c1ce1893 : SUCCESS in 12m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/74e112db678341e4acc5ec3011cef451 : SUCCESS in 13m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b41fb6624f174c1f8b4aa0784a0c6b15 : SUCCESS in 14m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/167ca7bba179462593670ef4a8f49a0a : SUCCESS in 15m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/004f98d8dcb340c085f90be8a6c00b13 : SUCCESS in 14m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6623ecfc2ddb415982a1465ba28633bc : SUCCESS in 11m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a19fc2e5a5444d8a46431a2cf89e743 : SUCCESS in 15m 18s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/3d29a09f773f4cc89613d6bef10dfaf4 : SUCCESS in 26m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/af6557014a4944c589317f6c00a6e079 : SUCCESS in 15m 35s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/1056c54d77ea4e919b0298d12556b2fc : SUCCESS in 22m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e86d4c92ebab4368a574dcb2b3df4a8a : SUCCESS in 12m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8bd4fd60018943108b5954ddb57d2090 : SUCCESS in 29m 32s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9626a95581704746b96a8e7ce0d6dec9 : SUCCESS in 21m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b20b27adf10f4ebcb7a72f467d2df6dd : SUCCESS in 12m 46s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/13a3d6bbd4094708bd003d5fb1e128a2 : SUCCESS in 17m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7090ebeeb82b4f029b83efcde18fd0af : FAILURE in 19m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/523e2c866ee74aabb2af1317d819b019 : SUCCESS in 25m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c4c3516daeb64e9b8996a79b216ce0a1 : TIMED_OUT in 1h 01m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/388f6899510b4890a48d5c09c75167c2 : SUCCESS in 42m 08s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b7ace73342774245831f3650dcf94de8 : SUCCESS in 31m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bef3652a62874cbcb23e83af1bfe809b : SUCCESS in 32m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f35f2b687aa6482cbcd0676ace4674fe : SUCCESS in 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c96bd25ad704a0f8e1a8424b9c94aa6 : FAILURE in 8m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f871c825a3c4a678747e0e81e9aa4bc : SUCCESS in 25m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d74d3164dd24c8d88a00f67c3e56cf7 : SUCCESS in 1h 13m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8f4938f2e0f911ca58a05f1e1d616a5cb198c53c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-15 02:29:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e5540d99c561c225c9431e912fe6f48b57da73f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 03:50:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c553405bc9c44d6bafba51e8752603ab : SUCCESS in 8m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d91d30b2df8d4839946676233e2552d3 : SUCCESS in 1h 08m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e96fe525eca4f1693cc3b66116adcec : SUCCESS in 48m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f4984b2f4ac449dbe772476e3c2387c : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a9947acacd3a4fd9ad592d7b7b0bb5db : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0ada4b13f775434e8fc96a59d72fd7d0 : SUCCESS in 13m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cc6f33740e6142c3be6adc219ca00221 : SUCCESS in 12m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2690a28b2a6842ce8a6eb1f9cb92027c : SUCCESS in 14m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e33677010cef4008a6c984f048d69ebe : SUCCESS in 13m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e1b85e6992f245fe90b849ac5226a2ce : SUCCESS in 10m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4d9d7832fd1f47c7b5193245cc97c348 : SUCCESS in 12m 54s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4f2cfa93a7b341ac8e799068d3b4ccdf : SUCCESS in 10m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cfb50100b3e7478890642e1eee7f8514 : SUCCESS in 11m 15s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a2ed7d3be5224b79a0c7188b35dbbc96 : SUCCESS in 15m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/562c05de0149456ca70e33013437b567 : SUCCESS in 10m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6cbde5935ff244bdaa9c4697bc7819c3 : SUCCESS in 12m 50s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a94ebdcff0ed4feab123a8c8eb07e45e : SUCCESS in 10m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4656bac6c91b45469fd063a4fec9a447 : SUCCESS in 11m 41s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/26a262e8643c43f3906b2b69f1ff5712 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/24c08d95b6ff468e8e9d1031e358d309 : FAILURE in 8m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb1299f72ab94f2e9c24a5fbced3891f : SUCCESS in 21m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bb7ec1d1879f4bc0b7be94b7ca8cd6a7 : SUCCESS in 47m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b338e48fe01c46dc8af0bfd0ebf83c63 : SUCCESS in 30m 18s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/3ae58c767a1f4bcbb89cab0af27c7221 : SUCCESS in 29m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5808a0eb3f054b93abd2368b86ac8711 : SUCCESS in 31m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a000cbd078e47c78e944f5f602eee82 : SUCCESS in 13m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b02eb3fd21144349b2d2612215a6a94f : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7b2a0358010411e96eacea21bbd6e84 : SUCCESS in 24m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9b083890d8f46cba7188ba4a30a2c0d : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":2},{"id":"d80eb3930e9b228297b7970e3f34acf55feaf076","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-16 00:13:39.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5e3d0b2c5242aefa3e20ab6c49a871152c4cf058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 01:31:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc0ca3f80b5e4b8d8b6528b35caba7a5 : SUCCESS in 8m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cc6bdca73f7b430f8908fb28a0f1b00e : SUCCESS in 1h 09m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4cb9cf6eb2194467919327b0bd64a394 : SUCCESS in 53m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6caff4ca5578493591efd1f0b9442a2f : SUCCESS in 9m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/569a20dcfd7346aeaa165b804b89494e : SUCCESS in 12m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c176eacf8191436ca30ae27b7956f890 : SUCCESS in 14m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b48685c2e320403e81b12fd8deec9157 : SUCCESS in 13m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9f89e9d8cade4cd89cae9f70c947a082 : SUCCESS in 13m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d31393b9cc22407db799855b27744125 : SUCCESS in 14m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/04a0f350c640491a9c352c4feccb4716 : SUCCESS in 10m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e66643db34334140b7be680cec1ad509 : SUCCESS in 12m 43s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b2c7d344d96e4699a48686d3da1eb95a : SUCCESS in 10m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fa7da4a2c5544d99b07065dfe77a4a64 : SUCCESS in 11m 18s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/868534b878a3496d913394fca07456e3 : SUCCESS in 15m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4b02fcfb211844c69b29388fed724db6 : SUCCESS in 12m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ff22a24f8f4b411c824939aa2d1a3fec : SUCCESS in 14m 09s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/acdb00627ba74e41873e31676663c6eb : SUCCESS in 10m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e775dda772e74449987f257ea656550e : SUCCESS in 11m 51s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6dc8d2e459f04adb9cfccda7f6a1ae5b : SUCCESS in 17m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7bbaaa77c4b045dbbe2a820beaee2446 : FAILURE in 8m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b06051f12f604bd2a60e2032c84dbe10 : SUCCESS in 20m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/441f693fcb784640a8d2c323a0fc9ec9 : SUCCESS in 43m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a43ecdc2cd5404cb5f409fa4bbdb00d : SUCCESS in 32m 32s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/186243a9867d4da29035466e89160fc0 : SUCCESS in 30m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7deada10b7e4d9e9ceb1fdb14f735f5 : SUCCESS in 32m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f08653c145884cd2b1eedf5fc8090790 : SUCCESS in 14m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/350ae23d09e14303bcab0a3afedf8d52 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b0a146b5401441af981b32de249da68f : SUCCESS in 42m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afec2911cd494e7bb16fb5fb78180a54 : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":3},{"id":"9bdd41d7ea518067e164c32eb69e1890f2ed7c3c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-16 17:06:43.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"c2d13d68b5604493068bee27f3f9a6ffb5c82779","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-16 17:41:05.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"f26a754b5d12fa8827309b55b7d3951e173c27d3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-16 20:48:48.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"e4de004114f5ed5d45880d5c21d5480733505625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 22:16:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc591c25e2a43589d49053a3135f345 : SUCCESS in 10m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f3cf4efc91394ce89b58afdf34678571 : SUCCESS in 1h 12m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e1159623e934098abaed8dcf4c95f48 : SUCCESS in 1h 00m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a78a964ef7204505919350c2f62712dd : SUCCESS in 12m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a31c47794b7141419887494504fb2eca : SUCCESS in 11m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a41664d9fb5d489f812677f91e64e785 : SUCCESS in 22m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/160bce2266da445ababbf116e703a6ac : SUCCESS in 31m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6d58863dcf78480db5527bcf0d5242bb : SUCCESS in 14m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/43aaabc9a1e845d7b1a382a1e0daa610 : SUCCESS in 13m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0365221012f547f8ab6ee42d64ad9925 : SUCCESS in 11m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3b897b5587b74cb0b59db26df3d368ad : SUCCESS in 14m 56s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/e2bc966126db4853a52c56d44b9c88ca : SUCCESS in 15m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cd1cbb954c9a4f8e9bbfa018a0cf7878 : SUCCESS in 17m 22s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0c23497c281445baaca232b428a70098 : SUCCESS in 21m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/60c0e4b6d7e7423fa8a38a9a2b7c2027 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3c9e1df3fd6e44bbad0e3eb4ce37f9c7 : SUCCESS in 20m 27s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d04146bbe7554db1849cbbde9df636e5 : SUCCESS in 19m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9a831e77e24a40a3bbd788cb8064f718 : SUCCESS in 17m 51s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d31ce9a2bb7843bb8321a5758dbbf3db : SUCCESS in 23m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7f7ecd2f7733435d9dc3f4719d93b213 : SUCCESS in 7m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b02bbb9eaae41e2ac7fa97006b3f0eb : SUCCESS in 21m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/91ad00b146cd41179ec467b09ecc293c : FAILURE in 42m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8358d9156d934fc988801a57ff4da035 : SUCCESS in 30m 20s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/082a2947944948b382dfd58ca6c6db68 : SUCCESS in 34m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3fe8ae7ae8064f87bc52b8093cbdab4c : SUCCESS in 43m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad6e21d38f3e403cbe560f8c0eb0bd18 : SUCCESS in 30m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614434cc32fa474cb476d57b197d1c42 : SUCCESS in 10m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/debd80f05a4d44d38f4f3327a16e5faa : SUCCESS in 25m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7176cf32b9ec4dc3bcec3294f59e6461 : SUCCESS in 1h 05m 05s","accounts_in_message":[],"_revision_number":6},{"id":"ab28b5bd5c0daa26b878a24239f76a514fbdf0f3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-30 05:31:50.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"f2b48b5556cbb4176e9c0c4cb2ca9acfee7f3917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 06:40:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f23fd35f09d44d591cfbbe21e9efed3 : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/bbbac72dd370403f804c69f304edd747 : SUCCESS in 1h 02m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ee1611faf2a4a4880a92618fc86e7cc : SUCCESS in 45m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e70dd2a10f084316a942350a6a98074f : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c66367f5cdb64313b06bd526c1aa2a12 : SUCCESS in 8m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/55f3ef8bdcce44b7949c55db85885142 : SUCCESS in 13m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b73839201f664a8ab3c60c8e6ee8f300 : SUCCESS in 12m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d60e08bc36194ed6ba64bc83827041db : SUCCESS in 12m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e1be385b053b4d36849d0cedecd28074 : SUCCESS in 27m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4d99cad016004facbcfa64ab10753e51 : SUCCESS in 9m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/838c36110eed400fb6491b4763d8c814 : SUCCESS in 29m 45s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/02c66729ef124a2dbfdb455a471713d2 : SUCCESS in 9m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/35e7fb374e8c4256bdc6335501479809 : SUCCESS in 11m 07s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/9c84cc3746fa44e1a72d37eaee993237 : SUCCESS in 15m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d30c6a8798694598b848e12d2b4451c9 : SUCCESS in 10m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0bd5ebd7326c440dbd55b99dabef2e74 : SUCCESS in 12m 39s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/966c8c294fc448dfb15698dd1c161d06 : SUCCESS in 25m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/975dec3d2fd247d28a3c19baed4b24a6 : SUCCESS in 11m 42s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2b55dab632f84d979f45e7b845cd213a : SUCCESS in 16m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a5edf40d2b4b4ff4ab0e9894d7aefb65 : SUCCESS in 7m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6a0965b9aea24babbc966abb3be8aaa9 : SUCCESS in 40m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/287ec57eaa02493fab6cf49e026a43ce : SUCCESS in 43m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ae597a4c82743fbb59dc08456713d98 : SUCCESS in 41m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6abf46e00ce49f889343574d5530ed1 : SUCCESS in 41m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3974456cfc6b4edba2d4d5b9845c08cc : SUCCESS in 26m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63ebea6712b545a9a6e301635a4ecd72 : SUCCESS in 7m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/febf6f171e704a249a42591adb7ef94d : FAILURE in 7m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1d94184b8034bbaabaf14b0659f6d78 : SUCCESS in 1h 04m 02s","accounts_in_message":[],"_revision_number":7},{"id":"7374964b039047a553ab3f26a9b0d024df5c0bc1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-14 00:17:34.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"f147a809b7ea4b2aed39a08af8b059c73bb42a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 01:27:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a03443aa95b45529fbf0b74632428fb : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/c44d54ab6b4c4c219289f26fe42d8939 : SUCCESS in 58m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/667c0aabe4204fe7b94503b1326ca165 : SUCCESS in 49m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9294261e5a904b3aa92590ac641d4d77 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3fc91b5978f24f83bd91b0bda0f0b711 : SUCCESS in 9m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/99f57e2dc34b4efebe378b4f3b0d57f9 : SUCCESS in 13m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7048e54e08244e76b2a6753fa61f46ff : SUCCESS in 13m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d8bde9ba387e4ca2aa69c804573ad94d : SUCCESS in 14m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/502f08d3ee3a4328b22d1a4473838d34 : SUCCESS in 13m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/13ca4668146b485c950f8b8a45010bd8 : SUCCESS in 25m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/29605c4f916c45a8955a755f8e4081aa : SUCCESS in 23m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fd157cece29345e9bb4e52a863d71e8a : SUCCESS in 24m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d2eed559665e4acea8193ed55c511042 : SUCCESS in 24m 12s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bfb4add6390e4523ae6d1320a95bbb77 : SUCCESS in 20m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/401f2d51a85f4b3693eff807a5dddd35 : SUCCESS in 23m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1f6918f0b56f49a2ade28a16f9bca3fe : SUCCESS in 5m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b95cb5afcb4946e98484bd4879eb08e0 : SUCCESS in 21m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f1d4e7755ee54d18b2785a59ff74ae9e : SUCCESS in 50m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/70e311297ee84910be451e178397d286 : SUCCESS in 46m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d3fd4eb0dae540639cb6ce8005aa9be4 : SUCCESS in 42m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc0f6ec893ca479d95c86aec2736bc85 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/269731220fa04099839c6a4ca21d4d2d : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2dcb84c6f6d54adcbfb3c5d36fc3faf1 : SUCCESS in 22m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d84c7f1500044fdac399a6c8e8a6b7e : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":8},{"id":"e2e06d68a2cce526d97ae1ba5969185cf67ce4a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-21 18:42:49.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"88f1a1e6ade1d3f372cca16f33f7cde75a8ac2d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 23:10:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b810dc6d07a4a308b4168407c1319a0 : SUCCESS in 10m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/58cd73777f244f2789470d621408d538 : SUCCESS in 1h 05m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33209381d3a34ef889a12c32bfc44dea : SUCCESS in 48m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94f0aa1614d34ad48237edc5e27134fe : SUCCESS in 9m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9d7af2ff40e5449283adcf972aa1a2ea : SUCCESS in 9m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/63e01ef4e378403f8a77d63f99dcce23 : FAILURE in 19m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f6aeee7ec07d40efae7d380dc9f56044 : FAILURE in 16m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e56a1fac0188422b8c9126dd47c7bee9 : FAILURE in 16m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d95c73f82ec141f49dcfdfb0e4194e7c : FAILURE in 16m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/db55fa6ae7bf468ba1cef6f5e05407fa : RETRY_LIMIT in 18m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3fdfaaae5ca44071bc82688717a08607 : SUCCESS in 18m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c215279768f040b387ebf8a65b0b0958 : SUCCESS in 26m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/50ddba039ad54464a7d0a0f7c97410ca : SUCCESS in 20m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b64e220cdbc34e7fac29f1ac162aebde : SUCCESS in 20m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/994f1337f0e64334bd3b92ca93b89a04 : RETRY_LIMIT in 24m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/499dc8c442db432eb7f49b8e07b0262a : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f8845da705df49aca299467d183bfd38 : SUCCESS in 29m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/64813206f16548f193b0ae46f5eea19b : RETRY_LIMIT in 29m 27s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/fff886bb3dbd4b3295f511da5360a4d6 : RETRY_LIMIT in 23m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/326770cba7d84b87a37db61df5ebff98 : RETRY_LIMIT in 20m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3be04f2433974a3b93ac46ce12748b57 : RETRY_LIMIT in 28m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28a825047f634027b8ed1f763340955e : FAILURE in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70058558dbf04b3e8b79db752607e86e : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e6a180797f244cdc8a323b13d6a1142d : FAILURE in 9m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26553213baca4fe5a9eddba7350438cd : RETRY_LIMIT in 27m 45s","accounts_in_message":[],"_revision_number":9},{"id":"b96ba82dbf1a3eb565ba4b759348c91bae172490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-01 19:05:35.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"21b3b1e8d08c747f4cfa901a4a2174ddadc3ee4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 21:11:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a6139e4eff4bf199eaa70e4b1b1543 : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/741d21510ac1425e8ae3d620444056af : SUCCESS in 1h 26m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8031b169dbb14fe7a3f0c859712b73e1 : SUCCESS in 44m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/20b6bd47b50f4f05b73cbdea2d16a7d6 : SUCCESS in 9m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dc76f6b409e14382a31fbee0a1fbb323 : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/55d968400d56437989970036f7ca3c92 : SUCCESS in 30m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab10b9f9c9ee494da9869bf8c9aa0365 : SUCCESS in 14m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2c2cb97735a34e18ac0dad3e7e77791b : SUCCESS in 14m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ea0a1480e00543468eadefc60014f12d : SUCCESS in 19m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4e9d0bd2706c48a2beef80ff35591666 : SUCCESS in 15m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25d1edb49e5d4e92b372aab97c322397 : SUCCESS in 21m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c086268f1143462d8573edf05a935932 : SUCCESS in 17m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c87bf90e1d474db6929f5a81ae638ccb : SUCCESS in 18m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f4f9dedd58da4a3c8bd64252c4745b0c : SUCCESS in 18m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ffc23dfa19614bf18ceef0a05b8a712b : SUCCESS in 17m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/af180effa29549afa80d36dbc93f996e : SUCCESS in 23m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b2ac91dd18a74b4b8aa51903a797f0a9 : SUCCESS in 16m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/922d7d211234477aa37592f8f306300a : FAILURE in 43m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f645d699708f44a0878b1ae28a844c14 : SUCCESS in 1h 12m 29s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a0451b7693a647b1b6b71f6a75885300 : SUCCESS in 47m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2f66c52f914f4b6a9932af52174a2638 : SUCCESS in 40m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9adca8506b1a436bb4c1f11d4d5e6102 : SUCCESS in 35m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/357e4e8901e94ede90c72fe8108974fa : SUCCESS in 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2779dd048445461ea9084fdb06329537 : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/54cec61ba598455d89ad4ba43e05b1f5 : FAILURE in 10m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea6caefdee0244178b19b2e7fef4c6c2 : SUCCESS in 50m 44s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"325cff580436b84cf73256825f5bd1ca4486e837","revisions":{"b50493146e3d52658dea98c75b4ee70a345f8a63":{"kind":"REWORK","_number":1,"created":"2020-04-14 23:01:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/720098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/720098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/720098/1"}}},"commit":{"parents":[{"commit":"ccbf6d699213c6ff37bd163a3755d10d2ec4bf4b","subject":"s3api: Allow CORS preflight requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccbf6d699213c6ff37bd163a3755d10d2ec4bf4b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-14 22:58:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-14 22:58:49.000000000","tz":-420},"subject":"WIP: s3api: Allow MPUs via CORS requests","message":"WIP: s3api: Allow MPUs via CORS requests\n\nStill need to:\n\n* drop the `Access-Control-Expose-Headers: *` in favor of something more\n  targetted\n* at least *try* to respect the CORS settings on the container, and\n* think about how best to deal with the `+segments` container so we\n  don\u0027t need configure CORS via swift.\n\nChange-Id: Ic765305da394a45e73ec17dcecb58da0c0b54a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b50493146e3d52658dea98c75b4ee70a345f8a63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b50493146e3d52658dea98c75b4ee70a345f8a63"}]},"branch":"refs/heads/master"},"fe64def6699b4bb29ab00b2ded7227c5aea837bf":{"kind":"REWORK","_number":2,"created":"2020-04-15 02:29:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/720098/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/720098/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/720098/2"}}},"commit":{"parents":[{"commit":"17230cc509d37e2da690eee1e8b64ee6d241ed9b","subject":"s3api: Allow CORS preflight requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17230cc509d37e2da690eee1e8b64ee6d241ed9b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-14 22:58:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-15 02:28:51.000000000","tz":-420},"subject":"WIP: s3api: Allow MPUs via CORS requests","message":"WIP: s3api: Allow MPUs via CORS requests\n\nStill need to:\n\n* drop the `Access-Control-Expose-Headers: *` in favor of something more\n  targetted\n* at least *try* to respect the CORS settings on the container, and\n* think about how best to deal with the `+segments` container so we\n  don\u0027t need configure CORS via swift.\n\nChange-Id: Ic765305da394a45e73ec17dcecb58da0c0b54a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe64def6699b4bb29ab00b2ded7227c5aea837bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe64def6699b4bb29ab00b2ded7227c5aea837bf"}]},"branch":"refs/heads/master"},"e2cbd6b1261964c9f8798f252f450ff3d8fa116c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-16 00:13:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/720098/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/720098/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/720098/3"}}},"commit":{"parents":[{"commit":"6f930bbd0b163c48e28916fedac0113aaf99bb6b","subject":"s3api: Allow CORS preflight requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f930bbd0b163c48e28916fedac0113aaf99bb6b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-14 22:58:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-16 00:13:23.000000000","tz":-420},"subject":"WIP: s3api: Allow MPUs via CORS requests","message":"WIP: s3api: Allow MPUs via CORS requests\n\nStill need to:\n\n* drop the `Access-Control-Expose-Headers: *` in favor of something more\n  targetted\n* at least *try* to respect the CORS settings on the container, and\n* think about how best to deal with the `+segments` container so we\n  don\u0027t need configure CORS via swift.\n\nChange-Id: Ic765305da394a45e73ec17dcecb58da0c0b54a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2cbd6b1261964c9f8798f252f450ff3d8fa116c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2cbd6b1261964c9f8798f252f450ff3d8fa116c"}]},"branch":"refs/heads/master"},"a811eafb46c52fabd2710f248703d16ba1677e19":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-04-16 17:06:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/720098/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/720098/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/720098/4"}}},"commit":{"parents":[{"commit":"cd92d0c727d802bbce3274301e4d2b9523a72484","subject":"s3api: Allow CORS preflight requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd92d0c727d802bbce3274301e4d2b9523a72484"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-14 22:58:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-16 17:05:22.000000000","tz":-420},"subject":"WIP: s3api: Allow MPUs via CORS requests","message":"WIP: s3api: Allow MPUs via CORS requests\n\nStill need to:\n\n* drop the `Access-Control-Expose-Headers: *` in favor of something more\n  targetted\n* at least *try* to respect the CORS settings on the container, and\n* think about how best to deal with the `+segments` container so we\n  don\u0027t need configure CORS via swift.\n\nChange-Id: Ic765305da394a45e73ec17dcecb58da0c0b54a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a811eafb46c52fabd2710f248703d16ba1677e19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a811eafb46c52fabd2710f248703d16ba1677e19"}]},"branch":"refs/heads/master"},"74b9b8f28ff2c0e53d1c85edafa2541f15749fb1":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-04-16 17:41:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/720098/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/720098/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/720098/5"}}},"commit":{"parents":[{"commit":"5dd4019b226fddcb4f52267e3ca7fa53496b48d1","subject":"s3api: Allow CORS preflight requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5dd4019b226fddcb4f52267e3ca7fa53496b48d1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-14 22:58:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-16 17:40:51.000000000","tz":-420},"subject":"WIP: s3api: Allow MPUs via CORS requests","message":"WIP: s3api: Allow MPUs via CORS requests\n\nStill need to:\n\n* drop the `Access-Control-Expose-Headers: *` in favor of something more\n  targetted\n* at least *try* to respect the CORS settings on the container, and\n* think about how best to deal with the `+segments` container so we\n  don\u0027t need configure CORS via swift.\n\nChange-Id: Ic765305da394a45e73ec17dcecb58da0c0b54a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74b9b8f28ff2c0e53d1c85edafa2541f15749fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74b9b8f28ff2c0e53d1c85edafa2541f15749fb1"}]},"branch":"refs/heads/master"},"92120f478bbb2983d2f7f9db0d8b667d498c382f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-04-16 20:48:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/720098/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/720098/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/720098/6"}}},"commit":{"parents":[{"commit":"353aa7be3ac029a4f52a1ac81a6adca4c806a958","subject":"s3api: Allow CORS preflight requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/353aa7be3ac029a4f52a1ac81a6adca4c806a958"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-14 22:58:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-16 20:48:38.000000000","tz":-420},"subject":"WIP: s3api: Allow MPUs via CORS requests","message":"WIP: s3api: Allow MPUs via CORS requests\n\nStill need to:\n\n* drop the `Access-Control-Expose-Headers: *` in favor of something more\n  targetted\n* at least *try* to respect the CORS settings on the container, and\n* think about how best to deal with the `+segments` container so we\n  don\u0027t need configure CORS via swift.\n\nChange-Id: Ic765305da394a45e73ec17dcecb58da0c0b54a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92120f478bbb2983d2f7f9db0d8b667d498c382f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92120f478bbb2983d2f7f9db0d8b667d498c382f"}]},"branch":"refs/heads/master"},"be93d0f38752059304eb07fdf64826940fdf6bed":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-05-30 05:31:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/720098/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/720098/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/720098/7"}}},"commit":{"parents":[{"commit":"843d917ccb7aa0aa89f749571cb6eaa11ad1058f","subject":"s3api: Allow CORS preflight requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/843d917ccb7aa0aa89f749571cb6eaa11ad1058f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-14 22:58:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-30 05:31:34.000000000","tz":-420},"subject":"WIP: s3api: Allow MPUs via CORS requests","message":"WIP: s3api: Allow MPUs via CORS requests\n\nStill need to:\n\n* drop the `Access-Control-Expose-Headers: *` in favor of something more\n  targetted\n* at least *try* to respect the CORS settings on the container, and\n* think about how best to deal with the `+segments` container so we\n  don\u0027t need configure CORS via swift.\n\nChange-Id: Ic765305da394a45e73ec17dcecb58da0c0b54a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be93d0f38752059304eb07fdf64826940fdf6bed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be93d0f38752059304eb07fdf64826940fdf6bed"}]},"branch":"refs/heads/master"},"4e8f6937093bcd3818eb2d25a6dcd3c1a06b1319":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-10-14 00:17:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/720098/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/720098/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/720098/8"}}},"commit":{"parents":[{"commit":"99c4273ab2cf64408bdcb5d44b80e96968c80120","subject":"s3api: Allow CORS preflight requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c4273ab2cf64408bdcb5d44b80e96968c80120"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-14 22:58:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-14 00:17:23.000000000","tz":-420},"subject":"WIP: s3api: Allow MPUs via CORS requests","message":"WIP: s3api: Allow MPUs via CORS requests\n\nStill need to:\n\n* drop the `Access-Control-Expose-Headers: *` in favor of something more\n  targetted\n* at least *try* to respect the CORS settings on the container, and\n* think about how best to deal with the `+segments` container so we\n  don\u0027t need configure CORS via swift.\n\nChange-Id: Ic765305da394a45e73ec17dcecb58da0c0b54a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e8f6937093bcd3818eb2d25a6dcd3c1a06b1319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e8f6937093bcd3818eb2d25a6dcd3c1a06b1319"}]},"branch":"refs/heads/master"},"0034e3b98625c58d9e3615e8dcc61829005b0419":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-01-21 18:42:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/720098/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/720098/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/720098/9"}}},"commit":{"parents":[{"commit":"e639ff534baf1e7d59bb0e6f3a8d31e380a21619","subject":"s3api: Allow CORS preflight requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e639ff534baf1e7d59bb0e6f3a8d31e380a21619"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-14 22:58:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-21 18:42:14.000000000","tz":-480},"subject":"WIP: s3api: Allow MPUs via CORS requests","message":"WIP: s3api: Allow MPUs via CORS requests\n\nStill need to:\n\n* drop the `Access-Control-Expose-Headers: *` in favor of something more\n  targetted\n* at least *try* to respect the CORS settings on the container, and\n* think about how best to deal with the `+segments` container so we\n  don\u0027t need configure CORS via swift.\n\nChange-Id: Ic765305da394a45e73ec17dcecb58da0c0b54a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0034e3b98625c58d9e3615e8dcc61829005b0419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0034e3b98625c58d9e3615e8dcc61829005b0419"}]},"branch":"refs/heads/master"},"325cff580436b84cf73256825f5bd1ca4486e837":{"kind":"REWORK","_number":10,"created":"2021-03-01 19:05:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/720098/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/720098/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/720098/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/720098/10"}}},"commit":{"parents":[{"commit":"516e75fbb037a2fe04b945d4171fc0e6e28151bc","subject":"s3api: Allow CORS preflight requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/516e75fbb037a2fe04b945d4171fc0e6e28151bc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-14 22:58:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-01 19:04:36.000000000","tz":-480},"subject":"WIP: s3api: Allow MPUs via CORS requests","message":"WIP: s3api: Allow MPUs via CORS requests\n\nStill need to:\n\n* drop the `Access-Control-Expose-Headers: *` in favor of something more\n  targetted\n* at least *try* to respect the CORS settings on the container, and\n* think about how best to deal with the `+segments` container so we\n  don\u0027t need to configure CORS via swift.\n\nChange-Id: Ic765305da394a45e73ec17dcecb58da0c0b54a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/325cff580436b84cf73256825f5bd1ca4486e837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/325cff580436b84cf73256825f5bd1ca4486e837"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
